)]}'
{"id":"openstack%2Frequirements~746324","triplet_id":"openstack%2Frequirements~stable%2Fqueens~I01408d067f83eab4e23f6096893f0cefb8d08d4f","project":"openstack/requirements","branch":"stable/queens","hashtags":[],"change_id":"I01408d067f83eab4e23f6096893f0cefb8d08d4f","subject":"[stable-only] Upper constrain wheel to 0.34.2","status":"MERGED","created":"2020-08-14 14:11:55.000000000","updated":"2020-08-17 11:44:09.000000000","submitted":"2020-08-15 22:13:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746324-1597529619569-44e8c23f","meta_rev_id":"f181059e02dfee3f5323120adbac20ac0ef53ed3","_number":746324,"virtual_id_number":746324,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2020-08-14 14:22:36.000000000","_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-15 22:13:39.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-08-14 15:02:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-08-14 16:59:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-08-14 16:59:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-14 15:02:46.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-14 16:59:27.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-08-15 22:13:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ac6711b7d9f0e0103e51bc60341513f7f4807c8c","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-08-14 14:11:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"95ba0bef4a45f908a0cb825e9e1cde3f6b240114","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-08-14 14:22:36.000000000","message":"Patch Set 1:\n\nwell, this comes at build-openstack-sphinx-docs job runs in Queens, Pike and Ocata, but not at every project.\n\nSee failure:\nthis: https://zuul.opendev.org/t/openstack/build/3f81947267904ae0ad1e6e86625824cd/log/job-output.txt#639\ncausing this: https://zuul.opendev.org/t/openstack/build/3f81947267904ae0ad1e6e86625824cd/log/job-output.txt#676-689\n\nfor more errors, see:\n* https://zuul.opendev.org/t/openstack/builds?job_name\u003dbuild-openstack-sphinx-docs\u0026branch\u003dstable%2Fqueens\n* https://zuul.opendev.org/t/openstack/builds?job_name\u003dbuild-openstack-sphinx-docs\u0026branch\u003dstable%2Fpike\n* https://zuul.opendev.org/t/openstack/builds?job_name\u003dbuild-openstack-sphinx-docs\u0026branch\u003dstable%2Focata","accounts_in_message":[],"_revision_number":1},{"id":"4eecc0b00f64c1f861ad586fa328f4b12c9947a7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-14 15:02:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7b39e4fe940cf9f02d091c8e43d277cc86e21e67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 15:36:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ade825e92ccb4fa3b8cf76ff0088fc5e : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3ee41c0a0a214855af696a4c84606504 : SUCCESS in 3m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f1325ceff0054ceb91e95331fa13a4f0 : SUCCESS in 3m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/046ba21982d4425cbd18e5829ea98544 : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/eb03f4d84bc64602b32fc5159d807ec8 : SUCCESS in 4m 18s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/8db1e31f32c64b8abe93667c9e7bca2f : SUCCESS in 5m 58s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/6f24593894a546b0ba3a1e50a2a9e8c6 : SUCCESS in 6m 42s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/03502bd701f74f9b866d839d01e1b9cf : SUCCESS in 9m 30s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/ace7352dcbb84bf28b3f57fa807ddd93 : SUCCESS in 9m 14s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/af65fa5f0ab3439da3655c8c633c1eb0 : SUCCESS in 4m 47s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/d5d44faad15541d6ac191e9f7a4892e0 : SUCCESS in 4m 35s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/c037eb2f3e6a461a8c0d7825941ba206 : SUCCESS in 8m 43s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/979678b9a1e2477b90f121b901440c28 : SUCCESS in 7m 45s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/1d68fa5d7e854473bf7072335bf6926d : SUCCESS in 9m 32s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/bda7dd22eef847b9ab59d19e23695b5b : SUCCESS in 11m 51s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/75d692315a954f8ca30aa330beca89ce : SUCCESS in 20m 00s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/ae185a38f34a462fa1106837ef362b90 : SUCCESS in 24m 32s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/2daa9aad0b82474785d303d21d947daa : SUCCESS in 14m 30s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/a25a5265c75549338c8dcbee178885dc : SUCCESS in 9m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/18d4ff9d12444cd6b0ad728a0fbca394 : SUCCESS in 22m 55s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1839a1b5c6ec4280855058ae0e5a5ef4 : SUCCESS in 6m 27s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/d61193897f634165a785da129d5aa4d3 : SUCCESS in 4m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0b0caf8328194589bd1add5e3e4c7838 : FAILURE in 48m 58s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/8ef78f52d1b147c7ab2207b05e72a94f : SUCCESS in 1h 18m 24s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9e7762160f154e16985c7e929bdf515a : FAILURE in 32m 38s","accounts_in_message":[],"_revision_number":1},{"id":"b3a6cc5c3c719c5d43d83eb2860c2abad83b39c0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-14 16:59:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck\n\nadding this should work because generate-constraints won\u0027t remove it (custom adds to upper-constraints would be removed from master)","accounts_in_message":[],"_revision_number":1},{"id":"b65a205c2ced01e4a84e6bff2d78bb3005aed923","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 18:27:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/deae4e0708dc42a2a3a74bb103a6d5a8 : SUCCESS in 5m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/24a904d696ce47feb2b8b41bce70f2d3 : SUCCESS in 4m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/91fa854be2dd4abaa0444e917d0938df : SUCCESS in 6m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1b736f1a71e64536ae634f9379993912 : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1b671de211e24bef8d9f57c0791e83ea : SUCCESS in 4m 58s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/3ec6bce3046c4f5db099e40886bde863 : SUCCESS in 8m 49s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/f9edbfedbd7840f8a24c2367da650ee5 : SUCCESS in 6m 18s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/9b2d89418d314799a47fd0ae251e1100 : SUCCESS in 9m 18s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/7160ba41c5094676aa4631caec57563c : SUCCESS in 9m 17s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/019bb03a0ccf4c34aae8c79c9317577d : SUCCESS in 5m 20s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/62f3b5ab8dc54a3288adcda92e3aa971 : SUCCESS in 5m 44s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/a2e5e74c123644409f255fbfb3089cc9 : SUCCESS in 9m 29s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/1cf6260d5cc043f6bef80eaf0c1af633 : SUCCESS in 10m 44s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/edfc968b2e9749668edf187ccae5b893 : SUCCESS in 12m 01s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/872980f304c0414c8191173002eaa79f : SUCCESS in 11m 57s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/95ccfbaabacd4cf2964f3fde90c51ae0 : SUCCESS in 33m 32s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/f85cca9bd52f495ba3eecb9fbe038f86 : SUCCESS in 34m 52s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/ac5d8b71a7ba4d709ac7c3ade287cf0a : SUCCESS in 13m 31s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/68ddf053f4e04242b558086ec4d90aeb : SUCCESS in 24m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/557acb8f71094cb99939aae0db1cfa39 : SUCCESS in 19m 21s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4a2df8e511244c4d8df11cbfdebd72aa : SUCCESS in 6m 30s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/e0d8894637a2454abebcaf3bcd513bad : SUCCESS in 4m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/265431f3647d45bc922989b30df1fa3c : FAILURE in 54m 29s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/c627287bfb1f4e68a45c688c3b1fc29f : SUCCESS in 1h 25m 32s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5fce11c4700a4698a6af7fe393bfa9a4 : FAILURE in 42m 54s","accounts_in_message":[],"_revision_number":1},{"id":"a7ee30ffba4d7786f6193af93f481e4f5a13d0f5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-15 19:39:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cd19de4b961ff7ca1c5f3caa43937a1e17a5eba1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-15 20:52:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6943b0c99b034a38b72a00e284a9aa52 : SUCCESS in 3m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e3e03a8b56b34aef8162913731c71580 : SUCCESS in 3m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2061feebb36f4a818f7da4abe550e5d3 : SUCCESS in 3m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bab7ce95f0e1412d9e0d652a8dde2d50 : SUCCESS in 3m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ac34c813820145a29265753c97847108 : SUCCESS in 3m 20s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/4eed4a8b4ff547bf94a215e1c76955ba : SUCCESS in 6m 03s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/b33e63d8a17c453db931cbab5a3bb629 : SUCCESS in 8m 16s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/1c8eb492eb6f42e49c0063470eaa12a5 : SUCCESS in 6m 54s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/886ddf4fa6184798b097f40a5d516d55 : SUCCESS in 14m 05s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/b50336bd5a634aafba30dc4c31f89706 : SUCCESS in 3m 55s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/20b1c98f03c64da890affd6e4bb36f85 : SUCCESS in 4m 58s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/c311e4c46f114af99668e150b0b61091 : SUCCESS in 6m 04s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/dc7e11077beb44daa5f7dd34a88ec75f : SUCCESS in 9m 50s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/bd5dd35f3c7044bd821d10466f82f190 : SUCCESS in 9m 42s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/40703f450f7b42248054b23ab889a2c3 : SUCCESS in 15m 39s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/61c7574a9d78409497de1d4eaa14bfbb : SUCCESS in 13m 02s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/1638bbae2a8a48b7912c698de3d5c998 : SUCCESS in 28m 20s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a0a635e629a140128eea0c98d0e9b919 : SUCCESS in 8m 40s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/918bf25106bc48368c8190b6ad341772 : SUCCESS in 13m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ed6ead5b1dc6465590fc90c3fbac6337 : SUCCESS in 14m 05s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9e7217cb38f943208af9d1370648327d : SUCCESS in 8m 18s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/260e5bb3af8b4b3bb9df77ffd3973c51 : SUCCESS in 3m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b05dc56d6723496eb6ba2cc73307f2cd : SUCCESS in 1h 02m 26s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/5f019a36398d4062800f5577b1ef584a : SUCCESS in 49m 54s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b8f3f4790aa34dc2ac811c58b305a022 : SUCCESS in 1h 11m 40s","accounts_in_message":[],"_revision_number":1},{"id":"da084a6fcc790d287433786053a6588d485709a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-15 20:52:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9c7ec279f6702dcbd7ecb29d309cc6521ac78ec0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-15 22:13:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e608a6ed621a19889ecab5d8866eb4fed97a773f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-15 22:13:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91850cd8ee924220beaaf2fd095f9054 : SUCCESS in 3m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9a3a4ea84d8b424daaf7e312ea8807f6 : SUCCESS in 4m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f1c2eab3af6c4ffe84744b53d2c92017 : SUCCESS in 4m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/292870b0ccb14d21973aefacb3307482 : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fd262eb9e9eb4de6914019aa6dda7582 : SUCCESS in 3m 57s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/a8aa7a1a78f84c089f4412d6dd649687 : SUCCESS in 7m 51s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/db0db553445a4a96b8b50a6b8032f21a : SUCCESS in 8m 10s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/5f47c4a95d0d470bb1e5908db6737965 : SUCCESS in 9m 24s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/a0f9230a65e04f37b3fb0ca7744578f4 : SUCCESS in 8m 54s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/23ecb5f4ca024881bc423bc1414b6987 : SUCCESS in 5m 47s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/2e11cd0eefd645fbb428c98372d7df45 : SUCCESS in 5m 38s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/301cea281c324f2f8655a7fbec671b7e : SUCCESS in 10m 39s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/d50fab739ede4b9ab50b29f2e39ab6c1 : SUCCESS in 9m 04s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/5c7c688f1fcf439fad629072512088e6 : SUCCESS in 18m 58s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/2571b91cb9cb4da0a92829de25cd7c44 : SUCCESS in 18m 43s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/6686850ec5c443a1a7194924cc7b4cd0 : SUCCESS in 20m 17s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/39ba481fd8a64ac0a30d24315806d16d : SUCCESS in 25m 17s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/9a676ce50dbe466a878d345999a80e7a : SUCCESS in 14m 07s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/4c45d701f6a146bcaba52678106aa917 : SUCCESS in 14m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fe5eec17b5fb4e03bd18f6dcb06592b2 : SUCCESS in 18m 40s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a45c5eea0a6c44d5955aaa61d9fe35ec : SUCCESS in 5m 22s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/5e0aeb50f2144e75899c9ee6d0dfaecb : SUCCESS in 4m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cc1c98b653cb4d2393a2d25bcc969ef1 : SUCCESS in 1h 00m 49s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/836c3c89a8e7489191ec99d362715464 : SUCCESS in 1h 14m 36s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/cd7c5413db0048eb9166b29525418e25 : SUCCESS in 59m 29s","accounts_in_message":[],"_revision_number":1},{"id":"0a339c0da464f677b2617c5ade812fc178603884","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-08-17 09:26:01.000000000","message":"Patch Set 1:\n\nShould we revert this now for consistency reason, as we don\u0027t need it anymore? Wheel 0.35.1 was released [1] which (maybe as a side effect) solves the issue for us [2] (changelog is not up-to-date, but at least it shows the relevant change).\n\nLet me know about what do you think and if you say so then I\u0027ll push a revert patch for this!\n\n[1] https://pypi.org/project/wheel/#history\n[2] https://wheel.readthedocs.io/en/stable/news.html","accounts_in_message":[],"_revision_number":1},{"id":"e10c3435ac017f1ac9c45ddb4ff1d7d2a47172d4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-17 11:03:26.000000000","message":"Patch Set 1:\n\n\u003e Should we revert this now for consistency reason, as we don\u0027t need\n \u003e it anymore? Wheel 0.35.1 was released [1] which (maybe as a side\n \u003e effect) solves the issue for us [2] (changelog is not up-to-date,\n \u003e but at least it shows the relevant change).\n \u003e \n \u003e Let me know about what do you think and if you say so then I\u0027ll\n \u003e push a revert patch for this!\n \u003e \n \u003e [1] https://pypi.org/project/wheel/#history\n \u003e [2] https://wheel.readthedocs.io/en/stable/news.html\n\nSure, I would say if it is not needed anymore, it probably would be good to revert it.","accounts_in_message":[],"_revision_number":1},{"id":"f181059e02dfee3f5323120adbac20ac0ef53ed3","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-08-17 11:44:09.000000000","message":"Patch Set 1:\n\n\u003e Sure, I would say if it is not needed anymore, it probably would be\n \u003e good to revert it.\n\nThanks, revert patch proposed: I1959fd598eae7da1f9a531cc9b9c27f9558c8e0a","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9bab19ff893be201b3124b797107ca3b948f0877","revisions":{"9bab19ff893be201b3124b797107ca3b948f0877":{"kind":"REWORK","_number":1,"created":"2020-08-14 14:11:55.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/24/746324/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/24/746324/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/746324/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/746324/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/746324/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/24/746324/1"}}},"commit":{"parents":[{"commit":"bc7833bb37d89a0973cef88621ba7ade3be574a1","subject":"Ensure pip is installed locally","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bc7833bb37d89a0973cef88621ba7ade3be574a1"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-08-14 14:07:01.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-08-14 14:11:30.000000000","tz":120},"subject":"[stable-only] Upper constrain wheel to 0.34.2","message":"[stable-only] Upper constrain wheel to 0.34.2\n\nwheel is not upper constrained but packaging is constrained to 16.8.\nThe latest wheel (0.35.0) requires packaging\u003e\u003d20.2. Strangely the pip\ninstall does not fail, but later causes this error:\n\nFile \"/home/zuul/.venv/local/lib/python2.7/site-packages/wheel/bdist_wheel.py\", line 23, in \u003cmodule\u003e\n    from packaging import tags\nImportError: cannot import name tags\n\nChange-Id: I01408d067f83eab4e23f6096893f0cefb8d08d4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9bab19ff893be201b3124b797107ca3b948f0877"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9bab19ff893be201b3124b797107ca3b948f0877"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
