)]}'
{"id":"openstack%2Frequirements~746434","triplet_id":"openstack%2Frequirements~master~Id6853599492ad869d25ee8b7f99afa9416926c2b","project":"openstack/requirements","branch":"master","topic":"new/adjutant_confspirator","hashtags":[],"change_id":"Id6853599492ad869d25ee8b7f99afa9416926c2b","subject":"cap toml library because of nested dict bug","status":"MERGED","created":"2020-08-17 03:40:53.000000000","updated":"2020-08-21 10:08:00.000000000","submitted":"2020-08-21 10:06:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746434-1598004392632-3ae6a7cc","meta_rev_id":"7346b61ba309b0ad83d0f3183e0bdf164af3d55c","_number":746434,"virtual_id_number":746434,"owner":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-21 10:06:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-08-21 05:17:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-08-21 08:39:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-08-21 08:39:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-21 05:17:40.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-08-21 08:39:26.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-08-21 10:06:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3afd4db859de197e255606bea00aaf6ad89dd3aa","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2020-08-17 03:40:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4e4ec299ec7dfdf6b679b95e81caded9cdbf4f33","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 04:47:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4680f06602bc42d385fe8dafa593d7a6 : SUCCESS in 1h 02m 27s","accounts_in_message":[],"_revision_number":1},{"id":"0c0b275df5498c8914bde2d2b63266bec82bfedc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 05:03:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac20c3f0efa84983a9696864353e4558 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fce792de7e874afc93a2a9cf5d5e53ee : SUCCESS in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9e5c1b5a8f24cdba30cf71b49ddff73 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de0bc77804114696ada523744c27007e : SUCCESS in 3m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d65b90c176dc42df891bde529a984073 : SUCCESS in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/47c38e8704a34ba381200dc37df1bea2 : SUCCESS in 3m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/969409c9c1644258b54ad3dd5ed31566 : SUCCESS in 5m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/95b6aa00ea474c34bbc724f33f58f8b3 : SUCCESS in 8m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/41f3df58a79349c9ba216f0b8e6d11b1 : SUCCESS in 8m 52s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ddcaf85ac0ab478ab724901b007c225b : SUCCESS in 4m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/984e5abb41fa42cb813ce0cd1006ef98 : SUCCESS in 10m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/34a1c5c5143840d1b041ae656e9c7610 : SUCCESS in 6m 30s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8de45582ced24eb28a284c083cd2b047 : SUCCESS in 6m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b17235c49bc84e75885eedb71578236c : SUCCESS in 12m 07s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cd8e9f6031bc46f0bad471d0ee983e94 : SUCCESS in 26m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/82cdae7fdefb430fa9e22c17c48fc70b : SUCCESS in 11m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3e0cf1f03aee4c5da314a90cfd516c94 : SUCCESS in 3m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/97fb53ca3ebc4ab6992b6f82fe2966d6 : SUCCESS in 16m 11s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6e836a7085f84ed3aaeba633965114b6 : SUCCESS in 4m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2fdaeb8f30d34d56a50e36fcaaf498bf : SUCCESS in 5m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d4f4ae5bf50b4b13ae9302ac8b1f7a7f : SUCCESS in 8m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f522f052fb114cadbc8d3b3158a2e2fb : SUCCESS in 54m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3255cd595afb48deb8fafb0e46506ae2 : SUCCESS in 1h 16m 27s","accounts_in_message":[],"_revision_number":1},{"id":"52d3925de0e549be2112c194413807c33592bff0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-17 19:01:25.000000000","message":"Patch Set 1:\n\nIn general I\u0027m ok with this but I think we should track which libs are not officially approved but only in global-requirements.txt due to needing to mask known bad versions.","accounts_in_message":[],"_revision_number":1},{"id":"084727ac331f7e54e24b9c7f1dbb8a6a82fb1ea0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-19 10:17:32.000000000","message":"Patch Set 1:\n\nLGTM\n\n \u003e In general I\u0027m ok with this but I think we should track which libs\n \u003e are not officially approved but only in global-requirements.txt due\n \u003e to needing to mask known bad versions.\n\nI\u0027m not sure to seen what you want to do and what we need to update. I suppose you want to keep 0.10.1 in upper-constraints.txt, right?","accounts_in_message":[],"_revision_number":1},{"id":"64fc9ddb65fb18f98cdba38ea55521e073887453","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-19 15:39:23.000000000","message":"Patch Set 1:\n\nheh, thanks for jogging my memory, move the toml entry in global-requirements.txt to the bottom into the section headed by the following line\n\n# NOTE(prometheanfire): python3 caps, are not approved for use in OpenStack","accounts_in_message":[],"_revision_number":1},{"id":"b9044b08b88dcf38366c74469e91c6d8e4d2aa3d","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2020-08-21 05:15:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"76815832775c631ed663748769aad6a5939e1176","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-21 05:17:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"21ee4036388e463edb91ea283c3f5e9804db8173","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 06:30:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5eadea774529480d9db12f97f2e0085c : SUCCESS in 1h 09m 16s","accounts_in_message":[],"_revision_number":2},{"id":"aad0c48640c3b8b863551ce61b9d77b423e7e135","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 06:40:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76d029215da848f0a019bf5d3ee83a7b : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbd1c989a1df4a0598e79249f15511a2 : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7fc33a3c2b3547a498771f5870fe6e1b : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37ef53f4cb5a4bdeabcc073dd318678e : SUCCESS in 3m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/62523da38ffb4a8ca8db765eaf44a483 : SUCCESS in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bda01eda07f24a6ba330f69de110bc4a : SUCCESS in 3m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/75764ea75e5a4efea655fd55c008395c : SUCCESS in 6m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/309441a93c38489390ea1e5740fd91c3 : SUCCESS in 8m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2c0d1d6dc089430e9f593ed01953155c : SUCCESS in 7m 33s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ad074c150e584800a17be8133baf80a1 : SUCCESS in 4m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3733cef04a7a4f2f8ec6c194061bb3df : SUCCESS in 10m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/af3325e6f44f41af84aafe1f70ce0c83 : SUCCESS in 6m 05s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ac4b05ebc7954e62a7a722f9f2b6aa37 : SUCCESS in 5m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/910fa8070916408fa4164d98cb1d3b77 : SUCCESS in 11m 25s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a9fed9cf7b854b9a838553e830d53b5d : SUCCESS in 24m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/66d2cfc7f8df408aa97ae0b89ad939fe : SUCCESS in 11m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7f812bc8afcd474bbc16ac12bb070aae : SUCCESS in 2m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dd4d4f910f064c41bedcbb244234d0a8 : SUCCESS in 16m 44s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2aeee29c3c754958ae5ee83e3029959c : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5b844f2c12ab428c881b3ca1db5762e2 : SUCCESS in 7m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5b44e65703f04f04966e326808d5ccfe : SUCCESS in 9m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f447d7cea23b49d5a823d7f87e939e89 : SUCCESS in 56m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8da0ae0c178d4e4daef1736295b8a588 : SUCCESS in 1h 19m 12s","accounts_in_message":[],"_revision_number":2},{"id":"c912bd6642f70e68c44a3f1fe681e47a806ffda5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-21 08:39:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"36dc2479f95f4dd1d6b4969e4d86b2778188e1d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 08:39:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"30022472f340036f2edaa417f9e808eb1bd4c962","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 10:06:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/980c1ffaa749475d92156d827fe72ced : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55bdb0c1c7764fe88ebbd5ee94ab9f9f : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47b8b142d4b5403e9e143f6a283d6f94 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/447cb0c7b4ba45abb468b8559cdc269f : SUCCESS in 5m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8458a1be34874b0398649f1e5c6adf2a : SUCCESS in 4m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cdc533fe087b4cb98776200b9f6a2f33 : SUCCESS in 6m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9371aff9d0774d54bfa51ff72ad513d7 : SUCCESS in 9m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e3c165b849544518aa3a15fb4b15a672 : SUCCESS in 10m 54s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/00f64ea6b0774cd9a3bf986258088904 : SUCCESS in 10m 31s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a0ed148432c1486e83d96d5772a3247a : SUCCESS in 6m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/40c77c5767fb4235aafba9ed7d3e48b3 : SUCCESS in 11m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b53be78d74cb4f94a39b0cbcd0246dbf : SUCCESS in 8m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/74123e9acb0842a0a6ecc57c0fe7ef43 : SUCCESS in 7m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8f6397a1eb464cd4ad5ce4982b05bef7 : SUCCESS in 29m 41s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4c48694258924396916d7b470b8d2766 : SUCCESS in 29m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/35dc762eda7b462da975f322e9525e71 : SUCCESS in 19m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ac6093ff1f144bd5b57df146ee38bcb4 : SUCCESS in 4m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4862309efb1441dc95a66cfd52b3966b : SUCCESS in 17m 26s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4c0a6693f5ff41bea637c101efb94dd6 : SUCCESS in 5m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2f3e89febf384e2cb6342bdd46d32b6c : SUCCESS in 8m 05s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4646bc7778e54a8381a32c7f34255441 : SUCCESS in 11m 44s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ab1f4436dd79484da39d05e292352231 : SUCCESS in 10m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb31ed81d1644423bc2215ff175c3276 : SUCCESS in 56m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0440343de15042beadf58ca756938919 : SUCCESS in 1h 20m 34s","accounts_in_message":[],"_revision_number":2},{"id":"823592f1501629e759181f8372d56cf017495f2e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 10:06:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"37022c3f0d25114dc919f117171ed0141516cf7c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 10:08:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0633db139c0f4544b7b321791e3c0ea6 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"293e6502fa960bc21db55d96cc9071a052007d1a","revisions":{"b5f0397c1fa87a28ed1626630be247769e78d4c5":{"kind":"REWORK","_number":1,"created":"2020-08-17 03:40:53.000000000","uploader":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"ref":"refs/changes/34/746434/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/34/746434/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/746434/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/746434/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/746434/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/34/746434/1"}}},"commit":{"parents":[{"commit":"68b891c85e247e09733b8577490d4484743814d5","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/68b891c85e247e09733b8577490d4484743814d5"}]}],"author":{"name":"Adrian Turjak","email":"adriant@catalystcloud.nz","date":"2020-08-17 03:36:33.000000000","tz":720},"committer":{"name":"Adrian Turjak","email":"adriant@catalystcloud.nz","date":"2020-08-17 03:40:50.000000000","tz":720},"subject":"cap toml library because of nested dict bug","message":"cap toml library because of nested dict bug\n\nversion 0.10.1 causes a bug with how nested dicts are handled\nand we should stay with 0.10.0 until it is fixed:\nhttps://github.com/uiri/toml/issues/295\n\nChange-Id: Id6853599492ad869d25ee8b7f99afa9416926c2b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b5f0397c1fa87a28ed1626630be247769e78d4c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b5f0397c1fa87a28ed1626630be247769e78d4c5"}]},"branch":"refs/heads/master"},"293e6502fa960bc21db55d96cc9071a052007d1a":{"kind":"REWORK","_number":2,"created":"2020-08-21 05:15:39.000000000","uploader":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"ref":"refs/changes/34/746434/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/34/746434/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/746434/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/746434/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/746434/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/34/746434/2"}}},"commit":{"parents":[{"commit":"258cf2cb18e66b70511662402dbb6652a5e0fc31","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/258cf2cb18e66b70511662402dbb6652a5e0fc31"}]}],"author":{"name":"Adrian Turjak","email":"adriant@catalystcloud.nz","date":"2020-08-17 03:36:33.000000000","tz":720},"committer":{"name":"Adrian Turjak","email":"adriant@catalystcloud.nz","date":"2020-08-21 05:14:54.000000000","tz":720},"subject":"cap toml library because of nested dict bug","message":"cap toml library because of nested dict bug\n\nversion 0.10.1 causes a bug with how nested dicts are handled\nand we should stay with 0.10.0 until it is fixed:\nhttps://github.com/uiri/toml/issues/295\n\nChange-Id: Id6853599492ad869d25ee8b7f99afa9416926c2b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/293e6502fa960bc21db55d96cc9071a052007d1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/293e6502fa960bc21db55d96cc9071a052007d1a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
