)]}'
{"id":"openstack%2Frequirements~746436","triplet_id":"openstack%2Frequirements~master~Ibdb8785b3fa6234c8fadbeaa3a672eda373c7707","project":"openstack/requirements","branch":"master","topic":"new/adjutant_confspirator","hashtags":[],"change_id":"Ibdb8785b3fa6234c8fadbeaa3a672eda373c7707","subject":"Add CONFspirator to requirements","status":"MERGED","created":"2020-08-17 04:11:04.000000000","updated":"2020-08-29 01:46:27.000000000","submitted":"2020-08-29 01:45:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746436-1598665506652-9d651cf1","meta_rev_id":"30285173596b0b98520ed41217a080677db6528c","_number":746436,"virtual_id_number":746436,"owner":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-29 01:45:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-08-28 21:22:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-08-28 21:24:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-08-28 21:24:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-19 10:07:15.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-08-28 21:22:31.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-28 21:24:31.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-08-29 01:45:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"365e9bfc4e672d29335b166b9060111b2907ff2c","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2020-08-17 04:11:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"efd8522ada844506111592e6b6048c6a691d70b1","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2020-08-17 04:12:35.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9d98ecfd3878431ec986a84c5500a2eb80efd1ff","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2020-08-17 04:19:12.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"dcfd7282a3aa915c65125dbf9b877e9babbdd23e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 05:25:01.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f1ef15efce364d1c9c7acf7ae9ce1555 : FAILURE in 1h 05m 12s","accounts_in_message":[],"_revision_number":3},{"id":"6fa46f24a16b4b27cf4f90da35b64a024726914a","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2020-08-17 05:26:42.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"e7d87a86e4fe34ee1db428d0f440095ec97a54b1","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2020-08-17 06:08:00.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"87ba028376b64635c955ea9975a5c7bc0c1d4b43","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 07:41:45.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/11c99628581d451abf59028e31571036 : SUCCESS in 1h 25m 42s","accounts_in_message":[],"_revision_number":5},{"id":"ae2d28a9e3d89f02b579a0715f2c1868b88a2121","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 08:21:23.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ac3c418a84e477a9a3fa21381aaf8b5 : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a01e33245b54823a4eaefc73f4d9eb7 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d47740fa39942109c8b2bff1bd7c51e : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2bb7ad08e7848b295b1fb9a8a2d96f9 : SUCCESS in 6m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7d94259aee8845e39eeaccbb530f06cb : SUCCESS in 5m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dc839e7bc4fb4be7825d2a3b92bc74b9 : SUCCESS in 4m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4c9ed8de1a2b41b7b1805ed88856f256 : SUCCESS in 8m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0315209b9b2a4ea5b18ce6f0e3a7c895 : FAILURE in 5m 32s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/18504ed01c5a41fbb15e14a34fd170e6 : SUCCESS in 9m 32s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a5d00db5458f46459cbef8966b20937e : SUCCESS in 5m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/82453e40dcc84edfbdacde968c14c0e8 : SUCCESS in 11m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/287d60bf5c9944059b234e0967fd24e8 : SUCCESS in 8m 41s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5180f9ffd39e4429a4dade646275ca6d : SUCCESS in 9m 31s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8f06de9a5fb046de9e627acc1500a156 : SUCCESS in 14m 10s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9474f61731e242dc8d4b7326a32056a8 : SUCCESS in 31m 41s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/13cc7382e8574fb4ae5d421ec6a9f474 : SUCCESS in 14m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/632a88cce11747cda53d2188f31e370c : SUCCESS in 4m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2316614cdcb54ad38e394916f5a819bc : SUCCESS in 20m 42s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/bf2d6c77f3ea43bc843e68bf86d25e08 : SUCCESS in 6m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bdea4e4e38164629af985ff79cd3ad32 : SUCCESS in 7m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5838a8865755406eb5e4f8efe8fb36cc : SUCCESS in 13m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/96655769070a4ce79434499a9bc23b4f : SUCCESS in 42m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3d90d0ebbdb948789443cb286e74fc13 : SUCCESS in 1h 19m 08s","accounts_in_message":[],"_revision_number":5},{"id":"3379fbf4e3352b75712a0055e0cc423a785328a0","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2020-08-17 23:06:15.000000000","message":"Patch Set 5:\n\nRECHECK","accounts_in_message":[],"_revision_number":5},{"id":"089b2cef3fcfc4a0ce4f0fe66e943d1a63137c53","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 00:19:14.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/0d7489cbab1f426ea4a20eacf46fdd98 : SUCCESS in 1h 08m 19s","accounts_in_message":[],"_revision_number":5},{"id":"702124721ced10bf3a765fe8af08f92d082b02f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 00:47:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6e5a1b8504643c39005a9a36d251109 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/172fac5d9464434f980f71f43964fb0c : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93a09e9bddc14848b4e2c73b73c3c182 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a758bf47dc54084b8d9b0b7545feb1f : SUCCESS in 4m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f17fefd8626e438996a8cfaaca451b00 : SUCCESS in 2m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/408909436b704a64b8bd3da212223a66 : SUCCESS in 3m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8237fd2483c949b38df9da388f246077 : SUCCESS in 7m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b79ee2c21c694868af9eaa9d2bbbcdc0 : SUCCESS in 7m 06s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/205f650336704c87a8838fe830f5799b : SUCCESS in 8m 16s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6bcb86092f554c319c610d2e87520ba6 : SUCCESS in 4m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d6dc8ab8af50456cbdf257fdcc7f8da1 : SUCCESS in 12m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/41afc2b719194324a91fde77a0249009 : SUCCESS in 5m 58s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/78598a86342042f198d0d6b796e9191d : SUCCESS in 7m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9288d32ef973439fb133cb03c62903c8 : SUCCESS in 11m 41s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ed43bd40d979433e80892c282aca79bd : SUCCESS in 26m 19s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0825207a622540b2be53d75bb68cf1bc : SUCCESS in 8m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5665f30624c54dcc831fddebdeb25235 : SUCCESS in 3m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4717b61890f54ee491977b4dafcc8b7d : SUCCESS in 12m 05s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/55d263752348438b9ba40b8c340152e1 : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/99a61d5ea6f843f5b02e46c50d721d74 : SUCCESS in 5m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/97e412340bbc44309d5a2d8f9e78f571 : SUCCESS in 9m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2aaf4837513d4b36a83b9674e048bb24 : SUCCESS in 40m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/40cd512c2131491181817a2123ae6e5f : SUCCESS in 1h 34m 05s","accounts_in_message":[],"_revision_number":5},{"id":"fc6f1065369b01f2b54d3682f328fd4435944537","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-19 10:07:15.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"0642ec498147b28617b2544bbed359ff752f1928","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-19 20:37:31.000000000","message":"Patch Set 5:\n\nHave you spoken to the oslo team to see if your assumption about not supporting these things in oslo.config is true? I\u0027d much rather we keep things there and augment that library than add a custom alternative.","accounts_in_message":[],"_revision_number":5},{"id":"f9b153962ac598c5daf80435d20ccb360ecf9f6d","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2020-08-20 04:23:47.000000000","message":"Patch Set 5:\n\n\u003e Have you spoken to the oslo team to see if your assumption about\n \u003e not supporting these things in oslo.config is true? I\u0027d much rather\n \u003e we keep things there and augment that library than add a custom\n \u003e alternative.\n\nI\u0027ve posted in #openstack-oslo again to see if anyone has an opinion.\n\nI\u0027m fairly certain I asked in the past, and either got no really useful responses, or what mostly made me assume that I was better off just building my own thing.","accounts_in_message":[],"_revision_number":5},{"id":"4fc4117a779d962e49a192252e57d4ff0f8ac08b","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2020-08-21 05:01:47.000000000","message":"Patch Set 5:\n\n@Sean \nhttp://lists.openstack.org/pipermail/openstack-discuss/2020-August/016715.html\n\nMy reply to Ben, and I am still of the opinion that having this as a separate library is probably saner.","accounts_in_message":[],"_revision_number":5},{"id":"ab6b32f5bb002f8e3790066d511855d8d8c2ceb5","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2020-08-21 05:15:39.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"05cf266b07d889941612cebe1f4f9b9438c4babe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 06:33:05.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/0725768bac1c4252950565914541ed85 : SUCCESS in 1h 11m 15s","accounts_in_message":[],"_revision_number":6},{"id":"a5505864685eef3a9faa34ddca322cfc8046f2ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 06:49:35.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/583e5dde03a34fbb9c42d3a855db6515 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3e2b71445964120b47ac00815511437 : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8d103644d3942b1b2c22933084b8f8b : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/162d5a01678d4870ae2cbce2df347806 : SUCCESS in 4m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5affcd6b98b545e5b61d6f3b5e1fc86e : SUCCESS in 4m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e6b3ab2e64d64f22863dda525c02cc00 : SUCCESS in 3m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/eb7d919a2aff42c08ec06401e296525b : SUCCESS in 7m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cacb50f0cdd443388e3efe6881855e23 : SUCCESS in 9m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/20aebc08d4494814b85870b59a87727c : SUCCESS in 8m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/870d9082a4774ba391edbe2591b11147 : SUCCESS in 4m 09s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d7ac448eda2248f382700199282c6613 : SUCCESS in 9m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/78dff6d2232e42e2abf82efcf4b8c9ca : SUCCESS in 7m 21s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9a5de87a498440a5a008a279f58496f4 : SUCCESS in 5m 35s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a20bc21c0c3948bfae3517fee54dfb67 : SUCCESS in 11m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/346eef7493fa41488aef707c976d4a0a : SUCCESS in 26m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/455dbb033aa54b92a388ee0a7e8a14b2 : SUCCESS in 12m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b4024454596c4961a9d41da96893bf70 : SUCCESS in 3m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7b82a691405c4b9aaca7729269c5957b : SUCCESS in 11m 36s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a55bf61353b2401fb3393455bf6430b6 : SUCCESS in 4m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ee45b037fe184cdf8c0dc2193d6646af : SUCCESS in 6m 16s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/871b229dcd954434ae50358d77cfbf3a : SUCCESS in 10m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6738dd0a33274097ba5dcb2d52327f15 : SUCCESS in 40m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ac331dd7d3714a928e539eac2b40005f : SUCCESS in 1h 26m 50s","accounts_in_message":[],"_revision_number":6},{"id":"531a1fe1629546b38c1a502088ebeb1e2aaa08b6","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2020-08-28 10:57:32.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"58286fbf9be4684d9c9e5dee92e79673f60c1b45","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2020-08-28 10:57:59.000000000","message":"Patch Set 7:\n\nWho else should I potentially bug for a review on this?","accounts_in_message":[],"_revision_number":7},{"id":"aed6ed5660fe41c8bf57540d9aa1d71177dc6efa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 12:07:57.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e909ef79373f4c8a8e77e6cd9feecbb9 : SUCCESS in 1h 06m 20s","accounts_in_message":[],"_revision_number":7},{"id":"3de734acab019d0dc11abfb7c91bd58fb4f1d851","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-28 13:01:25.000000000","message":"Patch Set 7: Code-Review+2\n\nThanks for starting the conversation with the oslo team. It looks like there is at least no easy fix to get something like this into oslo.config, but at least the idea has gotten more visibility. So I think we\u0027re probably fine adding this. Would be great if some day we don\u0027t need multiple config libs, but there\u0027s no harm in it.\n\nWill leave for Matthew to approve.","accounts_in_message":[],"_revision_number":7},{"id":"645a577f447cfc272e6b27ddd3ee59915e7d1e10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 13:28:39.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27a9ee963704420db35b0eb07511119c : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/650e166239b441e897fcd0b25788b9c8 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d9f131f93bf4f478219c3b537817aa7 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26db7725cc184a4d847347bd2b31cc56 : SUCCESS in 4m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d7df4465dc5b496c9c104a3fd2da772b : SUCCESS in 5m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e4da2b53f3f74a53acaf8834c78bca5d : SUCCESS in 3m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0fa43851574f47718b54b569da086ae4 : SUCCESS in 8m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/324106035475497e9807780cee0ac2a4 : SUCCESS in 8m 32s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d670421c4d4248708c96fcf602dffd34 : SUCCESS in 11m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ed13d4ea70f1406293e03f6067cf94a2 : SUCCESS in 6m 32s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/96829664ccc34ac19e03dfade7918458 : SUCCESS in 10m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d3138c6d3cb2451a80944f99f16cf528 : SUCCESS in 12m 15s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/cce55fdf8aa14c848387f715105cc587 : SUCCESS in 8m 35s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/472ef3d44bb244ac80bd63fe3e7361db : SUCCESS in 15m 08s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/797d300748cc4dfc995c9a711e04a411 : SUCCESS in 31m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d89420aa5132434892fabb918cc529f6 : SUCCESS in 13m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b8d2a3bbbb9f489c8c775477640307fe : SUCCESS in 4m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9f4dc5eee9bf4517aa1ccea2ca69f7a7 : SUCCESS in 20m 10s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1913406ddaac47cbbfff20503c45d771 : SUCCESS in 6m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0a24cca633ea4ac498f6e52518cbb440 : SUCCESS in 10m 44s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ea3b4cd4263d4ee6981205d4d3684754 : SUCCESS in 15m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5dc050bc234c4c8e97b0a97b22a30c15 : SUCCESS in 41m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4149e8f6e37b4a79a4f6955015a76264 : SUCCESS in 1h 24m 39s","accounts_in_message":[],"_revision_number":7},{"id":"ef226fc74454b48cee9851ea3796a8d1a2b44e75","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-28 18:24:05.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"f0c88bdba10bc3d40492222ecd68835160f67c82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 21:08:45.000000000","message":"Patch Set 7: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":7},{"id":"a079c9dc6e6f451357dce95824567f792dd109f7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-28 21:22:13.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"7952cd459c95b828e7f8a571172305bf90fdb763","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-28 21:22:31.000000000","message":"Patch Set 8: Code-Review+2\n\nJust rebased due to conflict with the addition of httpx.","accounts_in_message":[],"_revision_number":8},{"id":"a70abbb3c5c359a7228c3493fbaf9eb5246dc818","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-28 21:24:31.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"68de9060d44c81f1c9e8e6a17c3fe2e6cff5d0b9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 21:32:31.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5da67841346e4bd1a56298116f7bb53e : RETRY_LIMIT in 0s","accounts_in_message":[],"_revision_number":8},{"id":"034b5136cb47f8fa9bd0e1a088a5d604779c7b5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 22:54:58.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/852e62a4f5c4491fa75af575a029620b : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4d3e21dc5294281aa4c35af7bf8c58a : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8fce34f8713a492196865254a4547208 : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ac7580df4c84e2da651055b48a88947 : SUCCESS in 5m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/89fa44a9b8944cb281ad279ab52d45b8 : SUCCESS in 4m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/299544501acd47ddb5d460496bb3b146 : SUCCESS in 4m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0a4e5327d0594463ad1f530ab910ab44 : SUCCESS in 6m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d03c66a5de9e483b83a6370e6ed30371 : SUCCESS in 10m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/982adf6dbfee42b5a15c9342433c314b : SUCCESS in 10m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/727734174a0341fc934938da6b361cea : SUCCESS in 5m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7f686c7bfa824596b0010fe08bd5eba4 : SUCCESS in 11m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/16639e0b0d9041dd8963d51db9d3cc9f : SUCCESS in 7m 46s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/238d74af71ac4e5eb94ae8a2e833fd95 : SUCCESS in 10m 58s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ddd38482b2c940d0a9ea630de899c57d : SUCCESS in 13m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/23c166b6669343bb8c4735f8d9a5b63c : SUCCESS in 17m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d1d71f7c3c374dddb664c8918c78e9a5 : SUCCESS in 10m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0d1b2c53db1240f385bd4cc13bd76ce4 : SUCCESS in 4m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/81a0d0ae863f4f56bcaa10a0237065e5 : SUCCESS in 16m 43s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/315a9e523ad54157903a7ce8b9570a70 : SUCCESS in 6m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c4ccd7e028fa471ba0dfd7c42d31b605 : SUCCESS in 7m 44s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/acb0e844e7ad48d0895342ecea508062 : SUCCESS in 11m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a3a3e0809b5241e5a1fcc5f33f1e208f : SUCCESS in 53m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/413b839590934ac9a73492c435adc48a : SUCCESS in 1h 16m 26s","accounts_in_message":[],"_revision_number":8},{"id":"22e5f2bade89e5288038f88ceb0940cc01464abe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 22:55:20.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"994b0086b37e8adf5fe612b672f0febfa27c6602","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-29 01:45:06.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/707109c608b447fd86b4b597dc3d6845 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc0289ded4d1417f85ee920cc227ebfe : SUCCESS in 3m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1aa9174663344a8cbbd67c9570999c1d : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ca447d05e8e4b0d89ec7607e639f3e8 : SUCCESS in 4m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5cd9e3556d764c55982f49da4456f9e8 : SUCCESS in 3m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c428cf1983414a1ba834686cfdb4fa0c : SUCCESS in 4m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/53e40b3cbec24c808c9f22dd75316f37 : SUCCESS in 6m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/713f9c08b6fc4967b0a88cfc10fad786 : SUCCESS in 11m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/41d836d628f24097b372bbf24fbd6a6e : SUCCESS in 7m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/78179415fa84465a9bcf2e4fb692eb3f : SUCCESS in 4m 42s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/02e6657eb542424689a7eaea6c2d4052 : SUCCESS in 9m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a65efdf83e6142e3bdd3e0e2d37123f5 : SUCCESS in 5m 54s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9d6de89e48934903ab85e3df7c96687d : SUCCESS in 6m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c5909f9842ba40ca82925ec062aee71c : SUCCESS in 13m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7ea5ba18180d401f89ff87c8e798c9d6 : SUCCESS in 24m 58s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/89770156eedd4e32aafb8ae3a1bd426e : SUCCESS in 12m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/83adc3dad7ff4b9aa0f43dc066a3788a : SUCCESS in 3m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f4d6ababb8184499baf0c16ec8e8a292 : SUCCESS in 16m 42s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ae48f0867c2e41c98571aa3d52fddd17 : SUCCESS in 4m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1e10fa4462534927ae216d12267a0d78 : SUCCESS in 6m 12s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/90ae19dbf42548a9b4919394816e6901 : SUCCESS in 9m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c181502f8baf42249499f060bd544a23 : SUCCESS in 9m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2c410670e05c42dcb91e28aead859d52 : SUCCESS in 50m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e816738d4fef4109ab7ed0cf02164f25 : SUCCESS in 1h 25m 07s","accounts_in_message":[],"_revision_number":8},{"id":"5388c6f0c2d00c2ccb7eebebd67869e72bf87cd2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-29 01:45:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"8b5e6bc601d4f3ab21d8c80f7ef7341fc682bf1a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-29 01:46:27.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b74fb30982294bb4844809d40f2fcef7 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"babfd104ad886d8180ff3c2f287e6ef414096d1c","revisions":{"dc5686dfde83792c0ea1dd38a38c96147e7834e1":{"kind":"REWORK","_number":1,"created":"2020-08-17 04:11:04.000000000","uploader":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"ref":"refs/changes/36/746436/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/36/746436/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/746436/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/746436/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/746436/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/36/746436/1"}}},"commit":{"parents":[{"commit":"68b891c85e247e09733b8577490d4484743814d5","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/68b891c85e247e09733b8577490d4484743814d5"}]}],"author":{"name":"Adrian Turjak","email":"adriant@catalystcloud.nz","date":"2020-08-17 03:44:04.000000000","tz":720},"committer":{"name":"Adrian Turjak","email":"adriant@catalystcloud.nz","date":"2020-08-17 03:44:04.000000000","tz":720},"subject":"Add CONFspirator to requirements","message":"Add CONFspirator to requirements\n\nCONFspirator is a config library inspired by oslo.config\nbuilt to solve somewhat different problems.\n\nIt is license compatible, py3 only, written by the Adjutant\nPTL, and will be maintained along with Adjutant, as well\nas hopefully as a separate library going forward.\n\nIt is already in Adjutant as of:\nhttps://opendev.org/openstack/adjutant/commit/c750fd6d6c0ab5390d66428069a4752dfaae1182\n\nChange-Id: Ibdb8785b3fa6234c8fadbeaa3a672eda373c7707\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dc5686dfde83792c0ea1dd38a38c96147e7834e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dc5686dfde83792c0ea1dd38a38c96147e7834e1"}]},"branch":"refs/heads/master"},"6c008e26306eda89d55ca81fd7a809afa6069d0a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-08-17 04:12:35.000000000","uploader":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"ref":"refs/changes/36/746436/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/36/746436/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/746436/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/746436/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/746436/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/36/746436/2"}}},"commit":{"parents":[{"commit":"68b891c85e247e09733b8577490d4484743814d5","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/68b891c85e247e09733b8577490d4484743814d5"}]}],"author":{"name":"Adrian Turjak","email":"adriant@catalystcloud.nz","date":"2020-08-17 03:44:04.000000000","tz":720},"committer":{"name":"Adrian Turjak","email":"adriant@catalystcloud.nz","date":"2020-08-17 04:11:48.000000000","tz":720},"subject":"Add CONFspirator to requirements","message":"Add CONFspirator to requirements\n\nCONFspirator is a config library inspired by oslo.config\nbuilt to solve somewhat different problems.\n\nIt is license compatible, py3 only, written by the Adjutant\nPTL, and will be maintained along with Adjutant, as well\nas hopefully as a separate library going forward.\n\nIt is already in Adjutant as of:\nhttps://opendev.org/openstack/adjutant/commit/c750fd6d6c0ab5390d66428069a4752dfaae1182\n\nDocs:\nhttps://confspirator.readthedocs.io/\nCode:\nhttps://gitlab.com/catalyst-cloud/confspirator\n\nChange-Id: Ibdb8785b3fa6234c8fadbeaa3a672eda373c7707\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6c008e26306eda89d55ca81fd7a809afa6069d0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6c008e26306eda89d55ca81fd7a809afa6069d0a"}]},"branch":"refs/heads/master"},"c89cd4a70d359a989f2784709816a8e8dfb4b809":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-08-17 04:19:12.000000000","uploader":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"ref":"refs/changes/36/746436/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/36/746436/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/746436/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/746436/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/746436/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/36/746436/3"}}},"commit":{"parents":[{"commit":"68b891c85e247e09733b8577490d4484743814d5","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/68b891c85e247e09733b8577490d4484743814d5"}]}],"author":{"name":"Adrian Turjak","email":"adriant@catalystcloud.nz","date":"2020-08-17 03:44:04.000000000","tz":720},"committer":{"name":"Adrian Turjak","email":"adriant@catalystcloud.nz","date":"2020-08-17 04:18:28.000000000","tz":720},"subject":"Add CONFspirator to requirements","message":"Add CONFspirator to requirements\n\nCONFspirator is a config library inspired by oslo.config\nbuilt to solve somewhat different problems.\n\nIt is already in Adjutant as of:\nhttps://opendev.org/openstack/adjutant/commit/c750fd6d6c0ab5390d66428069a4752dfaae1182\n\nDocs:\nhttps://confspirator.readthedocs.io/\nCode:\nhttps://gitlab.com/catalyst-cloud/confspirator\n\n[Q] Is the library actively maintained?\n[A] Yes, and will continue to be maintained along with\n    Adjutant given it was written for it.\n\n[Q] Is the library good code?\n[A] yes, code has both tests and documentation and was\n    written by myself (current PTL for Adjutant)\n\n[Q] Is the library python 3 compatible?\n[A] yes, and it does not support python 2.\n\n[Q] Is the library license compatible?\n[A] yes, Apache-2.0\n\n[Q] Is the library already packaged in the distros we target\n    (Ubuntu latest / Fedora latest)?\n[A] Not as an OS package, but available via pypi.\n\n[Q] Is the function of this library already covered by other\n    libraries in global-requirements.txt?\n[A] Partly, but not without major changes that are unlikely to\n    be welcomed.\n\n[Q] Is the library required for OpenStack project or related dev or\n    infrastructure setup? (Answer to this should be Yes, of course) Which?\n[A] OpenStack project: Adjutant\n\n[Q] If the library release is managed by the Openstack release process\n    does it use the cycle-with-intermediary release type?\n[A] N/A\n\n[Q] Do I need to update anything else?\n[A] no\n\nChange-Id: Ibdb8785b3fa6234c8fadbeaa3a672eda373c7707\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c89cd4a70d359a989f2784709816a8e8dfb4b809"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c89cd4a70d359a989f2784709816a8e8dfb4b809"}]},"branch":"refs/heads/master"},"833e139e5ad462cd4a555c6636c7326303a6d1f5":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-08-17 05:26:42.000000000","uploader":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"ref":"refs/changes/36/746436/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/36/746436/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/746436/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/746436/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/746436/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/36/746436/4"}}},"commit":{"parents":[{"commit":"b5f0397c1fa87a28ed1626630be247769e78d4c5","subject":"cap toml library because of nested dict bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b5f0397c1fa87a28ed1626630be247769e78d4c5"}]}],"author":{"name":"Adrian Turjak","email":"adriant@catalystcloud.nz","date":"2020-08-17 03:44:04.000000000","tz":720},"committer":{"name":"Adrian Turjak","email":"adriant@catalystcloud.nz","date":"2020-08-17 05:26:06.000000000","tz":720},"subject":"Add CONFspirator to requirements","message":"Add CONFspirator to requirements\n\nCONFspirator is a config library inspired by oslo.config\nbuilt to solve somewhat different problems.\n\nIt is already in Adjutant as of:\nhttps://opendev.org/openstack/adjutant/commit/c750fd6d6c0ab5390d66428069a4752dfaae1182\n\nDocs:\nhttps://confspirator.readthedocs.io/\nCode:\nhttps://gitlab.com/catalyst-cloud/confspirator\n\n[Q] Is the library actively maintained?\n[A] Yes, and will continue to be maintained along with\n    Adjutant given it was written for it.\n\n[Q] Is the library good code?\n[A] yes, code has both tests and documentation and was\n    written by myself (current PTL for Adjutant)\n\n[Q] Is the library python 3 compatible?\n[A] yes, and it does not support python 2.\n\n[Q] Is the library license compatible?\n[A] yes, Apache-2.0\n\n[Q] Is the library already packaged in the distros we target\n    (Ubuntu latest / Fedora latest)?\n[A] Not as an OS package, but available via pypi.\n\n[Q] Is the function of this library already covered by other\n    libraries in global-requirements.txt?\n[A] Partly, but not without major changes that are unlikely to\n    be welcomed.\n\n[Q] Is the library required for OpenStack project or related dev or\n    infrastructure setup? (Answer to this should be Yes, of course) Which?\n[A] OpenStack project: Adjutant\n\n[Q] If the library release is managed by the Openstack release process\n    does it use the cycle-with-intermediary release type?\n[A] N/A\n\n[Q] Do I need to update anything else?\n[A] no\n\nChange-Id: Ibdb8785b3fa6234c8fadbeaa3a672eda373c7707\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/833e139e5ad462cd4a555c6636c7326303a6d1f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/833e139e5ad462cd4a555c6636c7326303a6d1f5"}]},"branch":"refs/heads/master"},"a18e315b194e5a4e04b8f98b0beea5e153bc4a48":{"kind":"REWORK","_number":5,"created":"2020-08-17 06:08:00.000000000","uploader":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"ref":"refs/changes/36/746436/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/36/746436/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/746436/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/746436/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/746436/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/36/746436/5"}}},"commit":{"parents":[{"commit":"b5f0397c1fa87a28ed1626630be247769e78d4c5","subject":"cap toml library because of nested dict bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b5f0397c1fa87a28ed1626630be247769e78d4c5"}]}],"author":{"name":"Adrian Turjak","email":"adriant@catalystcloud.nz","date":"2020-08-17 03:44:04.000000000","tz":720},"committer":{"name":"Adrian Turjak","email":"adriant@catalystcloud.nz","date":"2020-08-17 06:07:46.000000000","tz":720},"subject":"Add CONFspirator to requirements","message":"Add CONFspirator to requirements\n\nCONFspirator is a config library inspired by oslo.config\nbuilt to solve somewhat different problems.\n\nIt is already in Adjutant as of:\nhttps://opendev.org/openstack/adjutant/commit/c750fd6d6c0ab5390d66428069a4752dfaae1182\n\nDocs:\nhttps://confspirator.readthedocs.io/\nCode:\nhttps://gitlab.com/catalyst-cloud/confspirator\n\n[Q] Is the library actively maintained?\n[A] Yes, and will continue to be maintained along with\n    Adjutant given it was written for it.\n\n[Q] Is the library good code?\n[A] yes, code has both tests and documentation and was\n    written by myself (current PTL for Adjutant)\n\n[Q] Is the library python 3 compatible?\n[A] yes, and it does not support python 2.\n\n[Q] Is the library license compatible?\n[A] yes, Apache-2.0\n\n[Q] Is the library already packaged in the distros we target\n    (Ubuntu latest / Fedora latest)?\n[A] Not as an OS package, but available via pypi.\n\n[Q] Is the function of this library already covered by other\n    libraries in global-requirements.txt?\n[A] Partly, but not without major changes that are unlikely to\n    be welcomed.\n\n[Q] Is the library required for OpenStack project or related dev or\n    infrastructure setup? (Answer to this should be Yes, of course) Which?\n[A] OpenStack project: Adjutant\n\n[Q] If the library release is managed by the Openstack release process\n    does it use the cycle-with-intermediary release type?\n[A] N/A\n\n[Q] Do I need to update anything else?\n[A] no\n\nChange-Id: Ibdb8785b3fa6234c8fadbeaa3a672eda373c7707\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a18e315b194e5a4e04b8f98b0beea5e153bc4a48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a18e315b194e5a4e04b8f98b0beea5e153bc4a48"}]},"branch":"refs/heads/master"},"45e954b3eecb869a8b58c9c5ae438bdc36dfc071":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-08-21 05:15:39.000000000","uploader":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"ref":"refs/changes/36/746436/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/36/746436/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/746436/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/746436/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/746436/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/36/746436/6"}}},"commit":{"parents":[{"commit":"293e6502fa960bc21db55d96cc9071a052007d1a","subject":"cap toml library because of nested dict bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/293e6502fa960bc21db55d96cc9071a052007d1a"}]}],"author":{"name":"Adrian Turjak","email":"adriant@catalystcloud.nz","date":"2020-08-17 03:44:04.000000000","tz":720},"committer":{"name":"Adrian Turjak","email":"adriant@catalystcloud.nz","date":"2020-08-21 05:14:54.000000000","tz":720},"subject":"Add CONFspirator to requirements","message":"Add CONFspirator to requirements\n\nCONFspirator is a config library inspired by oslo.config\nbuilt to solve somewhat different problems.\n\nIt is already in Adjutant as of:\nhttps://opendev.org/openstack/adjutant/commit/c750fd6d6c0ab5390d66428069a4752dfaae1182\n\nDocs:\nhttps://confspirator.readthedocs.io/\nCode:\nhttps://gitlab.com/catalyst-cloud/confspirator\n\n[Q] Is the library actively maintained?\n[A] Yes, and will continue to be maintained along with\n    Adjutant given it was written for it.\n\n[Q] Is the library good code?\n[A] yes, code has both tests and documentation and was\n    written by myself (current PTL for Adjutant)\n\n[Q] Is the library python 3 compatible?\n[A] yes, and it does not support python 2.\n\n[Q] Is the library license compatible?\n[A] yes, Apache-2.0\n\n[Q] Is the library already packaged in the distros we target\n    (Ubuntu latest / Fedora latest)?\n[A] Not as an OS package, but available via pypi.\n\n[Q] Is the function of this library already covered by other\n    libraries in global-requirements.txt?\n[A] Partly, but not without major changes that are unlikely to\n    be welcomed.\n\n[Q] Is the library required for OpenStack project or related dev or\n    infrastructure setup? (Answer to this should be Yes, of course) Which?\n[A] OpenStack project: Adjutant\n\n[Q] If the library release is managed by the Openstack release process\n    does it use the cycle-with-intermediary release type?\n[A] N/A\n\n[Q] Do I need to update anything else?\n[A] no\n\nChange-Id: Ibdb8785b3fa6234c8fadbeaa3a672eda373c7707\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/45e954b3eecb869a8b58c9c5ae438bdc36dfc071"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/45e954b3eecb869a8b58c9c5ae438bdc36dfc071"}]},"branch":"refs/heads/master"},"39f7927518c47fe0dd45d716e2f7c2cfbf461209":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-08-28 10:57:32.000000000","uploader":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"ref":"refs/changes/36/746436/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/36/746436/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/746436/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/746436/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/746436/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/36/746436/7"}}},"commit":{"parents":[{"commit":"74828735b009ec62fb717d20d36f2ce387e4bd1b","subject":"Merge \"gnocchi client 7.0.7 breaks gate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/74828735b009ec62fb717d20d36f2ce387e4bd1b"}]}],"author":{"name":"Adrian Turjak","email":"adriant@catalystcloud.nz","date":"2020-08-17 03:44:04.000000000","tz":720},"committer":{"name":"Adrian Turjak","email":"adriant@catalystcloud.nz","date":"2020-08-28 10:57:32.000000000","tz":0},"subject":"Add CONFspirator to requirements","message":"Add CONFspirator to requirements\n\nCONFspirator is a config library inspired by oslo.config\nbuilt to solve somewhat different problems.\n\nIt is already in Adjutant as of:\nhttps://opendev.org/openstack/adjutant/commit/c750fd6d6c0ab5390d66428069a4752dfaae1182\n\nDocs:\nhttps://confspirator.readthedocs.io/\nCode:\nhttps://gitlab.com/catalyst-cloud/confspirator\n\n[Q] Is the library actively maintained?\n[A] Yes, and will continue to be maintained along with\n    Adjutant given it was written for it.\n\n[Q] Is the library good code?\n[A] yes, code has both tests and documentation and was\n    written by myself (current PTL for Adjutant)\n\n[Q] Is the library python 3 compatible?\n[A] yes, and it does not support python 2.\n\n[Q] Is the library license compatible?\n[A] yes, Apache-2.0\n\n[Q] Is the library already packaged in the distros we target\n    (Ubuntu latest / Fedora latest)?\n[A] Not as an OS package, but available via pypi.\n\n[Q] Is the function of this library already covered by other\n    libraries in global-requirements.txt?\n[A] Partly, but not without major changes that are unlikely to\n    be welcomed.\n\n[Q] Is the library required for OpenStack project or related dev or\n    infrastructure setup? (Answer to this should be Yes, of course) Which?\n[A] OpenStack project: Adjutant\n\n[Q] If the library release is managed by the Openstack release process\n    does it use the cycle-with-intermediary release type?\n[A] N/A\n\n[Q] Do I need to update anything else?\n[A] no\n\nChange-Id: Ibdb8785b3fa6234c8fadbeaa3a672eda373c7707\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/39f7927518c47fe0dd45d716e2f7c2cfbf461209"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/39f7927518c47fe0dd45d716e2f7c2cfbf461209"}]},"branch":"refs/heads/master"},"babfd104ad886d8180ff3c2f287e6ef414096d1c":{"kind":"REWORK","_number":8,"created":"2020-08-28 21:22:13.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/36/746436/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/36/746436/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/746436/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/746436/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/746436/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/36/746436/8"}}},"commit":{"parents":[{"commit":"8a4f5a76863c1282f4fd1b9113254b1f518b7ac0","subject":"Merge \"update constraint for sushy to new release 3.3.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8a4f5a76863c1282f4fd1b9113254b1f518b7ac0"}]}],"author":{"name":"Adrian Turjak","email":"adriant@catalystcloud.nz","date":"2020-08-17 03:44:04.000000000","tz":720},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-08-28 21:22:04.000000000","tz":-300},"subject":"Add CONFspirator to requirements","message":"Add CONFspirator to requirements\n\nCONFspirator is a config library inspired by oslo.config\nbuilt to solve somewhat different problems.\n\nIt is already in Adjutant as of:\nhttps://opendev.org/openstack/adjutant/commit/c750fd6d6c0ab5390d66428069a4752dfaae1182\n\nDocs:\nhttps://confspirator.readthedocs.io/\nCode:\nhttps://gitlab.com/catalyst-cloud/confspirator\n\n[Q] Is the library actively maintained?\n[A] Yes, and will continue to be maintained along with\n    Adjutant given it was written for it.\n\n[Q] Is the library good code?\n[A] yes, code has both tests and documentation and was\n    written by myself (current PTL for Adjutant)\n\n[Q] Is the library python 3 compatible?\n[A] yes, and it does not support python 2.\n\n[Q] Is the library license compatible?\n[A] yes, Apache-2.0\n\n[Q] Is the library already packaged in the distros we target\n    (Ubuntu latest / Fedora latest)?\n[A] Not as an OS package, but available via pypi.\n\n[Q] Is the function of this library already covered by other\n    libraries in global-requirements.txt?\n[A] Partly, but not without major changes that are unlikely to\n    be welcomed.\n\n[Q] Is the library required for OpenStack project or related dev or\n    infrastructure setup? (Answer to this should be Yes, of course) Which?\n[A] OpenStack project: Adjutant\n\n[Q] If the library release is managed by the Openstack release process\n    does it use the cycle-with-intermediary release type?\n[A] N/A\n\n[Q] Do I need to update anything else?\n[A] no\n\nChange-Id: Ibdb8785b3fa6234c8fadbeaa3a672eda373c7707\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/babfd104ad886d8180ff3c2f287e6ef414096d1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/babfd104ad886d8180ff3c2f287e6ef414096d1c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
