)]}'
{"id":"openstack%2Frequirements~746440","triplet_id":"openstack%2Frequirements~master~I896267fe3b49327a1ad54237b92a5e78491f0092","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I896267fe3b49327a1ad54237b92a5e78491f0092","subject":"Updated from generate-constraints","status":"MERGED","created":"2020-08-17 06:22:22.000000000","updated":"2020-08-18 22:52:19.000000000","submitted":"2020-08-18 22:51:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746440-1597791068967-0b925b31","meta_rev_id":"690f30363ea460932199576a48fee2158861a73a","_number":746440,"virtual_id_number":746440,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-18 22:51:08.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-08-18 15:42:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-08-18 17:55:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-08-18 17:55:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-17 15:07:16.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-08-18 15:42:01.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-18 17:55:28.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-08-18 22:51:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"361794e0b06987dd745a17096e8bb6423bb68d31","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-08-17 06:22:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c92e7948a701fc4f7b91b1ee9c5cdefd6d65b8c6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 06:46:46.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/da6eab922deb485e98927db7f9bf4f28 : FAILURE in 17m 01s","accounts_in_message":[],"_revision_number":1},{"id":"e45066136d88d151de25494d94fb0ac725cf1f69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 08:19:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/062de5314eb244c1865bc3bbc535b73d : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb6294b3c7d34a8da55af675add29c2a : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e8e91c6d44442799f5dfd45e0e8ae6d : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ae20707c17f46cf904687b01c1c3946 : SUCCESS in 5m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0a03861423ab4490b1c764b525b0f1b9 : SUCCESS in 4m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d778a53484b1459989af2c351d2a0071 : SUCCESS in 4m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/86c71adab84940cda4b3818cff2e2dce : FAILURE in 6m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/68c22bb6060a4318ac28ddd146024a29 : FAILURE in 5m 01s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f562b75eb541494091e379ae9105b0e7 : SUCCESS in 8m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bfc2abb5759d4a5b9b572d3dc0e87c50 : SUCCESS in 5m 00s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/890f823966ae42f0b5b49fe9b96ffb7d : SUCCESS in 11m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/daf7879126d749a1a00cc9e602976151 : SUCCESS in 6m 59s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f4982e7250ef4381ae2f0e016e729212 : SUCCESS in 7m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/741ff8afdae84af98d9d09de704665cc : SUCCESS in 13m 29s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/dce2138babe044dbb3db109372563502 : SUCCESS in 28m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/613d4b4756244c449c0a7f67c8bc2a0b : FAILURE in 12m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/250594de431e4285b21870ad74d92341 : SUCCESS in 5m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/65e7ddada28b4c3b88b4c1834e44e676 : SUCCESS in 17m 01s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9d7a7e03fc734458b00700563c9b2d0c : SUCCESS in 5m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d56038b726724082a750114f418f5ca4 : SUCCESS in 8m 08s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/608d078541a5444bacc1e2c4d30d9c8d : SUCCESS in 15m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5040aa784ca74ca0a8efff457a611a6c : SUCCESS in 54m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/334f1cfb56d040429e701effb9c397cd : SUCCESS in 1h 13m 18s","accounts_in_message":[],"_revision_number":1},{"id":"8673b629d7ddcbc2fd83176b3610f0d82b437419","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-17 15:07:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"62ac8668c131208852ec31ff823a50a7825009de","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-17 15:29:21.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"cdd93d55a8abdb81a944ff908023b98492a0036c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 16:21:25.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7d06b445b3ca40b78e99e57e8557942b : FAILURE in 1h 07m 26s","accounts_in_message":[],"_revision_number":2},{"id":"d3170beff76eb78d7d64c17dab8a0c0947049157","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 19:44:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/169190cf735243ad99cfb143662c9963 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36a294f4c48448b08aae25c5b05fad60 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1466210cb4474daa82e753a2392ae198 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5dd90ee86ce944248909cca57def9e35 : SUCCESS in 5m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2ce391ba51de43b8bb57b1a4b90af2b4 : SUCCESS in 4m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/641725e443584e4e817b9e81ff60ccd3 : SUCCESS in 4m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/75455a30928c43a694450af32038e135 : FAILURE in 10m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e7de4bcecbc34052a3b7b6c9d0ce3665 : FAILURE in 7m 19s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/35abbc5871474a0096fb6a9a7ae72e53 : SUCCESS in 11m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/30cdf3aff84445b0848b2f67e921a99e : SUCCESS in 5m 31s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9c306dd88fef4c0dab3992360e3c9cbe : SUCCESS in 11m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/009e6b2fa1bc451a9a150e5a9992b5e8 : SUCCESS in 8m 15s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ab38654634554460aa2640ce8281fd46 : SUCCESS in 9m 13s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2a462942133141fa8b3cca93d87c0129 : SUCCESS in 12m 54s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b54ca15572124b0f9e65f7450ef0949b : SUCCESS in 24m 03s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4a04a1438ece47f69b6dc43ab81339c4 : FAILURE in 14m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/78c3a2ba02f0485db611449c79ff2d59 : SUCCESS in 5m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b6fb83b8216d4125b257cdd2e1704662 : SUCCESS in 16m 17s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5a99d38c01a14bba9408e35257b4c49b : SUCCESS in 5m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d5264f87420641a4b394228538873716 : SUCCESS in 8m 06s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/de9ecbf4139d4cb5941702210798b21e : SUCCESS in 9m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7cfd6ef9530a4744b8f2ce6662e77a5f : SUCCESS in 1h 12m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c0ff33efeee346fb875058e79aacb7e6 : SUCCESS in 1h 21m 16s","accounts_in_message":[],"_revision_number":2},{"id":"e835ee55284b813aabae205cc9b3c13d4d93fc93","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-08-18 06:28:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e6a1ddfb413d36d6ce090a24490bc70541d6d9c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 06:51:47.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b4805c5aa48b4553903a920b35c51e35 : FAILURE in 17m 33s","accounts_in_message":[],"_revision_number":3},{"id":"15a3ec6fd5a576c131801aca8664a0b9a0e5e8b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 08:36:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a26dc4276f33479db865018f8cd3da79 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56ab69cbda6348158188ad8bc11b6e7f : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a32c226d6bd948bc8f7fb687c33ef266 : SUCCESS in 7m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/988776d7e2244cf28c53e6926451dde9 : SUCCESS in 7m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bc9cd4adb6004925b6de0950470f187c : SUCCESS in 4m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3b9cfc22bca542889ac1ce985f1a27cc : SUCCESS in 5m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0acc5ca39ac74edf8ffb20bd89e9017a : FAILURE in 7m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/27ce7e37e6d742cdb00972f7b137d5dd : FAILURE in 6m 25s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b7f2a39587e34fc6a5a177ec38b1fd9c : SUCCESS in 9m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5d558e27a10443c2907fe5046cfbd14e : SUCCESS in 4m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e6cbfbabfd2040758724af194358f3ff : SUCCESS in 11m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a3eeeca92d61425ebac35b7af0b4cde0 : SUCCESS in 7m 40s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3031d156adb64eea8979c976a2dc7c66 : SUCCESS in 7m 36s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d6fab9c01c634c3a80d6edfac5f3e6a7 : SUCCESS in 12m 57s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4a8772fa488340d09119b7e0a0c8eace : SUCCESS in 24m 12s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b0b2cec7389c4c319dbcbbf73cab3bde : FAILURE in 13m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9e08b56295cc4896a03edc4b6cb80af1 : SUCCESS in 4m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6152d8cfc18c48748dc2ffa2282c27a8 : SUCCESS in 19m 44s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a53bf75f09fb4388b3304230499618ba : SUCCESS in 6m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8f27201535904cd9b78a57fd94b3cfe6 : SUCCESS in 7m 13s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4b873af9a4cf42fb85b6dddee3b9a285 : SUCCESS in 10m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/392f3de7b1ca4f04b716c7c23677fdd8 : POST_FAILURE in 17m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fbab1b9089c24974b6fe718d02aa4372 : SUCCESS in 1h 19m 27s","accounts_in_message":[],"_revision_number":3},{"id":"c492df279fc0f733312305c0530ff66fa8b6720d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-18 15:36:17.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c2b7c4a8cdd695262c536074c16ce2057511751c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-18 15:36:21.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":4},{"id":"828d4b6d370bc74bebc5e456d1264a1d2ef0c38d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-18 15:42:01.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"a66ba8a8b30a037cccf02a34662116e1a0bfe7f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 16:59:17.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d2304c2ecdf54ae0b309851dc9c05e8e : SUCCESS in 1h 09m 28s","accounts_in_message":[],"_revision_number":4},{"id":"257920bf1f59196d4864f16a2040fba9bba3efee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 17:43:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d9738d83bca4761941e633464f1e6b6 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62147d5974cc40369dcb254d15ab0d07 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a9f039916f54c068beb650cd10d1d94 : SUCCESS in 6m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18575b8aed634019971c93f81112cf25 : SUCCESS in 7m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/149f2e16416840b99d0d6d0ecb2228a8 : SUCCESS in 5m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/985737b1168a4e6faff61772ba956da2 : SUCCESS in 5m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1dcb1fc93977477c8fa27cf3dda7e274 : SUCCESS in 8m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fb534067b9cf4335bc1f20d7c2ed99d0 : SUCCESS in 9m 20s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/64d4d726ebb2461a961a611208411c0f : SUCCESS in 12m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/45523d91357344c5977732df5e0f0f2d : SUCCESS in 5m 20s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/dd24b960d1fb46c5b0aacde8c4b99f5f : SUCCESS in 10m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b4a764d77efe455ea677c05a5eae0cf6 : SUCCESS in 9m 03s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1e9c05d984ff47fcb0d136f648ff91dc : SUCCESS in 11m 31s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a543829538f94364b7e589e25bf419f7 : SUCCESS in 20m 36s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/38f6d880ce444f509a71f73d4bef87e9 : SUCCESS in 30m 18s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7819b739463d4befb0775f4c685ba937 : SUCCESS in 12m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b7fa0d850cf64aab9392f58163d1ade9 : SUCCESS in 3m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c7b6feb270b647989f99d57500af8a3b : SUCCESS in 16m 26s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5aa5125b4a594a0d86c79b271c59fcf1 : SUCCESS in 5m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e3e55018b4874c8088874acf00191783 : SUCCESS in 6m 29s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/037a6b7f40ac4a578668b05c0c9eca90 : SUCCESS in 12m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/74c1904c090941079509887e0222ca1a : SUCCESS in 54m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a7b093094e9140bc8747852cac18b154 : SUCCESS in 1h 19m 19s","accounts_in_message":[],"_revision_number":4},{"id":"57e224b9aa82732dfa225d37c9982f58a19f58ce","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-18 17:55:28.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"a61f0943cfa7a4d8c8d7194adb6a91087de47e94","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 17:55:50.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7dbc1b94ef988c14986c34fb4bd4ed8ca228388d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 19:46:02.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0048cd6da85a4816bc9b1f9c9993fe0c : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93c93c7383394dd6b2f1764f1bfe618b : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b8581165edf4dee8340f6897ebadda1 : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39d7a0413f5f426aac7ed2568ba9f98f : SUCCESS in 5m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4ae8e11efa9d4ee7ab3bcc458a5145e9 : SUCCESS in 5m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/06f7f0557c8845bd90045d4aef897c47 : SUCCESS in 4m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3281ee9455dc426a8d390619cdbbbdfc : FAILURE in 5m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9877b46856a745ce86dcb816a47c1797 : SUCCESS in 16m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ddc3f7daa93c48eaa3250bd7e9c06add : SUCCESS in 13m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/95586515086b432187054fbe895fe262 : SUCCESS in 6m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0bcd8892cc2448249561f137255ba240 : SUCCESS in 12m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a1cc4af47ee04af4850091378290e54e : SUCCESS in 8m 07s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/265b3e4c0b8c4ffa811135336a087b37 : SUCCESS in 7m 31s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/bd5f03f022584d72bc80b0fc8af5ff19 : SUCCESS in 21m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1f1903c13cd54d8586cd5793e01e4231 : SUCCESS in 26m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8a3e2af6193d47d186f9e0e300363a2c : SUCCESS in 12m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dd9fef6a7b1c4d888e0c7e6b5bb00f15 : SUCCESS in 3m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/123970e1324f4c8f9e12a63f1b721369 : SUCCESS in 16m 44s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3ee98fcba0094f5d93ad8b8bfff1fcad : SUCCESS in 5m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ff3360d6be774dfd87117773d0c65598 : SUCCESS in 8m 12s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/205084e0f22a45dcaa8e693710d0b9d3 : SUCCESS in 12m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3c7e2a8ce6d9405a94b20e572781ad5c : SUCCESS in 9m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa8317126ac441dd9bb5220a2f0600fa : SUCCESS in 1h 14m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9b2705a190bd418fbab6a33e0c5dd3f6 : SUCCESS in 1h 43m 14s","accounts_in_message":[],"_revision_number":4},{"id":"6bdd3f0d828b0f8582d5e53d586cd20e797eead0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-18 19:47:13.000000000","message":"Patch Set 4:\n\nrecheck\n\npypi caching issue with python-karborclient?","accounts_in_message":[],"_revision_number":4},{"id":"ef40cb18bd7e5ce91b446aa374fca6767ab157d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 21:00:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c917c6cb808b4011bf2c996d795879ad : SUCCESS in 1h 09m 08s","accounts_in_message":[],"_revision_number":4},{"id":"50dc1df27e4526966507f9cc93c361e2136a8638","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 21:11:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/950955a559614e03bd9a6d67f15b0f51 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c301d22f46740cbac09f4e391e92113 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98a1f7ee58ff4b70a740d773d8ba547f : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc44861d06df461cbd566ef41901bcca : SUCCESS in 4m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/578573ef7210449b9cecc4921a240197 : SUCCESS in 3m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/61023d0741184359af8e10caa119365e : SUCCESS in 3m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8f7912e2854b42d0987e16a2b9f11f23 : SUCCESS in 8m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b4f9111e179a482bae862f372257236c : SUCCESS in 8m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/451854629e2546bd89b3f3dffa85f35b : SUCCESS in 8m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/df416f0c6f7a4606b4eaebccf8b269b6 : SUCCESS in 5m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/455b7a009c514b62bc16be394b004830 : SUCCESS in 11m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ff46529b38da4892a58885b7a8967408 : SUCCESS in 6m 04s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a7272e4117254383b3f158e920157d8f : SUCCESS in 7m 49s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1ec54cc0b6004b08b0852f95d499e424 : SUCCESS in 13m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7ab5d3300cdf4e72ac5375e41bb5d4a8 : SUCCESS in 25m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3fd2bc3d019d49d19a59b6f7e12c2243 : SUCCESS in 14m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/28f04b1200314c40969313d5a737da7d : SUCCESS in 5m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5c7e97ba1caf458ca4ef9e07097b7a59 : SUCCESS in 17m 32s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f10e8a8efd80475180329ee36b4191ce : SUCCESS in 4m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f273b36650a84f0e860edec7b2d4d3be : SUCCESS in 7m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7bff0a1d04564461a54fb7d7df4f8b6e : SUCCESS in 12m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b1dfebaa2e74ef1b6561a78a988b16a : SUCCESS in 58m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1f5c7d175d5442f8b89549eea2cb4b73 : SUCCESS in 1h 17m 38s","accounts_in_message":[],"_revision_number":4},{"id":"c9afffb03f3c76ae72c8c287f7d287abdb52d147","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 21:11:41.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e56892ac2ad582506ba0450478789936081f6fd1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 22:51:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"ee2c36a0a2edd933f963942bd27faf280fe1bfb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 22:51:08.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b66f1834c53469491bd55b5f2003511 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d049b4b982c45b0bce959df1dfa4326 : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/432de3554fc34407910b3df1de522214 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b54a4de98e84317988d71764f07195a : SUCCESS in 5m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d2428e555d7a4318b92f2406e3758326 : SUCCESS in 3m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/76f2d8824ca04226a7ece39f71ff6dbe : SUCCESS in 3m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f4d40d342ed5484ca3100217714ba15d : SUCCESS in 9m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0e241f609026477fa0a3268ba516294e : SUCCESS in 11m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/877042ae4ac84bc9b99bd742491afae4 : SUCCESS in 11m 24s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f762aab31edb4d98ad052aeed87a8c72 : SUCCESS in 4m 57s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f18b8b3318c84519b384f49cdb842ca0 : SUCCESS in 10m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/308433371f2041d0a5d0d192729fa3aa : SUCCESS in 7m 14s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/346a91bfb8174593834e7cdada764a3a : SUCCESS in 9m 31s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ba6d4a5db5da4fd78b7943435d5b9462 : SUCCESS in 24m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a8c007af675c4cb3969d390c1592739c : SUCCESS in 24m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c2fe450425f742ceb909f943982135de : SUCCESS in 16m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/971e430cd7244d3fafba02e255210967 : SUCCESS in 3m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1c6f8cf87abc4f27bfac5cea5f8fd78b : SUCCESS in 20m 09s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e45dcbdcf9c34426bc5a801c5be7011e : SUCCESS in 5m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/aa4962b8d23343af8d3a597d640120d6 : SUCCESS in 7m 02s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/797fdfe356294a248fe9a4ae0be959ab : SUCCESS in 11m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/dd6a343a5e1944f6af6d1a0dc6c72347 : SUCCESS in 10m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed5510ff4fa44623947d21709f90e888 : SUCCESS in 57m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fe0a3fc2eede4c50b994ea997167cf8c : SUCCESS in 1h 11m 00s","accounts_in_message":[],"_revision_number":4},{"id":"690f30363ea460932199576a48fee2158861a73a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 22:52:19.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbc9c810e3914e9c8e2ca3d1489810d1 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ed90da38869752602477b2ec39191f80620ce4e0","revisions":{"5ba53e05bd7a3c031721136d5e27c3a997e2b2bc":{"kind":"REWORK","_number":1,"created":"2020-08-17 06:22:22.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/40/746440/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/746440/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/746440/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/746440/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/746440/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/746440/1"}}},"commit":{"parents":[{"commit":"1bd17ec2af4e76c8da98ab495ba515a1602c67a3","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1bd17ec2af4e76c8da98ab495ba515a1602c67a3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-17 06:22:21.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-17 06:22:21.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I896267fe3b49327a1ad54237b92a5e78491f0092\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5ba53e05bd7a3c031721136d5e27c3a997e2b2bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5ba53e05bd7a3c031721136d5e27c3a997e2b2bc"}]},"branch":"refs/heads/master"},"8a45acef808572e81c72c9f3dc2e26d8fefcefa8":{"kind":"REWORK","_number":2,"created":"2020-08-17 15:07:16.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/40/746440/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/746440/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/746440/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/746440/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/746440/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/746440/2"}}},"commit":{"parents":[{"commit":"1bd17ec2af4e76c8da98ab495ba515a1602c67a3","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1bd17ec2af4e76c8da98ab495ba515a1602c67a3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-17 06:22:21.000000000","tz":0},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-08-17 15:06:56.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I896267fe3b49327a1ad54237b92a5e78491f0092\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8a45acef808572e81c72c9f3dc2e26d8fefcefa8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8a45acef808572e81c72c9f3dc2e26d8fefcefa8"}]},"branch":"refs/heads/master"},"38eb18880c2784ca8191a624f28841697cbe47eb":{"kind":"REWORK","_number":3,"created":"2020-08-18 06:28:21.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/40/746440/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/746440/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/746440/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/746440/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/746440/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/746440/3"}}},"commit":{"parents":[{"commit":"1bd17ec2af4e76c8da98ab495ba515a1602c67a3","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1bd17ec2af4e76c8da98ab495ba515a1602c67a3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-18 06:28:19.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-18 06:28:19.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I896267fe3b49327a1ad54237b92a5e78491f0092\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/38eb18880c2784ca8191a624f28841697cbe47eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/38eb18880c2784ca8191a624f28841697cbe47eb"}]},"branch":"refs/heads/master"},"ed90da38869752602477b2ec39191f80620ce4e0":{"kind":"REWORK","_number":4,"created":"2020-08-18 15:36:17.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/40/746440/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/746440/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/746440/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/746440/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/746440/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/746440/4"}}},"commit":{"parents":[{"commit":"1bd17ec2af4e76c8da98ab495ba515a1602c67a3","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1bd17ec2af4e76c8da98ab495ba515a1602c67a3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-18 06:28:19.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-08-18 15:36:03.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I896267fe3b49327a1ad54237b92a5e78491f0092\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ed90da38869752602477b2ec39191f80620ce4e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ed90da38869752602477b2ec39191f80620ce4e0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
