)]}'
{"id":"openstack%2Frequirements~746484","triplet_id":"openstack%2Frequirements~stable%2Fqueens~I1959fd598eae7da1f9a531cc9b9c27f9558c8e0a","project":"openstack/requirements","branch":"stable/queens","hashtags":[],"change_id":"I1959fd598eae7da1f9a531cc9b9c27f9558c8e0a","subject":"Revert \"[stable-only] Upper constrain wheel to 0.34.2\"","status":"MERGED","created":"2020-08-17 11:31:25.000000000","updated":"2020-08-17 17:30:15.000000000","submitted":"2020-08-17 17:30:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746484-1597685415820-86100347","meta_rev_id":"9a122f6f78945358d4321cf24bd560510dbc890e","_number":746484,"virtual_id_number":746484,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2020-08-17 14:13:44.000000000","_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-17 17:30:15.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-08-17 11:39:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-08-17 16:26:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2020-08-17 16:02:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2020-08-17 16:02:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-17 11:39:56.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-17 16:02:22.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-08-17 16:26:55.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-08-17 17:30:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fd09b06946d34e519ad6c17e736a440f5b1ccec1","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-08-17 11:31:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a990b036c6ba46b44001d3d3a3db3b77ae23a28","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-17 11:39:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c64b24f544945de8c94455e18b392532a6d99410","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 12:51:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56c39d7d014345f88bce3c0ce5dec7b8 : SUCCESS in 3m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f0bcc4ed2f044a03826bbebc20ef5063 : SUCCESS in 4m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7ca705576a7f4e85bd2974990b0c0e37 : SUCCESS in 4m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0d040c407db3430989d4106821258f75 : SUCCESS in 4m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f5009157e51342aaa95ed96c5c733d5c : SUCCESS in 3m 38s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/49eeb66c06b9428bb68b4d7c5046682a : SUCCESS in 7m 26s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/8526442399c64152a1e8f44c3d0cb184 : SUCCESS in 7m 51s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/31c064230b284a678ceabc373a629326 : SUCCESS in 10m 46s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/e76691818d0640959d651d986e614776 : SUCCESS in 9m 43s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/8b0aaa51c71a40678e80b4596fb535b9 : SUCCESS in 5m 17s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/f6b51d49afb442deaf5bfb07bb16ade0 : SUCCESS in 5m 40s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/f28ec5f9da224cce896690e1fd04ec75 : SUCCESS in 11m 17s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/6987608cf19a4057b67c5a46be29a8d0 : SUCCESS in 10m 51s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/f4d35e31628e4ed2b7f851e11d4bc81c : SUCCESS in 13m 46s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/f404740e57994580941703b2f8fcfebc : TIMED_OUT in 40m 57s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/40832576303f4505806f29494c009d8e : SUCCESS in 25m 38s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/4de9edde031a43c0a97f7fd3fbb24a9e : SUCCESS in 14m 02s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/e9ac5e481c4640e28ca3f60b25f5e35e : SUCCESS in 9m 10s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/6a220f49e47448238534a6068f1faa9f : SUCCESS in 13m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fc55d1a53e3a43419bfd110d866595b3 : SUCCESS in 13m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/20b7228c7dd34036ba52548703556f84 : SUCCESS in 6m 33s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/347a2d858b9c4332b705f8ce626c597d : SUCCESS in 3m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fe97841ba0c14bddb935b72b3fbd4d2b : SUCCESS in 1h 14m 02s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/66eb570f20f549b9a5f7e3d7ac95ff7b : FAILURE in 44m 53s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ec54d1728f584bbab72fa22a8f478f2b : SUCCESS in 31m 57s","accounts_in_message":[],"_revision_number":1},{"id":"19c0e7f424aefbd6a141796dd28e6e98fd6a92e2","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-08-17 14:13:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a11c52d65508d1b1e2252d03b72a4c209bbcfb02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 15:59:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b074ce5d11f4b2f98096092cb294f8f : SUCCESS in 5m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/34fb2c88da944b1f91426b7fd6cdeecf : SUCCESS in 4m 44s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9a614f0056a6452c81d88b43eff4b1fa : SUCCESS in 7m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2ccf5b8d965842cc83e00633a791876a : SUCCESS in 5m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/57d2aa5bdc184dd28e6d01812074a007 : SUCCESS in 4m 53s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/176cc2a9c00c4345aee61c20714d9e76 : SUCCESS in 10m 29s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/79b79acb062644e79e8438195b3391ed : SUCCESS in 13m 10s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ddc42e61662b43d8a190bc9c766e932e : SUCCESS in 10m 49s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/c41b3def9c304849aa87855c8daa3277 : SUCCESS in 12m 34s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/4553787c9d9245b5b8ff1b08af375b1d : SUCCESS in 6m 13s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/e9c890fe2d8d4239acefa8c5acdf1364 : SUCCESS in 9m 33s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/2bd8002cdcd74ca8a847428c7ea5a2b8 : SUCCESS in 10m 56s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/e30a45ae72034a63a9fbd35c2d4716c1 : SUCCESS in 12m 42s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/959d1c7b379d4090835f15f52a692d50 : SUCCESS in 14m 12s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/c241f6312b774e29a4d8efb0d4904fe4 : SUCCESS in 13m 59s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/e59853ec3e9d49d9a09280fe670402a7 : SUCCESS in 22m 48s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/75ba1e74bda84cf09943832d459c9a3a : SUCCESS in 24m 56s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/03e22c03106e427798c70d3fd0fe6ac1 : SUCCESS in 11m 13s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/9ce6007d2a974aee880d3d6ba371fa37 : SUCCESS in 10m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5c29e31b70a44fb3b9a26b72b6aad8ac : SUCCESS in 22m 44s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8bc8eb1c141c427bb703a635ae77528d : SUCCESS in 12m 58s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/be7123ee582046539fc22e4fbda588c4 : SUCCESS in 5m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6d7217b1a5b04ed9a041a7b947371353 : SUCCESS in 1h 13m 48s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/d8ffd66e9cc4406990339252d0d835fb : SUCCESS in 1h 17m 55s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8238954771134a95a8fee31a840fb1ae : SUCCESS in 47m 36s","accounts_in_message":[],"_revision_number":1},{"id":"6466bedeb60fa331e3122aea4db133f005f6af7e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-17 16:02:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"58f8d7686499c2be691943b2ea60499478e11f6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 16:02:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"07e09963bd4efae2459401775fa0e654ccb736f2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-17 16:26:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1e33155ab4bb35c71d2cfbe15ba82a1e09e51f7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 17:30:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef7d44e9d3a94c72bb96d4e132862d0c : SUCCESS in 3m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0069202c1bc14d109c03bede8ad2c317 : SUCCESS in 4m 48s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b24f9877bd6b41a7ba8d1f9e61e36c7c : SUCCESS in 4m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e87805ef610a41ce9425e983734bf30e : SUCCESS in 4m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5c68ae2c02944e65b9536e08ec3e0912 : SUCCESS in 4m 40s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c0b0454a94d940a68ea2b608b148093c : SUCCESS in 6m 08s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/0da7ea8d4dc742a4a834771fa52bce03 : SUCCESS in 6m 36s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/703ba9505bf94a068f36fa5faa38d1c8 : SUCCESS in 11m 27s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/01704866762c4ed3a87597a33a2a4be6 : SUCCESS in 11m 04s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/dae87dbc31e2474d9f2ff0a2d0e2efd2 : SUCCESS in 6m 21s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/98920bfe92114225b2b3be4c9cb3afe6 : SUCCESS in 6m 41s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/92131b1233f34a72af00ab4a7a20c857 : SUCCESS in 9m 13s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/425407103abc433e80efbd1f364d4063 : SUCCESS in 9m 23s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/5284916cfd2f414a998d904b6fbff15c : SUCCESS in 30m 52s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/5a9e08fbfac54643ac7e79e2ae51445d : SUCCESS in 13m 34s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/78035745dc0f4a148b7c3d5fc439a701 : SUCCESS in 23m 20s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/25ea14a81c7b4cda91ba52b83dd54e45 : SUCCESS in 25m 02s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/9e871adb8b31454ab7ef6c3beb96c38e : SUCCESS in 15m 34s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/0f8914efa4b54c01a84a7f6a46abe4c6 : SUCCESS in 18m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9abd69aa5b224ecba96693dc928e7487 : SUCCESS in 21m 23s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/167c37e2e0de42db9b777f07ae849f95 : SUCCESS in 7m 35s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/743625d2f9f841788f2551d2a499475e : SUCCESS in 4m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a77b5211e6b3471bb9d5f81468540235 : SUCCESS in 57m 40s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/dd12571ed59047da922366b08be506b6 : SUCCESS in 1h 18m 46s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ce9a1212ee0942e8be7cb646fd7b11b9 : SUCCESS in 57m 55s","accounts_in_message":[],"_revision_number":1},{"id":"1b6096a32d95c3ce764339d474c587d9ff1aaecb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 17:30:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5cdb94c26bf26b085302517f409eac1d448f709f","revisions":{"5cdb94c26bf26b085302517f409eac1d448f709f":{"kind":"REWORK","_number":1,"created":"2020-08-17 11:31:25.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/84/746484/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/746484/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/746484/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/746484/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/746484/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/746484/1"}}},"commit":{"parents":[{"commit":"9bab19ff893be201b3124b797107ca3b948f0877","subject":"[stable-only] Upper constrain wheel to 0.34.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9bab19ff893be201b3124b797107ca3b948f0877"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-08-17 11:27:19.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-08-17 11:31:08.000000000","tz":120},"subject":"Revert \"[stable-only] Upper constrain wheel to 0.34.2\"","message":"Revert \"[stable-only] Upper constrain wheel to 0.34.2\"\n\nThis reverts commit 9bab19ff893be201b3124b797107ca3b948f0877.\n\nMeanwhile new release arrived from wheel (0.35.1), which makes the\nimport bug disappear. So this is not needed anymore in Queens (and\nolder branches).\n\nChange-Id: I1959fd598eae7da1f9a531cc9b9c27f9558c8e0a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5cdb94c26bf26b085302517f409eac1d448f709f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5cdb94c26bf26b085302517f409eac1d448f709f"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
