)]}'
{"id":"openstack%2Frequirements~747012","triplet_id":"openstack%2Frequirements~stable%2Fussuri~Ie30fe79e686f89fbdb246b5330d5b9f10aecf354","project":"openstack/requirements","branch":"stable/ussuri","hashtags":[],"change_id":"Ie30fe79e686f89fbdb246b5330d5b9f10aecf354","subject":"Add ansible to blacklist","status":"MERGED","created":"2020-08-19 18:31:41.000000000","updated":"2020-08-20 00:48:29.000000000","submitted":"2020-08-20 00:48:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747012-1597884509578-038e877c","meta_rev_id":"240492669a96715389ead0f99330e41545781d2a","_number":747012,"virtual_id_number":747012,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"date":"2020-08-19 18:42:50.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-20 00:48:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2020-08-19 18:35:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-08-19 18:41:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-08-19 20:00:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"date":"2020-08-19 18:31:41.000000000","_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-08-19 18:41:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-19 18:31:41.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2020-08-19 18:35:40.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-19 18:41:24.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-08-19 20:00:02.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-08-20 00:48:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b24c5b275bbf326a7d4ffcbc19f5394df53c5716","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-19 18:31:41.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8e2e5de3422e5193937c599a9ac4bdf1dbeaa4eb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-19 18:32:43.000000000","message":"Patch Set 1:\n\nit seems req-check reports only one problem at a time and I missed this other patch previously; please accept","accounts_in_message":[],"_revision_number":1},{"id":"c3a639ffb7d5a0ef5bb24a8bb0dea3ba407ffa3f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-19 18:35:40.000000000","message":"Patch Set 1: Code-Review+2\n\nDone on master in May.","accounts_in_message":[],"_revision_number":1},{"id":"a954b155b9b772553084fd174a8056e090f5d790","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-19 18:41:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"04ce891d10db01e89ccd29f16a3780783db02d70","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-19 18:42:50.000000000","message":"Patch Set 1:\n\nTYVM!","accounts_in_message":[],"_revision_number":1},{"id":"0c400671f00ff62b96509fac921fc57fab4fde98","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-19 20:00:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"19c2598783fae573d42cc3bbf5cc94970a338645","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 21:11:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/408d50ef28434129b5fffa2b6cf2c612 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b3461ad43c34544872e1404738c364b : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/acaa5de2425849bd9502d55449f591d0 : SUCCESS in 4m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8e363ae4bc9d449da9b47d23cb0e78bf : SUCCESS in 4m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a9e63769e66418fb2db14996a38914c : SUCCESS in 53m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/54711f64688e4dc4a887b2ea0240cece : SUCCESS in 1h 24m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2efdecabd947449b82c563463ad10063 : SUCCESS in 1h 24m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/da3fe02da7754598837c1cf35c95bb89 : SUCCESS in 1h 10m 03s","accounts_in_message":[],"_revision_number":1},{"id":"2aabb567a3a095019ec99345195f086d7c40fc06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 21:11:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"70ef6229e17204174a0c4130d092a52556d22566","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 00:48:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/665e6e3af0ae4080bbfceaacbb856c3e : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/842e8f2dfc2d4ea3a910556a6e6f3170 : SUCCESS in 4m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0a4396d2673d499eac50ee7bcd560519 : SUCCESS in 4m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4ef54b7338cb4deda48bee6826ab3cd9 : SUCCESS in 4m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2ea0e648b3794e5caab27ac1f8970ec1 : SUCCESS in 1h 09m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/eb67554c729648d297698b11c3b36ab7 : SUCCESS in 1h 17m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fb7e1c7a502543418d283bfb7299485d : SUCCESS in 1h 18m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/35ae231dab2c415e90489befa718250e : SUCCESS in 1h 25m 55s","accounts_in_message":[],"_revision_number":1},{"id":"240492669a96715389ead0f99330e41545781d2a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 00:48:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f047fe28b6eb5d5e4c982f41180b49e9948b5bf2","revisions":{"f047fe28b6eb5d5e4c982f41180b49e9948b5bf2":{"kind":"REWORK","_number":1,"created":"2020-08-19 18:31:41.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/12/747012/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/12/747012/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/12/747012/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/12/747012/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/12/747012/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/12/747012/1"}}},"commit":{"parents":[{"commit":"05f7a31434845ce8b33a7e44c0e2c6aea97dc150","subject":"Add yamllint to blacklist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/05f7a31434845ce8b33a7e44c0e2c6aea97dc150"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-05-11 09:45:09.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-08-19 18:31:41.000000000","tz":0},"subject":"Add ansible to blacklist","message":"Add ansible to blacklist\n\nSome projects, such as kolla-ansible, have a dependency on ansible. At\npresent it is not possible to add this to your requirements as it does\nnot appear in the global requirements file. However, it is useful to\nhave ansible as dependency to allow testing of plugins.\n\nThere is not a strong dependence on the openstack libraries, so it makes\nsense to include this in the blacklist and not in global requirements,\nso that the individual projects can specify the version required.\n\nThe reason for Ussuri backport: Kolla cut off the Ussuri branch\nalready in Victoria cycle, with this in place, and now it fails\non req-checks jobs in Ussuri.\n\nChange-Id: Ie30fe79e686f89fbdb246b5330d5b9f10aecf354\n(cherry picked from commit dcef447f6d5233ecc20ba19b7fa1e04e7dfb740c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f047fe28b6eb5d5e4c982f41180b49e9948b5bf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f047fe28b6eb5d5e4c982f41180b49e9948b5bf2"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
