)]}'
{"id":"openstack%2Frequirements~747419","triplet_id":"openstack%2Frequirements~master~Id5759fcc7e856c2acab8f0f4214017e7d344c0f1","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"5792":{"account":{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},"last_update":"2022-08-02 05:20:31.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id5759fcc7e856c2acab8f0f4214017e7d344c0f1","subject":"Revert \"Ban pecan 1.4.0\"","status":"ABANDONED","created":"2020-08-21 13:58:57.000000000","updated":"2022-08-02 05:20:31.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d3431061129b6d99bae20d78fa8fc3ea353e2834","_number":747419,"virtual_id_number":747419,"owner":{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-07-19 16:38:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"disliked":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"date":"2021-07-19 21:54:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":-2,"date":"2022-08-02 05:20:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2022-08-02 05:20:31.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-21 13:58:57.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-08-21 13:58:57.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2020-08-21 16:19:30.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-08-24 05:01:19.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"},{"updated":"2020-08-24 06:17:06.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-09-09 09:43:16.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-09-18 07:51:49.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-10-23 07:44:26.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-10-24 01:48:03.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-10-26 19:06:53.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-10-26 21:41:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-18 23:29:21.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"CC"}],"messages":[{"id":"35bdb50a0410e62aba6473a46741c692229f1fd8","author":{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},"date":"2020-08-21 13:58:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"715c8812263b88042ccfea7467a4643718f9be74","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 15:08:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/66fd4a3fd9ed47ef8ab0d306a293c915 : SUCCESS in 1h 05m 34s","accounts_in_message":[],"_revision_number":1},{"id":"77bb5e3337f7a03d9cd16860ad292d4f9c48829f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-21 16:18:40.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nHello","accounts_in_message":[],"_revision_number":1},{"id":"4f3917e3575d2a4b6287ca2495eba809b7a87fc0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-21 16:19:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"977c048870a76dc0aa61a169496d0685a60727ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 17:57:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/418d2132ad754ffbb970de239b8f5a01 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2b5ff3e163b4598b53d94461616e59a : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9adb7105cb04b9589ce4d7dc0f1c9d4 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5244c471ecdd45a689fafec0a25103cf : SUCCESS in 4m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/92c5f71ca005491abb9f8374b1fe53d7 : SUCCESS in 3m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/959ac0bcffe24fa48e03c00e0dcf791f : SUCCESS in 4m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/183188cd24104f31b118f0b53cf05bbc : SUCCESS in 11m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6c2c07d1e55342d9bb4f90b2c87eeeca : SUCCESS in 12m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/39e81fffdf5d42db9d649d5cfc5cb2a0 : SUCCESS in 10m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1daea98248c444ccae6792879a50403d : SUCCESS in 5m 15s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/709211f1c070412ebb59e7cac6cafdca : SUCCESS in 12m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/477f1339ef184e5fbb16d00d41d970d0 : SUCCESS in 8m 32s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/77710dfa4aae453cafd0e93a70a9d082 : SUCCESS in 9m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9e5e8452c5cc4e378aea9c0d0b447342 : SUCCESS in 13m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b63e42c549184d4dba92455d58cf8145 : SUCCESS in 29m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c4f76e51904446bcba687b29e24eb794 : SUCCESS in 11m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8782570fdf45429ab4f238edba7204b5 : SUCCESS in 4m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3a295c91f6114f4ea9f0cfd778bd0a1f : SUCCESS in 16m 42s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c736c2a8565d411f90720da44502266b : SUCCESS in 4m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f62d8bc254d64b9b9b33be25d178993c : SUCCESS in 8m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3ef42231f6f24caf9c43f97e6e6dd6cd : SUCCESS in 10m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a85bbe5731044610a56b4b53914b4fce : SUCCESS in 57m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/592fb12fcbbc444a9330277144298a3a : SUCCESS in 1h 17m 59s","accounts_in_message":[],"_revision_number":1},{"id":"215862bf3e0d924d1ea4f5cc6922aaaa45e38824","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-22 06:04:16.000000000","message":"Patch Set 1:\n\nshould we be banning the bad gnocciclient release then?","accounts_in_message":[],"_revision_number":1},{"id":"6754dc2065def47c5b28cfc020294a29981dafed","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-08-23 15:32:36.000000000","message":"Patch Set 1:\n\n\u003e should we be banning the bad gnocciclient release then?\n\nHi we in vitrage started to get gnocchiclient.exceptions.NotAcceptable: Not Acceptable (HTTP 406) since the new gnocchi client released\n\nI tried to depend on this patch but now I get more errors from aodh trying to parse response","accounts_in_message":[],"_revision_number":1},{"id":"f4eb47504b78b0bd456f03ff7bfeef50c76cfeb9","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-08-23 17:31:55.000000000","message":"Patch Set 1:\n\n\u003e \u003e should we be banning the bad gnocciclient release then?\n \u003e \n \u003e Hi we in vitrage started to get gnocchiclient.exceptions.NotAcceptable:\n \u003e Not Acceptable (HTTP 406) since the new gnocchi client released\n \u003e \n \u003e I tried to depend on this patch but now I get more errors from aodh\n \u003e trying to parse response\n\nneed to fix aodh client with the same fix of gnocchi client\nhttps://opendev.org/openstack/python-aodhclient/src/branch/master/aodhclient/v2/base.py#L21","accounts_in_message":[],"_revision_number":1},{"id":"03c9fc9c7d9bfb958d1c404ac3384084fcf134fd","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2020-08-24 04:36:51.000000000","message":"Patch Set 1:\n\n\u003e vitrage started to get gnocchiclient.exceptions.NotAcceptable: Not Acceptable (HTTP 406) since the new gnocchi client released\n\nYeah, DELETE API still seems borken and has to be fixed in gnocchi.","accounts_in_message":[],"_revision_number":1},{"id":"dfcb9487887a0332cc43f76c49bcb1653a0a7f19","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-24 04:49:32.000000000","message":"Patch Set 1:\n\nok, so should we hold off on this then?","accounts_in_message":[],"_revision_number":1},{"id":"b9593759c7d8f75a5fb44d89561ebf4d7adc744f","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2020-08-24 05:01:19.000000000","message":"Patch Set 1:\n\n\u003e ok, so should we hold off on this then?\n\nYes and revert the gnocchiclient uc bump unless we want all the project gates broken again.","accounts_in_message":[],"_revision_number":1},{"id":"60d28298b09400a7f4d329749f022a1657d37bce","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-08-24 06:17:06.000000000","message":"Patch Set 1: Code-Review-1\n\nthis breaks the gate","accounts_in_message":[],"_revision_number":1},{"id":"88d24c9125de8b08fe7e8322b9cd38242ffe32e0","author":{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},"date":"2020-08-24 13:04:25.000000000","message":"Patch Set 1:\n\nsubmitted a fix to aodhclient\nhttps://review.opendev.org/747722","accounts_in_message":[],"_revision_number":1},{"id":"d117c1b5093f40e502fd86052442a40f6c17e8f6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-09-05 05:04:51.000000000","message":"Patch Set 1:\n\nis this still needed?","accounts_in_message":[],"_revision_number":1},{"id":"2fc1280c86f0799765d798d73ccf0020bf1734bf","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-09 07:02:35.000000000","message":"Patch Set 1: Code-Review+1\n\nI opened a bug for neutron (https://launchpad.net/bugs/1894864 ) as some stadium gates are now broken due to pecan.\nNeutron needs pecan 1.4.0 as that has fix for the above bug\n\nI abandon my patch (https://review.opendev.org/750447 ) with the same goal.","accounts_in_message":[],"_revision_number":1},{"id":"76c2af418be38683b9c804f7b7369c4be6f43e44","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-09-09 09:43:16.000000000","message":"Patch Set 1:\n\nWith https://review.opendev.org/#/c/747722/ merged, aodh should be fine with new gnocchiclient? So in that case revert should also include this and then it should be OK\n(note: not tested personally, just interpretation of the relevant reviews)","accounts_in_message":[],"_revision_number":1},{"id":"3f150795d6640e42e7653524b0f82946f3602a5e","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-09-09 09:46:16.000000000","message":"Patch Set 1:\n\n\u003e With https://review.opendev.org/#/c/747722/ merged, aodh should be\n \u003e fine with new gnocchiclient? So in that case revert should also\n \u003e include this and then it should be OK\n \u003e (note: not tested personally, just interpretation of the relevant\n \u003e reviews)\n\nNo there are still issues in the DELETE REST API fro gnocchi and maybe aodh","accounts_in_message":[],"_revision_number":1},{"id":"1ae589266e85da5ac84da241a983957d7b8eb6a7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-09-11 14:30:20.000000000","message":"Patch Set 1: Workflow-1\n\nat this point we\u0027ll need an FFE request to get this in :(","accounts_in_message":[],"_revision_number":1},{"id":"16b0864f09f0ebd2cb7072aa382014e5cb0741e2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-17 16:12:38.000000000","message":"Patch Set 1:\n\nIs this needed for victoria yet? Time is running out, so if we need an FFE, that should get sorted quickly.","accounts_in_message":[],"_revision_number":1},{"id":"6ab980a319f28bc402cf1a6d78d02ada941a983a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-18 07:51:49.000000000","message":"Patch Set 1: -Code-Review\n\n\u003e Is this needed for victoria yet? Time is running out, so if we need\n \u003e an FFE, that should get sorted quickly.\n\nIn neutron, pecan bump was reverted (https://review.opendev.org/752058 ), and as I remember we agreed to check this (bump pecan to 1.4.0 or higher) again in Wallaby.\nThanks for attention.","accounts_in_message":[],"_revision_number":1},{"id":"6e36c3843342111f2c840e68ccca157483bf8e27","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-10-23 07:44:26.000000000","message":"Patch Set 1:\n\nhi reviewers, can we somehow move on with this now as we are in Wallaby cycle already?","accounts_in_message":[],"_revision_number":1},{"id":"2dd78bd1debc70d98ea561a532f30f7a1e84c61a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-24 01:48:03.000000000","message":"Patch Set 1: Code-Review+2 -Workflow\n\nremoving the feature freeze","accounts_in_message":[],"_revision_number":1},{"id":"677dcb4b85acdfd9536a227514869450cb9428c3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-26 19:06:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"639f82a221801f94f7293862fedd16f1ac2e6f88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 19:07:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8015b9ab799e74c8e72b92845272edf05f55818c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 21:41:25.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b726ad725cfa4ef48b1d316db9cac0cb : TIMED_OUT in 30m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f31891ac85f04f4c9f014ff301e37224 : SUCCESS in 30m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd22f71459c7473cbdf48bd8a0b789e7 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3c2ef8eb125485f8fe66ff47199758d : SUCCESS in 3m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c0ceb9990b184bfe9b82e663e49a0d70 : SUCCESS in 3m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7034ed218f7b47e2bc7a4d2477dbc015 : SUCCESS in 3m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fec5070d5d1f4c89baff893a7b2e4233 : TIMED_OUT in 40m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c8140717ed414d78b43b530da7bb4dd7 : SUCCESS in 8m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4bf4b82e547f40cc96ab1d78f378ae42 : SUCCESS in 9m 19s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f9eec261a68f4de09d35e05655ddcaaa : SUCCESS in 6m 19s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/48fbd7081f75467d9f37966c86edb57d : SUCCESS in 10m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ce249d8899d14c208c2b9b957e909346 : SUCCESS in 7m 28s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/84b686a161a345b3b55d2aac2bee15c2 : SUCCESS in 7m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/13506d5d655a4e528757af61e8b992e3 : SUCCESS in 14m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/02cfe60d41f84d478d0175f2e4710d74 : SUCCESS in 25m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/36d71fc05c564090a0692a3d0bb4b387 : SUCCESS in 12m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/62aa02f7efa44ab6a2df3645559f73c1 : SUCCESS in 3m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/53b0db0a36e6470d83b478f5429c3e97 : SUCCESS in 16m 49s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1d2fffbf51484377979382554acb95fa : SUCCESS in 4m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1b047dfe64c142b78ce4b074fe243bd2 : SUCCESS in 6m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8afe890ae3534419bf647548a45a46a6 : SUCCESS in 10m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9dfa61a7f5f74e06980d93f0bd7f3a5c : SUCCESS in 55m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a5686b52ce074611a2e5a1d3cc3de985 : TIMED_OUT in 2h 03m 46s","accounts_in_message":[],"_revision_number":1},{"id":"d968e89596ca77f9710c3ff3c1ce965eca9d15fe","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-10-27 09:19:28.000000000","message":"Patch Set 1:\n\nWas the DELETE rest api in gnocchiclient and aodhclient fixed ? \nif not please don\u0027t merge this since it will break the GATE","accounts_in_message":[],"_revision_number":1},{"id":"5ddc188cd83c8edf45e41ce717dd658a6df0318d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-27 11:40:39.000000000","message":"Patch Set 1: -Workflow\n\nThis should get updated with depends-on if there are things we need to wait on that are not covered by the current tests.\n\nOr we process this, and anything like that will just be forced to deal with it. We are at the start of the cycle, so that\u0027s not the worst approach.","accounts_in_message":[],"_revision_number":1},{"id":"69f55ce279027e126db45490aac6ac1c37850a8f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-07 01:18:13.000000000","message":"Patch Set 1:\n\nping?  I\u0027d like to get this in soon if possible","accounts_in_message":[],"_revision_number":1},{"id":"91a194fbbaa886fff77013b82f4fa61b34a993c4","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-03-10 11:11:13.000000000","message":"Patch Set 1:\n\nI don\u0027t think there were any updates on gnocchiclient/aodhclient in Wallaby cycle :/","accounts_in_message":[],"_revision_number":1},{"id":"e8f1cc74f9f91073a981674c891ff00a41bb6043","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-13 01:39:03.000000000","message":"Patch Set 1: Code-Review-2\n\nDNM for freeze period","accounts_in_message":[],"_revision_number":1},{"id":"f51ba8df238b01c75a9ba56fccae792229b3d357","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-07-19 14:33:36.000000000","message":"Patch Set 1:\n\nCan we try this one again for Xena?","accounts_in_message":[],"_revision_number":1},{"id":"964e8a9e51ee03180c6b78492d85d2daae5df491","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},"date":"2021-07-19 15:18:06.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"894a38f65147bb4bc199d2c61d4bd2763339e96b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 15:32:27.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/9deae134237549cbb7702bf6cd664d29 : FAILURE in 10m 59s","accounts_in_message":[],"_revision_number":2},{"id":"9a69760b0e628fce298e7e58d51ce1f72de8f071","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2021-07-19 15:34:24.000000000","message":"Patch Set 2:\n\nI dont think gnocchiclient and aodhclient were fixed","accounts_in_message":[],"_revision_number":2},{"id":"cb9cf1aaee9d317f0ebd98ab92a89bc1836896d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 16:38:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65b78f0f435b42e7ada8e2f2cd495185 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa69e43a39754706a812c086df6ed8aa : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d6b85e725af45f381c54d247b0bfcb4 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7f23e8f591145cbae72bee51270b016 : SUCCESS in 7m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd7f79b65bf14482b92cf0f572c6bcf3 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/135e1749d165469aa190def35d41bca3 : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/eba9f6d4e3c14f6e8d48dc0ca5f323ea : SUCCESS in 4m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/54f18a36ce3440248a2333122ec8a79e : SUCCESS in 9m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e63660362c8d4391a87c261d8621faaf : SUCCESS in 9m 09s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/b250229c5bf74ccdae4584c829895ea4 : SUCCESS in 5m 53s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/69237a16eb0c4b0a8ec04f32c934c2b6 : SUCCESS in 11m 41s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e56a8bf5481e4c2a8b3f93c55c4f84f2 : SUCCESS in 5m 35s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/a7a04a3e28144633a97d61b72ba6e843 : SUCCESS in 11m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ed606611676d44eba1902b438f57174b : SUCCESS in 12m 27s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d278f7057c23488e866f8458eca84bd9 : SUCCESS in 8m 29s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/45598e68dafc4c00a51bde7ce34e52f3 : SUCCESS in 14m 18s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/8a48802b960a4ef6bc09c47edd1f1740 : SUCCESS in 10m 10s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/0f6757b60b4242d1a00b2f5055bc5992 : SUCCESS in 5m 42s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e1bf781507aa413296380fe15b93427c : SUCCESS in 23m 18s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/1a81fe795da242fbacee18264fb24fde : SUCCESS in 16m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2685d3ad13d2469a8f0a123e6e24e640 : SUCCESS in 5m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f3d87ae74bd148cc84385ab57aa2bf8b : SUCCESS in 18m 30s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/6b1fdf3ff42c4f5cbe2123e9416b1ad5 : SUCCESS in 6m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ecdba9d9bd094f9eb6e0aec6b2d69a05 : SUCCESS in 7m 09s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e14697161f924c3ab80dc06b6b100710 : SUCCESS in 12m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8299288ab2ff478fafe9bae12b00b144 : SUCCESS in 42m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dad35a189b8040acad15552b0010d103 : SUCCESS in 1h 16m 11s","accounts_in_message":[],"_revision_number":2},{"id":"dcbbe44ec1898c0384300ee204203931a7559f15","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-07-19 21:54:41.000000000","message":"Patch Set 2: -Code-Review\n\n\u003e Patch Set 2:\n\u003e \n\u003e I dont think gnocchiclient and aodhclient were fixed\n\nI don\u0027t think we have any good visibility on this, so I am very reluctant to approve anything. Is there some way we can track necessary work so we know when this can actually be approved?","accounts_in_message":[],"_revision_number":2},{"id":"cb384bab75924c33cfc3f383381a2ab541e4916d","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-07-18 23:29:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f26004a08d561bb2d77e4b51eda4cc78b801c0fe","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-08-02 05:20:31.000000000","message":"Abandoned\n\nAs per: https://review.opendev.org/c/openstack/requirements/+/747419/2#message-cb384bab75924c33cfc3f383381a2ab541e4916d\nI\u0027m abandoning this.  It it is still needed we can restore it.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"05af4367c05f6b25b30dddfc460740114ca44c08","revisions":{"e1f41534f5553d74b79c87bbb0b21eb016a457b4":{"kind":"REWORK","_number":1,"created":"2020-08-21 13:58:57.000000000","uploader":{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},"ref":"refs/changes/19/747419/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/19/747419/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/747419/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/747419/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/747419/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/19/747419/1"}}},"commit":{"parents":[{"commit":"570307ea8f569fa73445b745325e9245dd7b42b0","subject":"Ban pecan 1.4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/570307ea8f569fa73445b745325e9245dd7b42b0"}]}],"author":{"name":"Dan Radez","email":"dradez@redhat.com","date":"2020-08-21 13:58:57.000000000","tz":0},"committer":{"name":"Dan Radez","email":"dradez@redhat.com","date":"2020-08-21 13:58:57.000000000","tz":0},"subject":"Revert \"Ban pecan 1.4.0\"","message":"Revert \"Ban pecan 1.4.0\"\n\nThis reverts commit 570307ea8f569fa73445b745325e9245dd7b42b0.\n\nChange-Id: Id5759fcc7e856c2acab8f0f4214017e7d344c0f1\nDepends-On: Ida2cdc401def97bd5be10018e309e0ce72beb972\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e1f41534f5553d74b79c87bbb0b21eb016a457b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e1f41534f5553d74b79c87bbb0b21eb016a457b4"}]},"branch":"refs/heads/master"},"05af4367c05f6b25b30dddfc460740114ca44c08":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-07-19 15:18:06.000000000","uploader":{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},"ref":"refs/changes/19/747419/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/19/747419/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/747419/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/747419/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/747419/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/19/747419/2"}}},"commit":{"parents":[{"commit":"f398f5874bf895e7aca4759bedfb214947ec8a90","subject":"Merge \"update constraint for metalsmith to new release 1.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f398f5874bf895e7aca4759bedfb214947ec8a90"}]}],"author":{"name":"Dan Radez","email":"dradez@redhat.com","date":"2020-08-21 13:58:57.000000000","tz":0},"committer":{"name":"Dan Radez","email":"dradez@redhat.com","date":"2021-07-19 15:18:06.000000000","tz":0},"subject":"Revert \"Ban pecan 1.4.0\"","message":"Revert \"Ban pecan 1.4.0\"\n\nThis reverts commit 570307ea8f569fa73445b745325e9245dd7b42b0.\n\nChange-Id: Id5759fcc7e856c2acab8f0f4214017e7d344c0f1\nDepends-On: Ida2cdc401def97bd5be10018e309e0ce72beb972\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/05af4367c05f6b25b30dddfc460740114ca44c08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/05af4367c05f6b25b30dddfc460740114ca44c08"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
