)]}'
{"id":"openstack%2Frequirements~747565","triplet_id":"openstack%2Frequirements~master~I1664b27e9ad612cea55426859f00a84541e7393d","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I1664b27e9ad612cea55426859f00a84541e7393d","subject":"Updated from generate-constraints","status":"MERGED","created":"2020-08-23 06:16:06.000000000","updated":"2020-08-23 20:48:26.000000000","submitted":"2020-08-23 20:47:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747565-1598215640948-01c56cfd","meta_rev_id":"15fea0c1e8b497eebf274d09677f70f8e94efb90","_number":747565,"virtual_id_number":747565,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-23 20:47:20.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-08-23 06:20:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-08-23 06:20:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-23 06:20:36.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-08-23 20:47:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f2b93e9f15765b6f767b85fd73775cffa4dcbfb3","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-08-23 06:16:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f2dbe4eb9cc829e890051f8af15e4594b68ad64","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-23 06:20:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e5054204d1a2c169ad40d519a28f696e8b0014af","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-23 06:20:22.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"243776749fb1252aa8ebfef40cec4b1fd9d52dd0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-23 06:20:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cd3146b7f34bdeabaa713ab330f71e2600c309e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-23 07:39:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4d47e0a0587243cead6a03a627ba1174 : SUCCESS in 1h 12m 29s","accounts_in_message":[],"_revision_number":2},{"id":"1ab443bbf1be2b80b89a4f858f8f7a3059ba5d6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-23 07:52:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d30c1b8015ea4e679a560ad4d6d3432c : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d40f2a104d904235a8d767bde12baea4 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5720f300cc7422c944ee0f9f06b4441 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/171c5830795d4b5ab7cbae85731fab02 : SUCCESS in 6m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3ddf1813bb8e4bc59df6d87819a5edd9 : SUCCESS in 4m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6c89ace694db443ea3dbc05f404a30dd : SUCCESS in 6m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/aeee98c1a69e4fd381adc9e1d8365d3a : SUCCESS in 10m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/69d6afb61dbc44a4a1e6065a009a5b43 : SUCCESS in 12m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/bf3405af907b4f0ea8ea94a0038c6960 : SUCCESS in 9m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a693cc93124a43e89c1123f0c0281d23 : SUCCESS in 6m 08s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8cf02e5ab3a64a429ca2f08274ec98f4 : SUCCESS in 12m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3b883bcab796486a974b8c5c22ddc700 : SUCCESS in 9m 37s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ac831770a2cd4020923d08e0d8c27793 : SUCCESS in 11m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e31874f3c8554a5985ab988c43a27539 : SUCCESS in 21m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a3903dcd839b44129fcaa07bfea803c6 : SUCCESS in 24m 59s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c3af2849ad3249fba3137dd0812478ec : SUCCESS in 16m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5d028020665c4d00a5789fd6f47c85ff : SUCCESS in 6m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/04ee68176e584287bf9219c2ccb657ea : SUCCESS in 19m 30s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b52a68b2e3c440ea85fc13f5aec343ae : SUCCESS in 7m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f809e1f66d3a4e40aca064c0fbcea245 : SUCCESS in 8m 38s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4ae4c54d0c714b7d86f8947f2d7881e5 : SUCCESS in 12m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8f400f21f5bc48479219fcdad8cd4629 : SUCCESS in 53m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4fee128cfc9d4abb9edd0272129bf229 : SUCCESS in 1h 23m 23s","accounts_in_message":[],"_revision_number":2},{"id":"aafdafbce1166dc629450a9baf713b10b0ef7707","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-23 07:52:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ccfa25779bbf43bbf8d423928a4af6ac933299af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-23 09:02:19.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46c084cd99324bd2877851ba4211f30c : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fd6af7290014a168297bb0b1a32563b : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b986d40b7cae4c098a674f9f38f6e31a : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d296d0c713842a3a8006a6189dc3aea : SUCCESS in 6m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8910718fe23c4f728d447ded8217a202 : SUCCESS in 5m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/db259179eb614871b7f12c8978812e04 : SUCCESS in 5m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9d383e6cbab04df79728ccf11dbab3b0 : SUCCESS in 8m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2c821f2e9d034105992395eea0696a7d : SUCCESS in 9m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1e6d72d59b28459b8c3fa9a9dabb5da8 : SUCCESS in 6m 18s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b1df1aa64ad040e4ac548f4a2e1d6634 : SUCCESS in 5m 10s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/027661349e9b41feaf3c50f88c0bf1df : SUCCESS in 10m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/18faf85c298644fd93cb28e477ba3963 : SUCCESS in 8m 26s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/67a0753fd0dc4eaaadf1d46dee6fc8d1 : SUCCESS in 6m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ebc73d32d279418d969016ae2be19d32 : SUCCESS in 19m 22s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5dddc30377e24080ad5380e3175594d5 : SUCCESS in 22m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d5c77e072d9f44ffa4ecbfd1da1e5f05 : SUCCESS in 14m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1632bb21276a47149479c5778ba838e1 : SUCCESS in 4m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/168c67dbd8a44c0d8fc7adf466557a66 : SUCCESS in 20m 06s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/17c5da5f9fc24cf38c491b7e8445a7d2 : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2a1982fe39fa48249d423949bfb8de2b : SUCCESS in 7m 15s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/103c425ea25a44169e85323e5518c5e3 : SUCCESS in 10m 08s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a380bda739d1445cb612e5d7606e30b9 : SUCCESS in 16m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d0ccc9628fe14ab68875fb8c3f919c1c : SUCCESS in 49m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/49e0c3a552ad4c1b8d2e3db3adcfc4b6 : FAILURE in 1h 08m 06s","accounts_in_message":[],"_revision_number":2},{"id":"c67fe135a71e34060d3224381261fbf3a2fda1a4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-23 18:46:53.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e6f33d089af42a08a08b0a009dedc73ef7e4bad8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-23 19:42:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/642877aba525425fb366dac231620cce : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/326e1d99108446949e82d1f4f953522a : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/234db070f1db4ab7a28380bde3d5d66c : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef88b1f171b94c398c7c748e376eae45 : SUCCESS in 3m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f2c8822aa988469888a9dc31242df03d : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6b2c9f50e33844ee88964610354730ef : SUCCESS in 3m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6ea0a5449fca4082bee4bb8d92b9238b : SUCCESS in 6m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5d5c89c66d8b41d1bb02b356fdf22371 : SUCCESS in 8m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6b21f4c128bf4872911539392bc9e7d2 : SUCCESS in 8m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/63b5fda4a8ea42339ce5913657426f39 : SUCCESS in 4m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/06a4da91be4c4c9b885382858f79fbbf : SUCCESS in 6m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/74a2787cf76b46c8994c91734914a684 : SUCCESS in 5m 26s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/af3fbe719ed34feba4d62bfb2d708dbc : SUCCESS in 5m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2071ca1b77d84a5784282464c0fe0c5e : SUCCESS in 10m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2dea08661f8b4a189393d418a079adbf : SUCCESS in 15m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fc9c730f2f7f454399d3bf6b5c486f80 : SUCCESS in 8m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/43426fe03d0b46b5924c7aae328befa4 : SUCCESS in 3m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fe40d196f29f499ab25902bc497b7b8e : SUCCESS in 11m 39s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ff77658d8360460cb34f80563532dfac : SUCCESS in 3m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f835ec8382f0431da4ff99248eadc0e9 : SUCCESS in 5m 44s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9e2bd4fbb0fe479d89b7b6325ef76b55 : SUCCESS in 7m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d25fe09388354fafa69f7eccc7ebe253 : SUCCESS in 38m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8997fe0766bd4020b52007df584fbcca : SUCCESS in 54m 28s","accounts_in_message":[],"_revision_number":2},{"id":"d956d566076b784372b7491d0bc637c5ab8f39de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-23 19:43:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f78f64a55a3484a1b41698000357937d283373bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-23 19:51:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/36d81d4cd5aa4332be637e3a1e254bcf : SUCCESS in 1h 03m 54s","accounts_in_message":[],"_revision_number":2},{"id":"65bd1d0ed3a01b78cabcce76b9d67a6aa7383beb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-23 20:47:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"e2fb65c8d6e15e340b8b808fdfd18f96be86e2de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-23 20:47:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67da95add3634fc08f1a2eb9b3718ff3 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc1069d2388b4f47ab7584b30c2c7c93 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe8b722851be4359ab4ac85ceeb74b5b : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c13f9b1c61784c64b7e2139b46962d62 : SUCCESS in 3m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3229f7c4d8e942099281ffe814a6f28b : SUCCESS in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6ae4d6215c554b389028c0c7bb86fa8a : SUCCESS in 3m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2910eca97bc14a8080191e66346cd38e : SUCCESS in 5m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/853e3c4bf20a4b0e954543c3832a8a23 : SUCCESS in 8m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/61a885de6cc64ba89f7bc33518d53f03 : SUCCESS in 8m 00s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/33f7daf9971d41a38c20fe3400effe36 : SUCCESS in 3m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/04f8b2ad8d334018ae844b43bacd15cc : SUCCESS in 9m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/02c840b837544600ad2fce3ef5b89421 : SUCCESS in 7m 18s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4ebf9dbf8ed24a7093418853ad209638 : SUCCESS in 7m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e37de79f8904441898c9e968c7e00aef : SUCCESS in 10m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7b3b20cedb5e42e6a7ff73a1eca5d5c1 : SUCCESS in 25m 49s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/bdde3def4479486cb89103a7ec81325b : SUCCESS in 11m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3c91114693c7424ca9ffdd848646aeaa : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/444a7f7ace26460ca7b5ca2f3ae94d6d : SUCCESS in 11m 56s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a031ffcd2aab4d1e98de014efe4fbc51 : SUCCESS in 3m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bda1d83f3a784873bd8a51c857eaccd5 : SUCCESS in 6m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/58462e37267d4012bb8aa7f64d47125a : SUCCESS in 16m 56s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/743531936c58438fb01df73547ddfc57 : SUCCESS in 7m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/06d7808779a3404393947006d774c75c : SUCCESS in 48m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/841d8f6c010c4b3ea1ee038a1f38bd37 : SUCCESS in 52m 39s","accounts_in_message":[],"_revision_number":2},{"id":"15fea0c1e8b497eebf274d09677f70f8e94efb90","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-23 20:48:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/414f616aa11447c39d7c9f3b8563f543 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5f1f308941397fa110663e66829fe00f99b2066f","revisions":{"76e8558c05baf802eb135a95a4e5156447158ed9":{"kind":"REWORK","_number":1,"created":"2020-08-23 06:16:06.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/65/747565/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/747565/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/747565/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/747565/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/747565/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/747565/1"}}},"commit":{"parents":[{"commit":"efa0f25c0f2e44ecd696b3418102ba404169506b","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/efa0f25c0f2e44ecd696b3418102ba404169506b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-23 06:16:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-23 06:16:04.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1664b27e9ad612cea55426859f00a84541e7393d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/76e8558c05baf802eb135a95a4e5156447158ed9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/76e8558c05baf802eb135a95a4e5156447158ed9"}]},"branch":"refs/heads/master"},"5f1f308941397fa110663e66829fe00f99b2066f":{"kind":"REWORK","_number":2,"created":"2020-08-23 06:20:13.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/65/747565/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/747565/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/747565/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/747565/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/747565/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/747565/2"}}},"commit":{"parents":[{"commit":"efa0f25c0f2e44ecd696b3418102ba404169506b","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/efa0f25c0f2e44ecd696b3418102ba404169506b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-23 06:16:04.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-08-23 06:19:57.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1664b27e9ad612cea55426859f00a84541e7393d\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5f1f308941397fa110663e66829fe00f99b2066f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5f1f308941397fa110663e66829fe00f99b2066f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
