)]}'
{"id":"openstack%2Frequirements~747669","triplet_id":"openstack%2Frequirements~stable%2Fussuri~Iec6b639fce37135f8840240b5d2a1a168bf9bb4e","project":"openstack/requirements","branch":"stable/ussuri","hashtags":[],"change_id":"Iec6b639fce37135f8840240b5d2a1a168bf9bb4e","subject":"Add validations-libs to the global-requirements","status":"MERGED","created":"2020-08-24 09:40:21.000000000","updated":"2020-08-25 21:10:15.000000000","submitted":"2020-08-25 21:10:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747669-1598389815395-8a9bf4a0","meta_rev_id":"0dec080620df061e34c35c170e7dd28ab2c51bb2","_number":747669,"virtual_id_number":747669,"owner":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2020-08-25 19:48:26.000000000","_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-25 21:10:15.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-08-25 19:53:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":2,"date":"2020-08-24 09:42:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-08-25 19:53:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-24 09:42:14.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-08-25 19:53:29.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-08-25 21:10:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b846455b1cec707b6ef078245575765f89492207","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-08-24 09:40:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d1c793bf418507cebb0611a603c3b97b8a98d325","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-24 09:42:14.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM waiting for CI","accounts_in_message":[],"_revision_number":1},{"id":"de7d63965a2a4a32a1f6f2d9dd74f43fce1f8552","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 12:52:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4815e0cfd1e4169a7c659b3d1c3c54e : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ea9a9123d1b4b3ca7be618f95cf3e3f : SUCCESS in 6m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/73bf6797a8f9402c8b3026f330e8619a : SUCCESS in 5m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dcd7fb349ff14917adeb5a2c19513f51 : SUCCESS in 5m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5668b8e3115f42ecb4dfe00db1bfdde9 : SUCCESS in 4m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/53174602720d4fb9912a50dddf040623 : SUCCESS in 10m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a803acba8c6d42df9b84af1fb30664e5 : SUCCESS in 10m 43s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/57a2e712c7d140488d18709c38a9760e : SUCCESS in 6m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c65ce2a247464f0ab6a9f74375c5b756 : SUCCESS in 9m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ee66c088dec44740ab83eb462a0b0c8f : SUCCESS in 7m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0ceb8dbf44a74ac1a554e95eed12e8a0 : SUCCESS in 18m 33s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/43ea84c5d2ef4b19ab8636a39aad2f43 : SUCCESS in 25m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8652d3c892e14a1aaa9c3e35c3a7e57c : SUCCESS in 13m 31s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/5c911623003544e8a3d369c049db509c : SUCCESS in 17m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dfd3f36fe0c34244863ca7642ea1526e : SUCCESS in 17m 07s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f80c9faa87e04e42b88d682a5fe2cf95 : SUCCESS in 4m 50s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/98b4557f8d71440a9b1866c6bdf123de : SUCCESS in 5m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b331e989d7ee43afb58cdaaa9dd25b92 : SUCCESS in 6m 34s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/54c8925b8ced4da2a418b8347ebd5711 : SUCCESS in 9m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/223f1e655bcb4c268e302168b793b063 : SUCCESS in 9m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/40f61e39c1d341f5b2769cf03b8df047 : SUCCESS in 50m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/41967189702d4f2f919d8c3f3ca058e1 : SUCCESS in 1h 22m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e6e0eba498874b3eae47ec2dd7d7c4f8 : SUCCESS in 1h 18m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/095d1d28070a45679f35c97410374b1a : SUCCESS in 1h 10m 10s","accounts_in_message":[],"_revision_number":1},{"id":"9907cb6e415e43e408522e3064cf896862693e01","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-25 16:58:05.000000000","message":"Patch Set 1:\n\nI\u0027m fine with this, but it doesn\u0027t look like this was backported, is there a reason why it was not backported from master?\n\nhttps://github.com/openstack/requirements/commit/5368475655b129ed71a6c207220570a36a93b3de","accounts_in_message":[],"_revision_number":1},{"id":"15af1d39ed0953afe293b1e0cf0fe3a292ae416e","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-08-25 19:48:26.000000000","message":"Patch Set 1:\n\nIt\u0027s a cherry pick, but I had to resolve manually some conflicts.","accounts_in_message":[],"_revision_number":1},{"id":"9c0f12e080e016016eda78664fb5bd03d465e532","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-25 19:53:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nfair enough","accounts_in_message":[],"_revision_number":1},{"id":"cbcb831ec46e7e1d544534a1226c3d49bc03d239","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 19:53:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"10fdeeeff02e02e615c9a1a4bb46e3404d77980b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 21:10:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0dec080620df061e34c35c170e7dd28ab2c51bb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 21:10:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/267fb763fd3d4f47a36a2c49d73dbc30 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72f9d9d5419c4e60b0dfa63290b053e9 : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0782d99043734af48cad5fe436e10c1e : SUCCESS in 5m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/721f42e9daf3424cba5f98858c8db26d : SUCCESS in 4m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4161952847254eedbb3737b71d705b45 : SUCCESS in 6m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ff69733dcd994006aaf1cef81cd94de1 : SUCCESS in 13m 25s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a931a10562bb49799b74e1acea80d078 : SUCCESS in 11m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a018e62aa6de489295fb032e86c25d90 : SUCCESS in 5m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/020ba801bb4d40aab000642f13d79bb3 : SUCCESS in 12m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e5b2f567c3014eff83eeee63b2d4f6ed : SUCCESS in 7m 05s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f103354a52964fd9ae9e812ea328eff7 : SUCCESS in 23m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d89d8dcbd178440aaf57ebd9fae3869c : SUCCESS in 28m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2951b7d6f997487b9f89a41a608e32d3 : SUCCESS in 14m 05s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/073420bc6bdb41349754020800ee557a : SUCCESS in 13m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/75ec2033037f4dd698103e70f1cb0a9e : SUCCESS in 12m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/efa90e12e58142a2b3478d4ff21e1d4a : SUCCESS in 6m 32s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/b0fbb40bba2b4d029015099dfef7cf2b : SUCCESS in 5m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ea21d6c044f244dc9271a3a7fd114dd9 : SUCCESS in 8m 33s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b57779bf0cdd4f42aeb6e4ce2995643b : SUCCESS in 10m 15s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5844cb103a1048d7ad9c34f889506d41 : SUCCESS in 11m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/35389ce9478942f0a4ebbfd4efcd8269 : SUCCESS in 48m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/888d588a8989400aa3ead6e719033a09 : SUCCESS in 1h 05m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a0aa1cefefc7413fa98aa131704cde07 : SUCCESS in 1h 14m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/f5fe7485d63f4ecc9b4e120cf44add8d : SUCCESS in 1h 10m 27s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4021cd02ca2b11a3177080b03639a76f77370ade","revisions":{"4021cd02ca2b11a3177080b03639a76f77370ade":{"kind":"REWORK","_number":1,"created":"2020-08-24 09:40:21.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/69/747669/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/69/747669/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/747669/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/747669/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/747669/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/69/747669/1"}}},"commit":{"parents":[{"commit":"f047fe28b6eb5d5e4c982f41180b49e9948b5bf2","subject":"Add ansible to blacklist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f047fe28b6eb5d5e4c982f41180b49e9948b5bf2"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-05-05 07:16:47.000000000","tz":120},"committer":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-08-24 09:40:18.000000000","tz":120},"subject":"Add validations-libs to the global-requirements","message":"Add validations-libs to the global-requirements\n\ntripleoclient needs the validations-libs as dependencie, we need to\nadd the validations-libs to the global-requirements:\nhttps://review.opendev.org/#/c/714629/\n\nAnswers to the basic requirements questions.\n\nIs the library actively maintained?\n\nYes, the libraries is actively maintained and mainly used by TripleO projects\n\nIs the library good code?\n\nYes the libraries is covered by CI and Unit tests\n\nIs the library python 3 compatible?\n\nYes\n\nIs the library license compatible?\n\nYes\n\nIs the library already packaged in the distros we target (Ubuntu latest / Fedora latest)?\n\nYes\n\nIs the function of this library already covered by other libraries in global-requirements.txt?\n\nNo\n\nIs the library required for OpenStack project or related dev or infrastructure\nsetup? (Answer to this should be Yes, of course) Which?\n\nYes, it\u0027s required for python-tripleoclient and TripleO projects\n\nIf the library release is managed by the Openstack release process does it use\nthe cycle-with-intermediary release type?\n\nYes\n\nDo I need to update anything else?\n\nNo.\n\nChange-Id: Iec6b639fce37135f8840240b5d2a1a168bf9bb4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4021cd02ca2b11a3177080b03639a76f77370ade"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4021cd02ca2b11a3177080b03639a76f77370ade"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
