)]}'
{"id":"openstack%2Frequirements~747671","triplet_id":"openstack%2Frequirements~stable%2Ftrain~Idcf64b4c83f9004bb326f66c8c0da80bb8817a2c","project":"openstack/requirements","branch":"stable/train","topic":"validations/libs-stable/train","hashtags":[],"change_id":"Idcf64b4c83f9004bb326f66c8c0da80bb8817a2c","subject":"Add validations-libs and validations-common to openstack_requirements","status":"MERGED","created":"2020-08-24 09:57:07.000000000","updated":"2020-08-27 01:37:56.000000000","submitted":"2020-08-27 01:37:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747671-1598492276763-f63fd487","meta_rev_id":"4ff919771f7d766650ba5bf92b645ce3dea21578","_number":747671,"virtual_id_number":747671,"owner":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2020-08-26 20:33:08.000000000","_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-27 01:37:56.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-08-25 22:43:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":2,"date":"2020-08-26 09:44:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-08-25 22:43:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-25 22:43:18.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-08-26 09:44:19.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-08-27 01:37:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4ff693a64e412f4bcadcd76e491830b17582fa49","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-08-24 09:57:07.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e43fe018263b187d9ad094418a09fbbaca6245d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 13:40:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5af71d4380d4f4c8341da34788fd4da : SUCCESS in 4m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5c30d499f037423280b309fcf581d4fa : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba31b26a9e9043e9b91e1ee20c9cf9c7 : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e8f18de6d452412b819d829ba2f72513 : SUCCESS in 5m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/574a0da66fc4443081fc008a3cc4bd06 : SUCCESS in 5m 13s\n- requirements-tox-validate-projects https://zuul.opendev.org/t/openstack/build/4e73df82db364daaa087f8341ada215b : FAILURE in 4m 16s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/67dd3ec67a274858ac3d11359084967a : FAILURE in 1h 00m 59s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ee7193e12232452591f11d2eeff96a20 : SUCCESS in 1h 37m 49s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6121bf5cbd71457884b8ee38716b5f68 : SUCCESS in 1h 09m 46s","accounts_in_message":[],"_revision_number":1},{"id":"6cf44f538095d72d1b0b5947ac90965ac21b2542","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-08-25 20:42:47.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7c81558fae45548518a86a4d6c58d44c8c10b894","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 21:50:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdf910b9e146436096382f947f74c7c8 : SUCCESS in 4m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e9fa182b44ed4f1a81df9a5af683b7fd : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52119c2c1a364c2d8f98c2f37a25a4bc : SUCCESS in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a5007eca3b2e43fe913e4ce796e4b7f7 : SUCCESS in 4m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/75a7b2cb9ade40f4a41bd01d31d6df84 : SUCCESS in 4m 25s\n- requirements-tox-validate-projects https://zuul.opendev.org/t/openstack/build/346d9874da50498a9a11fe79935d4918 : FAILURE in 3m 43s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6e50efc2563a4e64837da36261d1e59b : SUCCESS in 46m 56s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6526cc5e464444afbd44ab147017a720 : SUCCESS in 1h 05m 38s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/60a94f2c62a04c02ab51debd499026fd : SUCCESS in 54m 06s","accounts_in_message":[],"_revision_number":2},{"id":"c173121056d2fd44d0caab15754e20b0c95835de","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-25 22:43:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d1fe40d8995319a716e89c9eb7481c3eb5933c2b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-26 09:44:19.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bb6e32db096f14726a35f401fd95402e2c59771c","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-08-26 20:33:08.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"80f4892322c61a8ec5aee492683531f6485ee252","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 00:00:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0efb3eb750340aebef1843dad3825a1 : SUCCESS in 4m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/387461939d484af9955adbc609f17313 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f27981ab8694b29aad024282c2f2d85 : SUCCESS in 5m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7476b07264054fc79ba11bc1ee0935b0 : SUCCESS in 6m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6bece0bb9dab42329e8a64a5951eb9d0 : SUCCESS in 4m 05s\n- requirements-tox-validate-projects https://zuul.opendev.org/t/openstack/build/903f6e0736474cb8a9a48095e943c3b3 : FAILURE in 4m 12s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b47ded557e264a6481c653c1882b9ac8 : SUCCESS in 1h 11m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c2cf46b6775d4e36abc9a6711127db18 : SUCCESS in 1h 27m 57s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0fd50f19a2cf463e8e1db948bc952fa3 : SUCCESS in 59m 11s","accounts_in_message":[],"_revision_number":2},{"id":"1a943aa1bd238f5079e7c0bf12e5772e952400ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 00:00:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"06a6b0c419b2ecde24aceec3df164c574a842fba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 01:37:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"e4eb1987541746ea12d34f10c7e615852efe1436","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 01:37:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37b690df120941f1a8c5e2c6c9f8482a : SUCCESS in 3m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/00060e91735143f99ad540d41f074bbd : SUCCESS in 3m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6023490cbcda4a3abcbd8a147f200fbd : SUCCESS in 3m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/84ba164f9ffa4656a9d0d42fe8f61a33 : SUCCESS in 4m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e5a77be5fd484420bc573c4a82a571e5 : SUCCESS in 4m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c1b6bc289d484fecb8a5194eefb8c488 : SUCCESS in 1h 09m 29s\n- tempest-full https://zuul.opendev.org/t/openstack/build/821bec67d75e46b1b6a8d62936973dfe : SUCCESS in 1h 36m 28s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4da928310df5415aa4202322c99c3571 : SUCCESS in 1h 01m 28s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"446451757005c65ce970c03fb73f6d3b80e1fc02","revisions":{"406fe49f764f1694b434b74ceb9705b617dd55a6":{"kind":"REWORK","_number":1,"created":"2020-08-24 09:57:07.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/71/747671/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/71/747671/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/747671/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/747671/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/747671/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/71/747671/1"}}},"commit":{"parents":[{"commit":"17f212594d39d26fc3c4827055f9500edd111f14","subject":"Bump OSC Train version to 4.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/17f212594d39d26fc3c4827055f9500edd111f14"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-04-22 21:00:57.000000000","tz":120},"committer":{"name":"mathieu bultel","email":"mat.bultel@gmail.com","date":"2020-08-24 09:57:07.000000000","tz":0},"subject":"Add validations-libs and validations-common to openstack_requirements","message":"Add validations-libs and validations-common to openstack_requirements\n\nvalidations-libs and validations-common are needed to the\nopenstack_requirements\n\nAnswers to the basic requirements questions.\n\nIs the library actively maintained?\n\nYes, the libraries is actively maintained and mainly used by TripleO projects\n\nIs the library good code?\n\nYes the libraries is covered by CI and Unit tests\n\nIs the library python 3 compatible?\n\nYes\n\nIs the library license compatible?\n\nYes\n\nIs the library already packaged in the distros we target (Ubuntu latest / Fedora latest)?\n\nYes\n\nIs the function of this library already covered by other libraries in global-requirements.txt?\n\nNo\n\nIs the library required for OpenStack project or related dev or infrastructure\nsetup? (Answer to this should be Yes, of course) Which?\n\nYes, it\u0027s required for python-tripleoclient and TripleO projects\n\nIf the library release is managed by the Openstack release process does it use\nthe cycle-with-intermediary release type?\n\nYes\n\nDo I need to update anything else?\n\nNo.\n\nChange-Id: Idcf64b4c83f9004bb326f66c8c0da80bb8817a2c\n(cherry picked from commit d14c326fbce9cec18990b25405490a24a0200c50)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/406fe49f764f1694b434b74ceb9705b617dd55a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/406fe49f764f1694b434b74ceb9705b617dd55a6"}]},"branch":"refs/heads/stable/train"},"446451757005c65ce970c03fb73f6d3b80e1fc02":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-08-25 20:42:47.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/71/747671/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/71/747671/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/747671/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/747671/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/747671/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/71/747671/2"}}},"commit":{"parents":[{"commit":"17f212594d39d26fc3c4827055f9500edd111f14","subject":"Bump OSC Train version to 4.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/17f212594d39d26fc3c4827055f9500edd111f14"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-04-22 21:00:57.000000000","tz":120},"committer":{"name":"mathieu bultel","email":"mat.bultel@gmail.com","date":"2020-08-25 20:42:44.000000000","tz":0},"subject":"Add validations-libs and validations-common to openstack_requirements","message":"Add validations-libs and validations-common to openstack_requirements\n\nvalidations-libs and validations-common are needed to the\nopenstack_requirements\n\nAnswers to the basic requirements questions.\n\nIs the library actively maintained?\n\nYes, the libraries is actively maintained and mainly used by TripleO projects\n\nIs the library good code?\n\nYes the libraries is covered by CI and Unit tests\n\nIs the library python 3 compatible?\n\nYes\n\nIs the library license compatible?\n\nYes\n\nIs the library already packaged in the distros we target (Ubuntu latest / Fedora latest)?\n\nYes\n\nIs the function of this library already covered by other libraries in global-requirements.txt?\n\nNo\n\nIs the library required for OpenStack project or related dev or infrastructure\nsetup? (Answer to this should be Yes, of course) Which?\n\nYes, it\u0027s required for python-tripleoclient and TripleO projects\n\nIf the library release is managed by the Openstack release process does it use\nthe cycle-with-intermediary release type?\n\nYes\n\nDo I need to update anything else?\n\nNo.\n\nDepends-On: https://review.opendev.org/#/c/748023\nChange-Id: Idcf64b4c83f9004bb326f66c8c0da80bb8817a2c\n(cherry picked from commit d14c326fbce9cec18990b25405490a24a0200c50)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/446451757005c65ce970c03fb73f6d3b80e1fc02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/446451757005c65ce970c03fb73f6d3b80e1fc02"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
