)]}'
{"id":"openstack%2Frequirements~747932","triplet_id":"openstack%2Frequirements~master~Id832a0cd67eebc030407cde972360a1412a6a018","project":"openstack/requirements","branch":"master","topic":"http2","hashtags":[],"change_id":"Id832a0cd67eebc030407cde972360a1412a6a018","subject":"Add httpx to global-reqs and upper-constraints","status":"MERGED","created":"2020-08-25 13:23:21.000000000","updated":"2020-08-28 14:10:00.000000000","submitted":"2020-08-28 14:06:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747932-1598623582202-98515791","meta_rev_id":"d16b7402d765122c9964db294b8c938392042b01","_number":747932,"virtual_id_number":747932,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2020-08-27 13:02:24.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-28 14:06:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-08-26 23:14:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-08-27 08:12:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-08-27 08:12:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-26 23:14:45.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-08-27 08:12:51.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-08-28 14:06:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1a52ae07a8dc4c4f966d52f71082ac904e69ce2e","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-25 13:23:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"93ad970e33edaf4c3f24bd92b7df47efcd130d49","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 14:33:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/25d020a030d8425aa2e73bc7c116d141 : SUCCESS in 1h 05m 40s","accounts_in_message":[],"_revision_number":1},{"id":"0d8c0cdb9139f582d4ee6f5c2defc1642145e14b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 17:23:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/407f6790c0c04cd4893bbc8e03166a93 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c840ec592f1649eaba9b7e3e90de8a26 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20de0a45e84b41f1ae13530d6063fa90 : SUCCESS in 7m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3465bb503c084b4f8e129398d375d974 : SUCCESS in 3m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/21d5b55b8b594d3cb34e67e40d138f76 : FAILURE in 7m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3e7f67c4af614287818309b866054a9a : SUCCESS in 8m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9bfea27e72cb4543824c5dd9310ab1bc : SUCCESS in 10m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4f2b492e17aa4a6587ffafdb506df2ec : SUCCESS in 12m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8bfa842c8f694917b1fefe1cff65b514 : SUCCESS in 10m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1f2092f355df493fa404f080a112623a : SUCCESS in 6m 23s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c5b4277b78cd46b49152d41cf9c08c99 : SUCCESS in 11m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7c19e36fba0c4dccb537eab54c984b05 : SUCCESS in 8m 14s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/445114f21f4a4177a1f4e0a899d89ba6 : SUCCESS in 9m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/02d4d48e079749e88f186a5f0e98fdf1 : SUCCESS in 14m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ca420019564546ac8a7ed1285b167553 : SUCCESS in 32m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ae076ed903f541dbbdf736e4c78c4236 : SUCCESS in 13m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0c5bdceb31274d76be9d8de58bcf76e5 : SUCCESS in 8m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4eaa84368de3436f85bca495ec9012cc : SUCCESS in 13m 57s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0052a1e829604601b2ff6cc1c8ec551e : SUCCESS in 4m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/90f7e49d676f40bea70ef4e6ccef00eb : SUCCESS in 8m 00s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e61853d8358942a0b02b49d9212d9d47 : SUCCESS in 8m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e8f88a6001954469ba2e2e8c7e9ada1d : SUCCESS in 41m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7335f67056c143c1b694fe9c003f4bec : SUCCESS in 54m 55s","accounts_in_message":[],"_revision_number":1},{"id":"7403228e91a89addc1ebebe14d81d3d994dd0bf0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-26 20:42:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"27f6aafa2bfc782b180ba1891eb546fd1a93265b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-26 20:43:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3906a63ecfe4a14901e4b708d8794c60c5797ff3","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-26 22:06:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"86c090aadc5b13ae6d6987acdfac728c0d6e8301","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-26 22:06:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"acbbbb087d1f37b3441c8e5d0e18c400d71f2432","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-26 23:14:45.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6ca0a17a57d5826f4a586b96c306bbd527bf7ace","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 23:23:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c4df67a2009b42c3a4d5121d1a9ba9aa : SUCCESS in 1h 05m 44s","accounts_in_message":[],"_revision_number":2},{"id":"e942f2d6018585d043f026f84be4f73318f529bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 00:14:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52a53951f26e43dd8635e8323c92d733 : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd0217b593a14266b249696e02153d11 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/edfae6f859a54cd5a1f8815ce3b76ac6 : SUCCESS in 7m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/840284fbe0ae4ae3bc359cb809056290 : SUCCESS in 5m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4201a3e986924c7084864ffa219ce095 : SUCCESS in 5m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bbd1e668c44347e4a12d1711386d0d50 : SUCCESS in 6m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/051f64c73b9749de84ffcc3663c5f817 : SUCCESS in 10m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/206c24480c684b92ac73f36d4b15aaff : SUCCESS in 10m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/07c488f0ebed4f39b10733552bdccb95 : SUCCESS in 9m 55s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e88021b41f254c22affe1345b5b5030d : SUCCESS in 5m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/caa94a75effb482ebc0792bcd68b06f0 : SUCCESS in 11m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7cd7d612d8ae41ecb85017ba28d970bb : SUCCESS in 7m 46s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/69be2136c20e4b369b3e763a425490ce : SUCCESS in 5m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8a6d238d8efa4928be111ef80d869a41 : SUCCESS in 12m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f389f0c14cec4461bf6700d72a5f5724 : SUCCESS in 17m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ae236b8738b042699e9eff3bdcbcdb6b : SUCCESS in 9m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b176d01191624ccdb2ef6e23ba11c104 : SUCCESS in 4m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2c2ccb561968442ea5146acb23287283 : SUCCESS in 18m 08s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/033e7b2ca6eb43bbac8529f7207b84f4 : SUCCESS in 6m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4556da2117f2460a8ff79541a5cf2353 : SUCCESS in 6m 28s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/88ea636e414048c29d62cbf6e4af9b33 : SUCCESS in 10m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1e0731e2b47b43b69041371409e0ac4f : SUCCESS in 57m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5515e8051f2648e997932e69fd7ae2ac : SUCCESS in 1h 23m 06s","accounts_in_message":[],"_revision_number":2},{"id":"ef1c689e374fe31051c4f8bfd6ec16e5d9b091c2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-27 08:12:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b5bccd32dd25bd9e99da3f05b24572bcd5f3df24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 08:13:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"49038fedce132ef246ee6cfef277a785a3c13b0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 10:04:24.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0f4d9e315324da5ada6e821cde9d103 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/865cbe6b980a40bc8ff79e55cfc25482 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/677f5af8b57e4e22b55c398f6a536f7a : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9523660678c4d9c81f9942f54aecdbb : SUCCESS in 4m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bc6640eb7efb4f0bab0b835c07bfbb23 : SUCCESS in 4m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f1c5e1c821aa455c986134eafaf5aba5 : SUCCESS in 3m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/67c70f8c501747d78e1c3415158020f4 : FAILURE in 5m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e9d8199a760143479c7b610d49218e4b : SUCCESS in 8m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2f8420809643439b87b412212a554956 : SUCCESS in 9m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7c0e0689e83a453882bec3a82208aa2f : FAILURE in 3m 43s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/62fe40c08ace4bdcadc481910b9e9334 : SUCCESS in 10m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8dae2bcbfe6849b1bb80fa988e433994 : SUCCESS in 7m 43s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a9d81c58fbe8496d9a9096131c55f870 : SUCCESS in 7m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/756c25953f404c17a8963121073c9644 : SUCCESS in 12m 33s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/09dacbba7c5d47caa04e976e5248a86c : SUCCESS in 28m 31s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8fa6b89a2cd748d3924f9e7c168591d3 : SUCCESS in 12m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d927d15c1f704d84acae6f40ff22f162 : SUCCESS in 5m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8220de35a2ed4ae69a40601b06e26e2f : SUCCESS in 19m 04s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/aedc79b4f4184980a3b7f03e0ea6449a : SUCCESS in 5m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1f0ae19ce94c4316b7ff59b842b94d87 : SUCCESS in 6m 22s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ec14246e0e914edf96ffaa91bb1688c5 : SUCCESS in 9m 43s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d9f9ae013c1d4dce9e5ed35c410007f2 : SUCCESS in 10m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/13353656d3fc41ab9bbd750aa11c12b3 : SUCCESS in 1h 02m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/11d8e05c91a04ddd9be8dd22061f011a : SUCCESS in 1h 43m 09s","accounts_in_message":[],"_revision_number":2},{"id":"f0d7913a99886c751e8ad524822f9ab333121e43","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-27 10:05:43.000000000","message":"Patch Set 2:\n\nrecheck\n\npip cache issues","accounts_in_message":[],"_revision_number":2},{"id":"cda3fd1abf2884f724df1550ae763243d68b9af3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 11:15:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2b3029ffc8aa439a9e0f69112ae1f5ae : SUCCESS in 1h 06m 01s","accounts_in_message":[],"_revision_number":2},{"id":"2ebb3f1e97f721f8c7df8eaad9903a8f82c2169b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 12:59:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eac721fe95264c2aaee607821a201975 : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec6151b05b654db0a1503257d2d57e43 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a0f392b6b7e45f99eca7a2701951527 : SUCCESS in 7m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4db0b21d4d7a48e78384d7469e73cc40 : SUCCESS in 5m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e3265b62ec2541e19c45b1811cbeb520 : SUCCESS in 4m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/40d969868367493d95c8776634fa4227 : SUCCESS in 4m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3b5cbfcbdb7146069dd6f219fe78f6e2 : FAILURE in 6m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0dc1efaf8b354eeab2930b414684872d : FAILURE in 5m 45s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ae5fbe4f88af41f38e3a373e7d02e68e : SUCCESS in 11m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b70067e16bac465bb4577a3efd897eb7 : SUCCESS in 6m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a4834614f391433fa93d0674e7701d03 : SUCCESS in 11m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/54589c3538be485d8a5153a5552cf0e9 : SUCCESS in 9m 08s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b8b4291fffb14c188ede71e76eed5028 : SUCCESS in 10m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/65298a478cb24979922851a1135ca0bc : SUCCESS in 17m 19s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/45615cdae5fe4af782534eb94809f8c5 : SUCCESS in 25m 58s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/62cd6dea2e7c4a7aa6568765519c526e : SUCCESS in 12m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4c873f46a9364f0898addc3fca46693d : SUCCESS in 4m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/036151731e184f13a162f15dffc59ce2 : SUCCESS in 21m 48s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e73695b4fe0b4a3ebacff147cdb0fdbc : SUCCESS in 4m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4a4a04790d864a3b88a5f178964dfb2d : SUCCESS in 9m 11s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f993d89f43ec4725a6e46b1218ccbc18 : SUCCESS in 16m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dd40f93d5f6048e5897a668e8b479460 : SUCCESS in 57m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cc99bf8c84074b1285573446acd39a8d : SUCCESS in 1h 13m 27s","accounts_in_message":[],"_revision_number":2},{"id":"cd2698c87e1aee87f1945e46a82e354eb28bdd7a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-27 13:02:24.000000000","message":"Patch Set 2:\n\nrecheck\n\ninfra issues","accounts_in_message":[],"_revision_number":2},{"id":"b115dc0b153a741628015d39ae39dd1941279f9d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 14:09:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/33e514a35d134cd2840f623c38573854 : SUCCESS in 1h 03m 48s","accounts_in_message":[],"_revision_number":2},{"id":"8cf968853ef2f0f318be46f4afb4dd1fa5f28cb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 14:10:26.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/473d72d58aaf457f9c969bea485d355b : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/559296d44eff480c8307985a4f6b8866 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c42626c6581c41c1b3df5635e66097d6 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52b4261ab1c144c184bb06f82fe8da38 : SUCCESS in 8m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/271272ab819b4e65ace7cdf5168ee5df : SUCCESS in 5m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2bbee0495eaa47739836c606b53d1002 : SUCCESS in 4m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6a0ec61257f440dea374f499f6df2a73 : SUCCESS in 6m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/aa93878f4f244d6d850379399e907e20 : SUCCESS in 8m 49s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f155937f3b864abfb21306bc9ca4622e : SUCCESS in 8m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0b8ee647c8c14191b51fdb6c5e308b59 : SUCCESS in 5m 20s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e53f7adc95944b5887f4e481d7a1eff5 : SUCCESS in 9m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8e69017ec2184e8e896d35a59d403ddd : SUCCESS in 6m 38s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7703418a80714f4bbb2c741a8f94294b : SUCCESS in 7m 26s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2420327a128d49ef92fe49d15736598c : SUCCESS in 13m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/26fc4e320bc548db95d360a593b36afe : SUCCESS in 29m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e4a38416f9a14d5f9006617b14a11309 : SUCCESS in 11m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5c20b247ea07496181c172dd352c651f : SUCCESS in 4m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2f34eb936c4447f3b7ee2360773673e7 : SUCCESS in 18m 46s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9a3752f2a1ad4483ab13d213bb8ea57e : SUCCESS in 5m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2660c4099f7542f18d79ea0a6e577d24 : FAILURE in 4m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/81bac2ff1e4a41faaecb947f3d1d754f : SUCCESS in 11m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bd2ead6d6eea4a079768a7bda91938c6 : SUCCESS in 1h 01m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1c7bbbd316ba4175b2fc3b833be73052 : FAILURE in 11m 19s","accounts_in_message":[],"_revision_number":2},{"id":"21cda55400966e894f0e0b1194dcd1fb0dd0b96e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-28 08:36:01.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9135ef7c9a631c2315a5b64f77c991072d6bb270","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 09:53:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a967375a11dd4a3e86070986b847bfbb : SUCCESS in 1h 07m 06s","accounts_in_message":[],"_revision_number":2},{"id":"2a4990460e42794f234d25ccd2afd530e4bf8e22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 11:53:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28eceaa77b9b46cd960eb0aa0120e0f0 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5d9fc28abff426dbac1a7b55457e694 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c58dc1929e184dd798d292abc60eacbf : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c71a6e84a14947d789eab49d241e5c44 : SUCCESS in 5m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cb4266bebd894e04813841953d536994 : SUCCESS in 5m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/39aa2e6310c5429a8684d995763374e9 : SUCCESS in 6m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6f72cc940f9c426982912b61d745dc93 : SUCCESS in 12m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3d34efb8badf47658822a68371a3dc56 : SUCCESS in 14m 09s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/93239600e6af49738f8c177a6cddf215 : SUCCESS in 9m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/cc0467bf31684d94a99aedb909caeb8f : SUCCESS in 5m 46s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d9b7723da8234a218e91e8ce5c5026bf : SUCCESS in 10m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2757383df0a942b884d26896f9c053d4 : SUCCESS in 6m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/359cbc987d9246bfadd1535a734db2da : SUCCESS in 13m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f7b25b891e6449009d399ab5a19b363d : SUCCESS in 13m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/58cf443fe43d4a98be0195b460d0076e : SUCCESS in 24m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2a0798448121453b8bf9becdc728b31f : SUCCESS in 12m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1bf407991c954bbf80b45de3cc0fbfb5 : SUCCESS in 3m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fc7b037736254e988fe878d061f4ee2e : SUCCESS in 15m 51s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/52604913f14044af98e259c1cfcc50c7 : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e3ba185067f648eabc658a27c5ec03ab : SUCCESS in 10m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/66f998e48414475fa2cb3e4f7a129029 : SUCCESS in 12m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b5f65dabf9024689a85bec9e49eb29ff : SUCCESS in 52m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b0b3dca748b947278e5e13140d905023 : SUCCESS in 1h 18m 35s","accounts_in_message":[],"_revision_number":2},{"id":"5d3de7e61bba5e3c15a2c8930e73fe06bfe9fd90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 11:53:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"03ef2323bdf6402162904712583175c46b7d26cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 14:06:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/148ce8bb774147afa562fb33d02b6981 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5238bfc1dc5e480ebc062b116844459c : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36c7b68611584a33afe786cc973611a3 : SUCCESS in 7m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fde46fa485ec4a74a17da65bc3365c90 : SUCCESS in 6m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d19acf5562c1499f8320825e669556cf : SUCCESS in 4m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/60da4e719d504477897f2e61f9198926 : SUCCESS in 5m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b86e40ba59624ffea9517577299263b7 : SUCCESS in 7m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/878a90b2129148aab1b62f1233a64f8a : SUCCESS in 9m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6b4c7332bc184964b55a143174e56dea : SUCCESS in 9m 52s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fe0e21ef5b884f2abdbc24a1241477ea : SUCCESS in 7m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9377d899ca9548159b3f9522f6e11741 : SUCCESS in 10m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2c2704476d5144e7b66d6e8ecbf56726 : SUCCESS in 7m 21s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/86883e740ef243ca8e7e0146c108e551 : SUCCESS in 8m 25s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/62f746aeaabd4fffb2fee0a0d78f34ca : SUCCESS in 20m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/445d502e8c614bdb9d69ab1d300515a1 : SUCCESS in 29m 57s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/45d3749abf7948b196bf80a5ac11d033 : SUCCESS in 12m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f899363fce514794992ee00bf5e0f238 : SUCCESS in 6m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/327094477b074fda9f0d05e73281cdfe : SUCCESS in 20m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/def59aaa571847f6819d86eb27d6ab5d : SUCCESS in 6m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7a21037e525c458daca5841a248f5dbc : SUCCESS in 10m 31s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a6713cd559ff4ab7ba6dc15bdafc631e : SUCCESS in 12m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b4af0671139d4e749b7cfb89c7761bbf : SUCCESS in 10m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4ff3a06837564911abbc7b62484e3dfd : SUCCESS in 1h 04m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d560a42678e84f59bf29b43629598ef0 : SUCCESS in 1h 32m 32s","accounts_in_message":[],"_revision_number":2},{"id":"d755c4e68a6b70c914149635cea5bd58bc5a2776","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 14:06:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"5c0803a6ea84f1772960113a95c3edcc0cd59c80","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 14:10:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e2d20a06d1d41fdb93d3bccab1ae1e1 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5494ce4b174e197bcceb84a4d1502b7d57e7446f","revisions":{"37fc950d2d4b3f548eddddc45cc29c239d5cda01":{"kind":"REWORK","_number":1,"created":"2020-08-25 13:23:21.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/32/747932/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/747932/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/747932/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/747932/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/747932/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/747932/1"}}},"commit":{"parents":[{"commit":"aca419d774fbfcd6292f36892bc1b90b6bbca888","subject":"Merge \"Revert \"updating gnocchiclient\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aca419d774fbfcd6292f36892bc1b90b6bbca888"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-25 12:54:36.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-25 13:22:01.000000000","tz":120},"subject":"Add httpx to global-reqs and upper-constraints","message":"Add httpx to global-reqs and upper-constraints\n\nProject openstack/octavia-tempest-plugin requires httpx for HTTP/2\nTempest scenario testing [1].\n\n- Is the library actively maintained?\n  Yes\n\n- Is the library good code?\n  Yes\n\n- Is the library python 3 compatible?\n  Yes\n\n- Is the library license compatible?\n  Yes\n\n- Is the library already packaged in the distros we target (\n  Ubuntu latest / Fedora latest)?\n  Yes [2][3]\n\n- Is the function of this library already covered by other libraries in\n  global-requirements.txt?\n  No\n\n- Is the library required for OpenStack project or related dev or\n  infrastructure setup?\n  Yes (octavia-tempest-plugin)\n\n[1] https://review.opendev.org/#/q/topic:http2\n[2] https://src.fedoraproject.org/rpms/python-httpx\n[3] https://packages.ubuntu.com/search?keywords\u003dhttpx\u0026searchon\u003dnames\u0026suite\u003dall\u0026section\u003dall\n\nChange-Id: Id832a0cd67eebc030407cde972360a1412a6a018\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/37fc950d2d4b3f548eddddc45cc29c239d5cda01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/37fc950d2d4b3f548eddddc45cc29c239d5cda01"}]},"branch":"refs/heads/master"},"5494ce4b174e197bcceb84a4d1502b7d57e7446f":{"kind":"REWORK","_number":2,"created":"2020-08-26 22:06:56.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/32/747932/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/747932/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/747932/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/747932/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/747932/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/747932/2"}}},"commit":{"parents":[{"commit":"cc8489e1eb0f87de5fb17648a5802489d56e8aee","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cc8489e1eb0f87de5fb17648a5802489d56e8aee"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-25 12:54:36.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-26 22:06:36.000000000","tz":120},"subject":"Add httpx to global-reqs and upper-constraints","message":"Add httpx to global-reqs and upper-constraints\n\nProject openstack/octavia-tempest-plugin requires httpx for HTTP/2\nTempest scenario testing [1].\n\n- Is the library actively maintained?\n  Yes\n\n- Is the library good code?\n  Yes\n\n- Is the library python 3 compatible?\n  Yes\n\n- Is the library license compatible?\n  Yes\n\n- Is the library already packaged in the distros we target (\n  Ubuntu latest / Fedora latest)?\n  Yes [2][3]\n\n- Is the function of this library already covered by other libraries in\n  global-requirements.txt?\n  No\n\n- Is the library required for OpenStack project or related dev or\n  infrastructure setup?\n  Yes (octavia-tempest-plugin)\n\n[1] https://review.opendev.org/#/q/topic:http2\n[2] https://src.fedoraproject.org/rpms/python-httpx\n[3] https://packages.ubuntu.com/search?keywords\u003dhttpx\u0026searchon\u003dnames\u0026suite\u003dall\u0026section\u003dall\n\nChange-Id: Id832a0cd67eebc030407cde972360a1412a6a018\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5494ce4b174e197bcceb84a4d1502b7d57e7446f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5494ce4b174e197bcceb84a4d1502b7d57e7446f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
