)]}'
{"id":"openstack%2Frequirements~748362","triplet_id":"openstack%2Frequirements~master~Ie9e8423d29413e5ad39698cf16b8889eda715ef4","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"Ie9e8423d29413e5ad39698cf16b8889eda715ef4","subject":"Updated from generate-constraints","status":"MERGED","created":"2020-08-27 06:16:29.000000000","updated":"2020-08-27 10:16:31.000000000","submitted":"2020-08-27 10:14:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748362-1598523272878-fda0f000","meta_rev_id":"98fcce84cbd021af502f3bf57e0a5d52dd413ae0","_number":748362,"virtual_id_number":748362,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-27 10:14:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-08-27 06:48:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-08-27 08:09:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-08-27 06:48:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-27 06:48:34.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-08-27 08:09:33.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-08-27 10:14:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bdfbcb7da137350425a979d78df7d812dde574ef","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-08-27 06:16:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac61e0a77bec5919f06b79ededc1efccdbe76c6f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 06:35:52.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5da8ac5203d24e149450e589a44dd7fd : FAILURE in 14m 15s","accounts_in_message":[],"_revision_number":1},{"id":"871848b5392b814e996d85c9d78896eb4972983a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-27 06:48:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fc91fdddc0ae61fb9b8192a8925557ee6850b94a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-27 06:48:20.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"be80dfc596d635feadd985a7d7a064046264fccd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-27 06:48:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7f0424fd025ee15c8ef48b72131feb97f812ad5c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 08:06:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b8fcdca76a77480bab9e405f706ead6f : SUCCESS in 1h 10m 40s","accounts_in_message":[],"_revision_number":2},{"id":"66ccdaaebfaf60c2aeca1f3128f7bd41feafbb57","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-27 08:09:33.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6aa3e9d64a01f8bdf1cc66c24018ad5bdd64fca2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 08:35:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c471920696314d27b32938a3396c7d57 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0ead5b7da8f473bbd7695ac502b6bbd : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/828692eea4fc4a84bb48aa2b63c30018 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3066fbd2615e4b5193735638b90ae150 : SUCCESS in 4m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/441810ef37d44610ad60c951445d45b7 : SUCCESS in 4m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6ec10226e5884832b246a96f309dbe4b : SUCCESS in 5m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/df9e50a7d9124291912302dd86655dfa : SUCCESS in 7m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/031d0d90d8814010ac79a0433cade701 : SUCCESS in 9m 19s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d71e8884b42b40daa33960cb8620bcd4 : SUCCESS in 8m 50s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f273456690144bfd9f507881ab3124af : SUCCESS in 6m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/130b8948e325472c8f6f637c7af58000 : SUCCESS in 10m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9d8b3dd924f7471dac3bbc2a6b16f852 : SUCCESS in 7m 56s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f871a266fd1c4389b41ee78ce82d9ffa : SUCCESS in 9m 13s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1a777532d9e54efda9a4944daa60f0d3 : SUCCESS in 24m 27s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a9c0372a63634250829ea90e69dda324 : SUCCESS in 17m 44s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e09b2ad20e8d4b43a2873a572613175d : SUCCESS in 10m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9eb15acef85e4678b85e2128b6a2a6af : SUCCESS in 4m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/051d303a595d4d67bb4ac47394b4f12e : SUCCESS in 14m 05s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cc79575f06624834be54ae13c655e8bd : SUCCESS in 7m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/70231b3d00984d74a94f63c6abfdca47 : SUCCESS in 9m 40s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/853460ec53d94f60b7f58e60896c1b30 : SUCCESS in 12m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/03bb4fecb80a47c29953c150a62c319c : SUCCESS in 1h 02m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5ab774d04e9b4bd282ee7930048d94fc : SUCCESS in 1h 19m 44s","accounts_in_message":[],"_revision_number":2},{"id":"4f31e85fc857a798327e445313214bb2f1713e83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 08:35:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d0231832206f529f99adf8971d9e20241a239928","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 10:14:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cc6ffa4e2c84ac79833a29b8140a2a0 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31d14bed897549019d0552658bf7e6e8 : SUCCESS in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b27ed3fe1ad24c8b810f62c997c7304d : SUCCESS in 6m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82af1469c78a407c89d361d56b28b3ca : SUCCESS in 6m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a63d58a6b3ee41dca90e83267e88fdd4 : SUCCESS in 5m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3468a226cd4a4d8e8fe9c9b84d7ba170 : SUCCESS in 6m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/66749afe680c49e4b008add51b174961 : SUCCESS in 7m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7153f171bf04465489422ff0640a6312 : SUCCESS in 13m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4eb64ac8f9ba45d5ad8191fe2121f220 : SUCCESS in 18m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f39cb30f87c54000b78ec4cd779adc5c : SUCCESS in 9m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/747ad2c4196a4badb3426349172bbe30 : SUCCESS in 11m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6376db2f357b4fa6866a4fc388c0cf97 : SUCCESS in 8m 28s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c70efcdcd3e948d7b92fb3b710d79dc1 : SUCCESS in 8m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e48e1c37966b4bc29aaaf0ac710dc2ae : SUCCESS in 12m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e27ef12781824b22816e91b75c47cd57 : SUCCESS in 26m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0c32e277e3e744858a12984f33c4b8cd : SUCCESS in 13m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1dc2732bb40048dc9704f787b7d1a9c7 : SUCCESS in 5m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/647aa081f822445a85638f53929d5e0d : SUCCESS in 19m 11s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c0d73fb2057b462884c91b15ceb3a1f1 : SUCCESS in 6m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/37d92486e24c48089a3fd42ae1656bfe : SUCCESS in 10m 19s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f1819caef24d40668c56d7e4e8c4ee20 : SUCCESS in 13m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/979b9669144b49cda43d590a374d1b34 : SUCCESS in 10m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fb58f63250f6471ab0d2e1dc5ef5d74b : SUCCESS in 1h 09m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/affb9c28ec8d4859b6e67936e1f6a94c : SUCCESS in 1h 25m 44s","accounts_in_message":[],"_revision_number":2},{"id":"c7209208eb81d553986f93aa5494f9a8802ac6cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 10:14:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"98fcce84cbd021af502f3bf57e0a5d52dd413ae0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 10:16:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3f8a93b88c04c0bb4942695cdf824fb : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1402e7c93c84800ea4c5196cd96e8ec8cea9bf5c","revisions":{"de13aa5146d7be55aea0dffa9ae294bd7e4ebf96":{"kind":"REWORK","_number":1,"created":"2020-08-27 06:16:29.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/62/748362/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/62/748362/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/748362/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/748362/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/748362/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/62/748362/1"}}},"commit":{"parents":[{"commit":"b79e28846ccddb5c4e9f5d006209e5dc9a5f8d33","subject":"update constraint for oslo.messaging to new release 12.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b79e28846ccddb5c4e9f5d006209e5dc9a5f8d33"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-27 06:16:27.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-27 06:16:27.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ie9e8423d29413e5ad39698cf16b8889eda715ef4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de13aa5146d7be55aea0dffa9ae294bd7e4ebf96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de13aa5146d7be55aea0dffa9ae294bd7e4ebf96"}]},"branch":"refs/heads/master"},"1402e7c93c84800ea4c5196cd96e8ec8cea9bf5c":{"kind":"REWORK","_number":2,"created":"2020-08-27 06:48:13.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/62/748362/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/62/748362/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/748362/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/748362/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/748362/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/62/748362/2"}}},"commit":{"parents":[{"commit":"b79e28846ccddb5c4e9f5d006209e5dc9a5f8d33","subject":"update constraint for oslo.messaging to new release 12.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b79e28846ccddb5c4e9f5d006209e5dc9a5f8d33"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-27 06:16:27.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-08-27 06:47:54.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ie9e8423d29413e5ad39698cf16b8889eda715ef4\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1402e7c93c84800ea4c5196cd96e8ec8cea9bf5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1402e7c93c84800ea4c5196cd96e8ec8cea9bf5c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
