)]}'
{"id":"openstack%2Frequirements~748376","triplet_id":"openstack%2Frequirements~master~Id87c3dfa4b4e5990e1a48ca72d9a7309a2f881d9","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Id87c3dfa4b4e5990e1a48ca72d9a7309a2f881d9","subject":"update constraint for octavia-lib to new release 2.2.0","status":"MERGED","created":"2020-08-27 08:48:18.000000000","updated":"2020-08-28 03:55:09.000000000","submitted":"2020-08-28 03:53:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748376-1598586814117-98515791","meta_rev_id":"a3019923afba012d561c2f4176a10b6454eb4965","_number":748376,"virtual_id_number":748376,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-28 03:53:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-08-27 18:39:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-08-27 08:58:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-08-27 18:39:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2020-08-27 19:07:35.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-27 08:58:30.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-08-27 18:39:18.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-08-27 19:07:35.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-08-28 03:53:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ca356fc7b9dfb1ffad0f78d15d50648a02e06ebe","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-08-27 08:48:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a17daeb9dd95b73072051663421a69f4ae59c2e8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-27 08:58:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f4eb20c9a90305e9ad85a4d2601bfe6d673fb1aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 10:01:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/dffcbbd47c1d420f8868dc3d584605be : SUCCESS in 1h 07m 02s","accounts_in_message":[],"_revision_number":1},{"id":"9d0be654456038ed8c6061df21f2949fdcb16d79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 12:19:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/236c7f676cf44fe6b2ed7c07a3465a09 : SUCCESS in 8m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58cb9719a9bd4707b2cc160d7cc48ea2 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6be7b6b017844ed29b62a6d00c85bf1c : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8e19f0e33374ecb9576bac5a87f81dc : SUCCESS in 5m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cfda9bbcfbfb431182b23c4c4483e555 : SUCCESS in 4m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bc62ddb106924892bcc868cf56ec06cb : SUCCESS in 4m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0896823254fe4905b4ee1a656df0d62f : SUCCESS in 9m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/48ca9242e60e44f08dfbcaed2e19b376 : SUCCESS in 11m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d68e5ab9f093440a9b0246349ebf2b27 : SUCCESS in 11m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bdb28094800c45378ec84b184eeab295 : SUCCESS in 8m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3e234510542b477babdcf130d0aebc1d : SUCCESS in 8m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5f8ae1f9d11249c8b2fc803edcd03f69 : SUCCESS in 8m 47s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b81bd6ad36df48dc95c6933f471b3171 : SUCCESS in 8m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1f9718506f774b36b23dee51facec1ae : SUCCESS in 35m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2f1415eda0e941ea80c9b713ca796a56 : SUCCESS in 26m 09s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2da38884fe634280a93f23e5e7953b81 : SUCCESS in 12m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dc0ad42eaa3f423badf06b8aea18a8e5 : SUCCESS in 4m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/20fa0bf1541741d59eb0589117b2ada1 : SUCCESS in 19m 21s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/47ea7c3401f3422792adb6e73f19412a : SUCCESS in 7m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/104590fea3f646249d0f8c0c9f56b1f9 : SUCCESS in 6m 38s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bea5d7197ab14b47b9cce215c430ddc3 : SUCCESS in 11m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/38983a248d4c4f69a956eba9620bf135 : SUCCESS in 58m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b1c56560990b4a3ba90254a427bf5dba : SUCCESS in 1h 16m 31s","accounts_in_message":[],"_revision_number":1},{"id":"73ca49aa2980707d4cbca61695d70f4a88d03775","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-27 18:39:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5983c4fa5f25ee903426e60743b73cadd645c9bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 22:55:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e9d5dde9092f555872e6160829ece4af5194434c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 03:53:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85b42b5063b0426d8f2f3f7bb93e02ad : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f72fac9c0e2470d93702e4af284a3fc : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d213b81b79c4922a5a504fc8dab15d2 : SUCCESS in 7m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c47c897dfc14c6e9ac02b2ce1d8b938 : SUCCESS in 5m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a70d62b3271e421abef4fb62228a37d5 : SUCCESS in 5m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4be968cc0be84e1bb0cb719955be8652 : SUCCESS in 5m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fd9cae8bac224fb99634fb74ea1b20cb : SUCCESS in 6m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/eca55ed17efe42068b1425dad120136b : SUCCESS in 9m 45s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1525834645a94baba878c3bfac6f38ec : SUCCESS in 12m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2118dd14349f4b91b6ee1e7ff432d9f4 : SUCCESS in 6m 02s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8340bef449bc4aff93f2d56e3066d62c : SUCCESS in 10m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7a23bd7322c54169bf7c72de13db0997 : SUCCESS in 6m 44s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3b8a3e99136447199d6b1ab522074935 : SUCCESS in 10m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/97af63ec64fe4a749a3b8c991ddcea3d : SUCCESS in 13m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ae04a9b8893a434e879f058a29625aab : SUCCESS in 27m 00s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/20ba69665a5e4184b0e6190cd9dad511 : SUCCESS in 15m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/904e93c1cfe54f8ba8dd69d2daaef116 : SUCCESS in 3m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/48d6d4684d94424c8e921a1ec82ec3ea : SUCCESS in 16m 24s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/26885216fb3547528e68ba16947831f4 : SUCCESS in 6m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/745ab27b5686405c9ab868edfc5e0490 : SUCCESS in 6m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9da8bc3204a740a28011eaf2f86a9f7a : SUCCESS in 14m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/98894a4dcdf64db2923c7813f02290de : SUCCESS in 10m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e5ba46c297c84854935dd6e6447e36d2 : SUCCESS in 52m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6f1e96a07a184cb48c52efcba97a2d2a : SUCCESS in 1h 13m 30s","accounts_in_message":[],"_revision_number":1},{"id":"3f7a37ab5c4cc1f0901dfe530d1f03a6951104f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 03:53:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4e061712ef92fb192cb630427a01723c613bf718","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 03:55:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b775fb39514f4c5da5b45d361d2d4568 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7687323c09d0a1e4c53e4d39b3297e0504e7277e","revisions":{"7687323c09d0a1e4c53e4d39b3297e0504e7277e":{"kind":"REWORK","_number":1,"created":"2020-08-27 08:48:18.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/76/748376/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/76/748376/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/748376/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/748376/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/748376/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/76/748376/1"}}},"commit":{"parents":[{"commit":"b79e28846ccddb5c4e9f5d006209e5dc9a5f8d33","subject":"update constraint for oslo.messaging to new release 12.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b79e28846ccddb5c4e9f5d006209e5dc9a5f8d33"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-27 08:48:16.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-27 08:48:16.000000000","tz":0},"subject":"update constraint for octavia-lib to new release 2.2.0","message":"update constraint for octavia-lib to new release 2.2.0\n\nChange-Id: Id87c3dfa4b4e5990e1a48ca72d9a7309a2f881d9\nmeta:version: 2.2.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nmeta:release:Commit: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nmeta:release:Change-Id: If4040cfebc6670daa8f7b3955a9461d077b1a273\nmeta:release:Code-Review+1: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7687323c09d0a1e4c53e4d39b3297e0504e7277e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7687323c09d0a1e4c53e4d39b3297e0504e7277e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
