)]}'
{"id":"openstack%2Frequirements~748943","triplet_id":"openstack%2Frequirements~master~I97e8fe6dd296d8552e256dc1dd85e9d13a9feafe","project":"openstack/requirements","branch":"master","topic":"setuptools","hashtags":[],"change_id":"I97e8fe6dd296d8552e256dc1dd85e9d13a9feafe","subject":"Reintroduce setuptools \u003c50 cap","status":"MERGED","created":"2020-08-31 03:03:25.000000000","updated":"2020-08-31 16:59:31.000000000","submitted":"2020-08-31 16:57:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748943-1598893077468-038e877c","meta_rev_id":"d0d33566d814223a8bc873c76bdf0200b8736473","_number":748943,"virtual_id_number":748943,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-31 16:57:57.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-08-31 06:48:59.000000000","_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-08-31 12:29:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-08-31 15:38:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-08-31 12:30:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"date":"2020-08-31 12:33:59.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2020-08-31 12:19:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":1,"date":"2020-08-31 08:37:47.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-08-31 15:38:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"date":"2020-08-31 13:07:11.000000000","_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-31 08:37:47.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-08-31 12:19:56.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-08-31 12:29:27.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-31 12:30:26.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2020-08-31 12:33:59.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-08-31 13:07:11.000000000","updated_by":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"reviewer":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"state":"REVIEWER"},{"updated":"2020-08-31 15:38:46.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-08-31 16:57:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d90d1cb3aec76279b0a609b8215e8fdee903568a","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-08-31 03:03:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"223ab2ffa69072cb5b7379f027e50576ab36317c","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-08-31 04:21:35.000000000","message":"Patch Set 1:\n\ntesting with https://review.opendev.org/#/c/748934/","accounts_in_message":[],"_revision_number":1},{"id":"ee392f10fbe5bba29e35a32092972ce1a03a86e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 04:30:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e39a57a5f9a348cc9d2f5e620510f7e9 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea2b8c468d664fb9bf3659f73aa5c85b : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb864538cb7248c6b46e86c9bd77ad98 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47fd0e732d1841baa364d48f42b1232b : SUCCESS in 3m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0c0474d43c334ac2bcdaabe5bd2321b3 : SUCCESS in 3m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/382cc1d744724592953eb7a73121917e : SUCCESS in 1h 02m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/878acd0c19634634a184834658044f05 : SUCCESS in 1h 21m 32s","accounts_in_message":[],"_revision_number":1},{"id":"9ca1c01d7141146db5e689516d9725a61ec92099","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-31 06:05:46.000000000","message":"Patch Set 1:\n\nmy question is how we move forward from this (either we adapt or upstream adapts).","accounts_in_message":[],"_revision_number":1},{"id":"0136864d5aa08b3f51d71b5ec3850bd05cd1b381","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-08-31 06:43:38.000000000","message":"Patch Set 1:\n\n\u003e my question is how we move forward from this (either we adapt or upstream adapts).\n\nthat is of course the million dollar question :)\n\nI think we need to backport this to ussuri for grenade too","accounts_in_message":[],"_revision_number":1},{"id":"aaa8f7b0dd3546dc9b69e426a4707e018ad9be8b","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-08-31 06:48:59.000000000","message":"Patch Set 1:\n\nI\u0027ve provided some links to devstack changes in\n\nhttp://lists.openstack.org/pipermail/openstack-discuss/2020-August/016907.html\n\nPatching in SETUPTOOLS_USE_DISTUTILS\u003dstdlib to install calls will probably work.  It really just punts this down the road until setuptools drops that too, however.\n\nI think we will have to keep our eyes on the upstream issues to see how debian and setuptools come together on this; if at all.","accounts_in_message":[],"_revision_number":1},{"id":"bc319b78c4d36e5e783a546e55b86ce2a05f54d0","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-08-31 07:37:30.000000000","message":"Patch Set 1: Code-Review+2\n\nI think that this is a feasible workaround to unblock the gate and let us evaluate and implement other options.\n\nCurrently on Focal it seems like we could resort to using distro pip, not sure how long that would hold up, but maybe the python3 world can get stable over time (yes, I\u0027m still naive, I know).","accounts_in_message":[],"_revision_number":1},{"id":"ded6a15762e2cefdfa6832310d7d2bb2e2168f47","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-31 08:37:47.000000000","message":"Patch Set 1: Code-Review+1\n\nOnly +1 to let\u0027s Matthew have the time to take a look twice","accounts_in_message":[],"_revision_number":1},{"id":"4e2675db98f939c86693009e283fe68ed1c8193a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-08-31 12:14:14.000000000","message":"Patch Set 1: -Code-Review\n\nhttps://review.opendev.org/749007 seems to be an alternative for devstack-based jobs","accounts_in_message":[],"_revision_number":1},{"id":"71c85e85329d99d8b871aaab59c8b80f65847860","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-08-31 12:19:56.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bea53022aab8d821347d15218557871b8c0c06f5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-31 12:29:27.000000000","message":"Patch Set 1: Code-Review+2\n\nAlternative only failing on grenade tests, so maybe just needs to be merged into stable first.\n\nBut in the meantime, this should get things moving again. I don\u0027t really like having caps like this, but it seems to be a big enough issue.","accounts_in_message":[],"_revision_number":1},{"id":"51b253dc5bd9070d9f0483be1ccba09c5a707af7","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-08-31 12:30:26.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d20ea374c678766e55ba4ee874633c4740f8d8af","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-08-31 12:33:59.000000000","message":"Patch Set 1:\n\nTo fix grenade, this cap needs to be backported to stable branches, too. See https://review.opendev.org/748934","accounts_in_message":[],"_revision_number":1},{"id":"2ce0327de29721a604eef49b92c2d412562476f7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-31 15:38:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nI\u0027m ok with this (hate it, but OK).  I do think we need to work on a plan to uncap, I\u0027m following the mail thread but is there a better place for tracking the fix? (is there a bug)","accounts_in_message":[],"_revision_number":1},{"id":"6744508d800f04f0d39d44cfdbbdc6d551355354","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 15:40:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"08a5697a6f748fd5d4fd38911c84d2b82748ec27","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 16:57:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ca3b761763f257e62992bbc1e952288ab0831739","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 16:57:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f7cdc2855f04de8a8014485a26cf5ea : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1aa829794b9b4b84bebaca529def01fa : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc8ca7cf415d46289845aff2f89ae167 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8695491f5a140d0bc2256d1dfd26ce2 : SUCCESS in 3m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3496ed1c513344a1a333749c5a67231c : SUCCESS in 3m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b84dc4ab62af4999ba824b5056db4f68 : SUCCESS in 49m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6ef40f25799f4643a0b554ba81915922 : SUCCESS in 1h 11m 42s","accounts_in_message":[],"_revision_number":1},{"id":"a523ae0050a9ff89b6b5ab356eaba6ec195df439","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 16:59:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62b49039b32b4615a446b0f07eb67a1b : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5b2ee12846e05722a45cd7ca616f8254b7ed1ec1","revisions":{"5b2ee12846e05722a45cd7ca616f8254b7ed1ec1":{"kind":"REWORK","_number":1,"created":"2020-08-31 03:03:25.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/43/748943/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/748943/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/748943/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/748943/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/748943/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/748943/1"}}},"commit":{"parents":[{"commit":"6e9786fd2076d0d74726925623cc5f3c3c6c960a","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6e9786fd2076d0d74726925623cc5f3c3c6c960a"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-08-31 02:54:40.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-08-31 02:54:40.000000000","tz":600},"subject":"Reintroduce setuptools \u003c50 cap","message":"Reintroduce setuptools \u003c50 cap\n\nPer the linked github issue, the setuptools has re-introduced it\u0027s\nvendoring of distutils so it not longer obeys the overrides configured\nby Debuntu.  This was last pinned with\nca1eb8dd0ae9c7c7ac149e8b4ff9bbc9630eb43f; setuptools reverted their\nchange, but it is back now.\n\nIn short, Debuntu overrides distutils to install everything in\n\"dist-packages\", leaving \"site-packages\" in case you install a\nnon-packaged interpreter which has that all to itself (i.e. packaged\nPython doesn\u0027t mix/conflict with a non-packaged Python install).\n\nThe latest setuptools now ignores the overrides Debuntu ships in\ndistutils, and thus installs binaries to /usr/bin and libraries to\nsite-packages; the first causing problems for numerous scripts which\nassume /usr/local/bin/\u003cfoo\u003e and the second meaning that\n/usr/bin/python3 (i.e. Debuntu packaged python) can\u0027t find anything\ninstalled anyway (because it\u0027s in site-packages, which isn\u0027t\nsearched).\n\nThis pin affects the devstack pip install from [1].\n\nAt this stage, we do not have a solution.\n\n[1] https://opendev.org/openstack/devstack/src/commit/e7625fc72cbbbfaf37a6682166bc2180d4e50855/tools/install_pip.sh#L153\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2020-August/016905.html\n\nChange-Id: I97e8fe6dd296d8552e256dc1dd85e9d13a9feafe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5b2ee12846e05722a45cd7ca616f8254b7ed1ec1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5b2ee12846e05722a45cd7ca616f8254b7ed1ec1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
