)]}'
{"id":"openstack%2Frequirements~749290","triplet_id":"openstack%2Frequirements~master~If9da9491718b28ea697133922b86750f2b550c19","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"If9da9491718b28ea697133922b86750f2b550c19","subject":"update constraint for openstack-release-test to new release 3.3.1","status":"MERGED","created":"2020-09-01 14:11:00.000000000","updated":"2020-09-01 20:12:02.000000000","submitted":"2020-09-01 20:10:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749290-1598991034154-8a9bf4a0","meta_rev_id":"5c63d92347c9a6bd079b77755934ad8369f97bdb","_number":749290,"virtual_id_number":749290,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-01 20:10:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-09-01 18:32:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-09-01 16:29:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-09-01 18:32:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-01 16:29:01.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-01 18:32:58.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-09-01 20:10:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"23986ccf7193804380725e368f60d386ba40b6dd","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-01 14:11:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f784367a84b2acb4f3ac9e35baba96eaac708ce4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 15:27:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a9f7a7bd80c2496288ccda5c4ed1a837 : SUCCESS in 1h 05m 25s","accounts_in_message":[],"_revision_number":1},{"id":"215c4f3a900f86b5151110ab4090415dacec7e21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 16:24:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50805da09c6e48518e6745c3448f3f33 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1f458353bf748049f3ba6a2b50b115a : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03b1943db6224a56a7788fd959acb874 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9f0bcef0f614f46b375d764fa9f7923 : SUCCESS in 4m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2b2647bf1774428e8a0d68cff84c6f70 : SUCCESS in 6m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b021633cf6a24f22a7518b24146288d8 : SUCCESS in 7m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a1ee26a0579e403a941d97c746f74693 : SUCCESS in 8m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/87f4644fb4f841659bb63e2b97cece23 : SUCCESS in 10m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f20f84da22854f0c92a8d904557a8f97 : SUCCESS in 11m 12s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ea43481eb0024d8fb7f6ade90fed0532 : SUCCESS in 8m 45s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/88714c11f2de4314b5f8c643dff747c3 : SUCCESS in 11m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/88b1d5ade97a4daa87c9eb55fc182629 : SUCCESS in 13m 04s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b77804c16d10438a8428c57c4df0d5b8 : SUCCESS in 6m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/28dd4036387b4d5c95d91573b446cbc8 : SUCCESS in 12m 27s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5da0ae8acdcc4386bdf9c4b65939d9dd : SUCCESS in 27m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b36bfffc66c24dfab90701d4e55acb55 : SUCCESS in 17m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7a824f0096fe473b9ed990b2483f7a06 : SUCCESS in 6m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ad4ad086347b416dbd8ec1a4da7e2646 : SUCCESS in 14m 02s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/448d3f59e67647648d236e02310384ab : SUCCESS in 6m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a7b7ee3c7d7c422ea8f813107e653888 : SUCCESS in 8m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/51755bbaf9304267ab4a3e606c67c00f : SUCCESS in 11m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a1c43898391b4e97b5a01fea39b78f15 : SUCCESS in 53m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/88ff0891ec3d4094ade7cc20388e95b2 : SUCCESS in 1h 24m 15s","accounts_in_message":[],"_revision_number":1},{"id":"5f859e08f8be5d74550029e21b6290bed5d860fb","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-01 16:29:01.000000000","message":"Patch Set 1: Code-Review+2\n\nI suppose we need to merge this to continue our release test, right?","accounts_in_message":[],"_revision_number":1},{"id":"34b3deefb3f224658d31d94699ec139728b2ef64","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-09-01 18:32:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bad72c32d2087c0f861863fbd82c125b7676ed3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 18:33:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8a32d9387f29afb7faf1c18f997e48d225ae766e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 20:10:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a65582277e564328a84ba7d5bfac9e1c : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd52ed7e6dcb4db888bbd37076f480fb : SUCCESS in 10m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c8dd6e57e47414daa608427f85f5105 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d028ef0f095a42788a8c9f8f42a1b4de : SUCCESS in 6m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4fe779cbc86f4bb5b197a6a9e98a1044 : SUCCESS in 6m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3153342088d34e60aa2f3c7faaf09552 : SUCCESS in 4m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8e285c77e1e74eb29c31cb807c7a86c5 : SUCCESS in 8m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/80385d63a5514909b6b7696e9b56632c : SUCCESS in 10m 37s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fc63e699a8a74714a47f211fb261bce5 : SUCCESS in 16m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1c87665902da4c66b2b5daf62099f960 : SUCCESS in 7m 03s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d2f94f9082e7412186c90716f714e457 : SUCCESS in 12m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/10e683b40fa043729b631aded220f48c : SUCCESS in 7m 02s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c5738abbe2e340c98e80d33a137eea51 : SUCCESS in 21m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ddda97e515154f72923f5866c7485552 : SUCCESS in 12m 53s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7689f04a04414c7981eb69434612d407 : SUCCESS in 26m 58s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/30e4ace9584949f6baaebdbf2dcbb075 : SUCCESS in 18m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2700d7d150894a7ea13683e67e461aba : SUCCESS in 4m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a39fd7d842bd40c5b3cb6e6bd29d0e6c : SUCCESS in 25m 26s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6d8ab92c75de4af6ab7735104c6ce5aa : SUCCESS in 6m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1b37f9dd8019493caaad816283014bfe : SUCCESS in 6m 48s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7b44621339ee425aad8177ab15cde6c0 : SUCCESS in 17m 03s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3faf489d73bd45b9b2363906b00848e9 : SUCCESS in 12m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ef5c6cb7cadc4bf6a5a474c715088675 : SUCCESS in 59m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b46a3e8aa6a54b0cb8771fd4d112353d : SUCCESS in 1h 26m 45s","accounts_in_message":[],"_revision_number":1},{"id":"0a18896e9147e0f61eef4321d20f6798be98aed7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 20:10:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"34e77967946335a2cfa4928e3c0a28e3743d0811","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 20:12:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e08408cdfdc54d06b6dd496b616f2150 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d43e1614d3cc2fc57066b96acfb48316c802fc4d","revisions":{"d43e1614d3cc2fc57066b96acfb48316c802fc4d":{"kind":"REWORK","_number":1,"created":"2020-09-01 14:11:00.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/90/749290/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/90/749290/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/749290/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/749290/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/749290/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/90/749290/1"}}},"commit":{"parents":[{"commit":"5b2ee12846e05722a45cd7ca616f8254b7ed1ec1","subject":"Reintroduce setuptools \u003c50 cap","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5b2ee12846e05722a45cd7ca616f8254b7ed1ec1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-01 14:10:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-01 14:10:57.000000000","tz":0},"subject":"update constraint for openstack-release-test to new release 3.3.1","message":"update constraint for openstack-release-test to new release 3.3.1\n\nChange-Id: If9da9491718b28ea697133922b86750f2b550c19\nmeta:version: 3.3.1\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: Ic2519a0f7291dd447d45264804cbe3151cef7135\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d43e1614d3cc2fc57066b96acfb48316c802fc4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d43e1614d3cc2fc57066b96acfb48316c802fc4d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
