)]}'
{"id":"openstack%2Frequirements~749330","triplet_id":"openstack%2Frequirements~master~I43bbeab61d7414296da738c816d8b88862625429","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I43bbeab61d7414296da738c816d8b88862625429","subject":"update constraint for glance_store to new release 2.3.0","status":"MERGED","created":"2020-09-01 15:20:18.000000000","updated":"2020-09-01 20:43:40.000000000","submitted":"2020-09-01 20:42:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749330-1598992925456-941d6ed0","meta_rev_id":"a3ca6b70313d55138ad46ba95d05f851e678ad9e","_number":749330,"virtual_id_number":749330,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-01 20:42:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-09-01 15:22:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-09-01 16:25:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-09-01 16:25:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-01 15:22:41.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-01 16:25:24.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-01 20:42:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7d7ba8b7b5aeb0c7f52a8c10c41ba3cf4b9554f7","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-01 15:20:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8969f3bf9d6e86304f1adfd79ad9e53ecb56c459","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-01 15:22:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a25279507da18f9e9f882b14db3279e89ef45ee0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-01 16:25:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d3b97fe4f4488bab64f87197aab01d1534fbc714","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 16:30:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1083fe8d3f434c68843557e0eedbc100 : SUCCESS in 1h 05m 10s","accounts_in_message":[],"_revision_number":1},{"id":"59b79d689597b24f99e07a7e9932b8ae476741cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 19:10:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79dfea3d59734b1a9eb6700961637879 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19488f59467d4ae6bf7840e9ed25b9af : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9224f0c178f148e19ee2af7794ea9bf3 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b5fd80d07274c3594ffb23cf8c43328 : SUCCESS in 5m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0ec07da3743b4384856acc545fecfa7e : SUCCESS in 3m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e083c6c2e5ea4c79bf4a3c718339d074 : SUCCESS in 5m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d91105b58a764cd5ae6f01bf4d515c92 : SUCCESS in 9m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e224e881e3db427d9e07771eb6b77dd9 : SUCCESS in 14m 44s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4c16e9677778429c800bcbbd26ba6cc5 : SUCCESS in 8m 32s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b5fd50c42eb34a42a63c1bbe36f3acdf : SUCCESS in 6m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/dfa0f6c75b2a4fb7a62c9d5a3ab2e5a0 : SUCCESS in 10m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e44aacb7e23f4858b3ca82850f3f83bf : SUCCESS in 7m 06s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/25eddf41069449b3a82e780af6d2aec5 : SUCCESS in 9m 10s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c4d484ef599046ecb975301fef09b1b2 : SUCCESS in 21m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/617b4df4882a46d9844d0756b5d9f6ff : SUCCESS in 25m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1a988ff70590491b97525830b0920ab1 : SUCCESS in 16m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e900106b6361462cb8eac7e966dbf453 : SUCCESS in 5m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/736d7f5f07864497a984e4d07a36e42a : SUCCESS in 20m 54s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/379e9e1713ae4c8f9317306331c098b3 : SUCCESS in 4m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0eca5424d8814f68ae832f4256fac47b : SUCCESS in 7m 00s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6dd553f2f03d4bd799ba35308290c364 : SUCCESS in 9m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e94268cfb359420f9cca4b9d03f29b28 : SUCCESS in 58m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d165e6f4cd8d420f82497d6e5a92e124 : SUCCESS in 54m 32s","accounts_in_message":[],"_revision_number":1},{"id":"20df85b675dad4744490862d1f2fd23450eaecbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 19:11:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a0cc7dbdfbe009207c660aaaf1bd57d54508bf3a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 20:42:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1ab75edf6e58848541a7b5db7130e5501ca1e1fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 20:42:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed0cb3b2ff574ddca60bf72ed2aec5f8 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35fc262a553e45e1984255fbab364eb2 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e7b84987ab74b5c9d4ed9229d1cb2f5 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa686f0235f7421ca4e5045d8028a9ec : SUCCESS in 5m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/93f5db897ec84b73a28b00a2f6ddee58 : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/345a372c0f1c4efebc2d4b9bbac5db8f : SUCCESS in 3m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e5d304bfeb6f4d868afa7da6217fb670 : SUCCESS in 9m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7c5476be0fa34c4f9f7edb3fc18fb3fa : SUCCESS in 9m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/abb14762c8db4b839f22a4210f7544fb : SUCCESS in 8m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ed4440a089254744bce656315abf945e : SUCCESS in 6m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c05bdf6206be4903a35d4ad2259e9922 : SUCCESS in 10m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b956349a12874086ace5ee325de519d1 : SUCCESS in 8m 34s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/087f9fc7dfc84ccba347042736920d2d : SUCCESS in 7m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ed6bdc66911d4b10adb8ea9aafdbf3e6 : SUCCESS in 20m 59s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/be804cf9c7744850bee4dbbd700384c5 : SUCCESS in 28m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b2268ecd18bc4b22aa55cd942e82909d : SUCCESS in 18m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8aec8d93ddf649fa8d2660cb3ad9be44 : SUCCESS in 4m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d536c65044e5461da2ca9d1ac2334ac3 : SUCCESS in 19m 04s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5792159716904fc9962b7ba1c6732d8b : SUCCESS in 5m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5be0cb4db5584cb58abbdc3719c4a3a3 : SUCCESS in 11m 50s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/472bfe55565a4853abc0f18b730bde5d : SUCCESS in 11m 00s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/80d105b6f265415da2e404240bd94faa : SUCCESS in 14m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fb8cd1ae2e3a474cbca1fb563c0d2bcf : SUCCESS in 42m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ee2380f5ca9f4bef8227202bed0b0414 : SUCCESS in 1h 24m 05s","accounts_in_message":[],"_revision_number":1},{"id":"f1fccda318703f5cc83086010f8d8ed3ce3f3364","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 20:43:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb190b44ecf846e3a42fc8860510d813 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"10c3e441a52baa6bbb06221224950af92d7e98de","revisions":{"10c3e441a52baa6bbb06221224950af92d7e98de":{"kind":"REWORK","_number":1,"created":"2020-09-01 15:20:18.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/30/749330/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/30/749330/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/749330/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/749330/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/749330/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/30/749330/1"}}},"commit":{"parents":[{"commit":"5b2ee12846e05722a45cd7ca616f8254b7ed1ec1","subject":"Reintroduce setuptools \u003c50 cap","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5b2ee12846e05722a45cd7ca616f8254b7ed1ec1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-01 15:20:13.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-01 15:20:13.000000000","tz":0},"subject":"update constraint for glance_store to new release 2.3.0","message":"update constraint for glance_store to new release 2.3.0\n\nChange-Id: I43bbeab61d7414296da738c816d8b88862625429\nmeta:version: 2.3.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Abhishek Kekane \u003cakekane@redhat.com\u003e\nmeta:release:Commit: Abhishek Kekane \u003cakekane@redhat.com\u003e\nmeta:release:Change-Id: Idf841c03e06012edc62bf563f31d1deae9081671\nmeta:release:Code-Review+1: Dan Smith \u003cdms@danplanet.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/10c3e441a52baa6bbb06221224950af92d7e98de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/10c3e441a52baa6bbb06221224950af92d7e98de"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
