)]}'
{"id":"openstack%2Frequirements~749341","triplet_id":"openstack%2Frequirements~master~I3bafc04615e562328715599ecdc4136c6d3ac51e","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I3bafc04615e562328715599ecdc4136c6d3ac51e","subject":"update constraint for stevedore to new release 3.2.1","status":"MERGED","created":"2020-09-01 15:24:48.000000000","updated":"2020-09-01 21:01:26.000000000","submitted":"2020-09-01 20:59:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749341-1598993964213-42a114cc","meta_rev_id":"5bee6b111c5b315ed030fc335c91ce0bebdc66a1","_number":749341,"virtual_id_number":749341,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-01 20:59:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-09-01 15:33:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-09-01 16:26:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-09-01 16:26:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-01 15:33:05.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-01 16:26:40.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-01 20:59:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c0cdbac224cd7af0212560ad9a51c311afdfc9f6","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-01 15:24:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a12abe7381113ae6708462f9dd4c7e6f4673353","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-01 15:33:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8d2f5ba2298900e9311f44724bc84708c399c1d4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-01 16:26:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c6665b3aa88ae292a99d81b85746247471e2401c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 16:36:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/af4894f2cd4a48a9a3956e9ad2ccd372 : SUCCESS in 1h 05m 14s","accounts_in_message":[],"_revision_number":1},{"id":"eeca9785ab615e7d5d632b460fed6484b4dfa528","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 19:30:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f028ecebb1b34f75b0953d7909186282 : SUCCESS in 10m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8379ea4d1ff54d5e8a08112e59c613e6 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f968e8936e042b9abbf3ff62a694705 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12109d904d8c403aa4a006037e6be277 : SUCCESS in 4m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4ffd6a0623cc42c0ab861d415a64b5ca : SUCCESS in 4m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5b8c60b7332e4d668bc1b018cca38f35 : SUCCESS in 4m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7d4de35bf94b4266a3bdf512606398bd : SUCCESS in 8m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8f30067164fe4cdbb7060d57bc7c5800 : SUCCESS in 11m 16s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e3b52fb3d3064b61ab1bbda38387f5d4 : SUCCESS in 9m 06s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5a5819275e4a4847a134c09808d67b0a : SUCCESS in 5m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f44f8273b66f4e8083b3b9b34831ef10 : SUCCESS in 10m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/42f70db8749f4112a7fd1324b663e8e3 : SUCCESS in 7m 28s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/147fca84423f42cb9b175e0713449377 : SUCCESS in 6m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6ecb280f17484115912f28d4cf6232ae : SUCCESS in 11m 57s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fc99fb80da044977be0e93372abbf7d8 : SUCCESS in 27m 40s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4df95366a81f4419a4e5944d0a1de70a : SUCCESS in 13m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3b884c8488a64a82b915c4dd65c1e331 : SUCCESS in 4m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a108a1009bb045a497837e81059bb2d0 : SUCCESS in 18m 41s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/284c9161b5734a7481915a469c6e9b6c : SUCCESS in 7m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d9cd0c2a6f7b4a7589b3b1a63128c8f0 : SUCCESS in 6m 27s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/df6602c2fef64e2e8819489699820f06 : SUCCESS in 14m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c1e05148555a4d88a2f56b8ca6d47232 : SUCCESS in 1h 05m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d1c2ccff5f1d4badb4162f7f6d9294ab : SUCCESS in 1h 10m 33s","accounts_in_message":[],"_revision_number":1},{"id":"038269dc22ede38f1ab225563e10ff9b8fc2aa46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 19:30:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d4a5a2e71648e091ba28f8a34312e5a68a29a4c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 20:59:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f719abd0ac843afb21d72426d5fcdf6 : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a384b021e0d04d00a49976392ac5b680 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3a3818044a54df398dd72b19bff2c43 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe3737fedfff4d1787491c00ee45f0ec : SUCCESS in 4m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c62c033e0dd9478eb0e4974a33b7380a : SUCCESS in 4m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e7a8f9f5cf4e4e2d8d7376e1d8b3d44d : SUCCESS in 4m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/27246bafe3694cbabe32f485fce34fb2 : SUCCESS in 7m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bcc8ae2b2f0646efae07d5da47005367 : SUCCESS in 9m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/373cc6a056ed49c39a0ba01be93c346e : SUCCESS in 9m 55s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/376bbac918ad485d88554e03231bc24c : SUCCESS in 5m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/06a99d39e1514ccbbcd4566eab76fdeb : SUCCESS in 11m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7e466c63f344443e80dff8bc04fa755a : SUCCESS in 9m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8342df92bb1340028e971c7cfeec53c3 : SUCCESS in 13m 03s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d378666d0e4943449926a68a01895a56 : SUCCESS in 14m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/baaccc6e778b44c3a7608cc06936a690 : SUCCESS in 29m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f50d55db5e4849f4b5d07e7d00e4e48f : SUCCESS in 15m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/228bff4902a94f9d84420290ee28f625 : SUCCESS in 5m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dada6bc2f2a54ef3925ba87076dbfcff : SUCCESS in 20m 10s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b4745facad844b128b492ee0c82cc291 : SUCCESS in 7m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5353c0504297452494ef61191cf6cecb : SUCCESS in 7m 54s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f56cbc5760a84b168fb6be415897c4db : SUCCESS in 8m 16s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5cba3cc8b8264e96b740bac3f6d94309 : SUCCESS in 10m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1d274b0eef3848d5981f463572c65488 : SUCCESS in 1h 02m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/539a7ed8b4944255adcf0677afb6e06f : SUCCESS in 1h 18m 03s","accounts_in_message":[],"_revision_number":1},{"id":"033d0f7d07427bc1388eb9c1ef7ae88021b8d476","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 20:59:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2ecaa0b7b4a1ad65dbd577a2722176e6d395cab9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 21:01:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eec2a200e93f421da2a34959bb63a785 : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1a9cd0251e01f91387394e203dfcb63c75a0e682","revisions":{"1a9cd0251e01f91387394e203dfcb63c75a0e682":{"kind":"REWORK","_number":1,"created":"2020-09-01 15:24:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/41/749341/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/749341/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/749341/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/749341/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/749341/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/749341/1"}}},"commit":{"parents":[{"commit":"5b2ee12846e05722a45cd7ca616f8254b7ed1ec1","subject":"Reintroduce setuptools \u003c50 cap","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5b2ee12846e05722a45cd7ca616f8254b7ed1ec1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-01 15:24:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-01 15:24:46.000000000","tz":0},"subject":"update constraint for stevedore to new release 3.2.1","message":"update constraint for stevedore to new release 3.2.1\n\nChange-Id: I3bafc04615e562328715599ecdc4136c6d3ac51e\nmeta:version: 3.2.1\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Change-Id: I634e7d331d55fbbd83b1bb7e5059d4be6ec86aaf\nmeta:release:Code-Review+1: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1a9cd0251e01f91387394e203dfcb63c75a0e682"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1a9cd0251e01f91387394e203dfcb63c75a0e682"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
