)]}'
{"id":"openstack%2Frequirements~749343","triplet_id":"openstack%2Frequirements~master~If10d8e96895be088f43960596b9d62f96274e41a","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"If10d8e96895be088f43960596b9d62f96274e41a","subject":"update constraint for castellan to new release 3.5.0","status":"MERGED","created":"2020-09-01 15:25:08.000000000","updated":"2020-09-01 21:20:55.000000000","submitted":"2020-09-01 21:17:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749343-1598995061826-941d6ed0","meta_rev_id":"7f72151923de837816aafad825fdc46c7e478466","_number":749343,"virtual_id_number":749343,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-01 21:17:41.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-09-01 15:33:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-09-01 16:27:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-09-01 16:27:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-01 15:33:24.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-01 16:27:16.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-01 21:17:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6a68265ae8308a8b7df9c074d42d05ec9826eb1f","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-01 15:25:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7dca4df3d79e22196d86f7ad4efdadfda8d254e1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-01 15:33:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5e053feb1890678c1fc63061749b3ccfe1c04163","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-01 16:27:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0094de1287e16375e4f97f13e2c4aed7368c5f23","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 16:36:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5d762c7a20d943c19ee6f4482f5fdbbc : SUCCESS in 1h 05m 08s","accounts_in_message":[],"_revision_number":1},{"id":"c7ae26b7a7817065765df8381365fce0891d464c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 19:47:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d25e67d84df840f7bf84cfae30bd3a8c : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9485bf0dae6d4395a291cb1cb93697ea : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3a24b83215c4dc5b626cfeea381a62a : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/87091b316bae4a14934b27b3e90abbfd : SUCCESS in 9m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/73274751d85845dca2d81ec27c54cf82 : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fb03196c9f08475da7451685028340e7 : SUCCESS in 9m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5882fa1457734942acfabbc7d5e7d403 : SUCCESS in 8m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/919676f66055489590422c27e3f1fe9c : SUCCESS in 10m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c0a22d32d67245c5948e5d91410347d2 : SUCCESS in 9m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a6e55bf76b8f441c9a42c4fd8ba7eade : SUCCESS in 6m 20s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/de880b6bb98346db8de5f5c856779432 : SUCCESS in 10m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1e5b1390f909407ea2f2ef340d0336fe : SUCCESS in 12m 27s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/bfdc30aff65e45ea9483be6636eb4eab : SUCCESS in 6m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ce66fc3dd9d34e48a1391aa400905e56 : SUCCESS in 12m 44s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ab9669f3acf2414bbb7576ee1eea4189 : SUCCESS in 26m 09s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1c00f21a1b5c481daaa749b20435ce6b : SUCCESS in 13m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9403a2468a114200a01e56d37f5c43fa : SUCCESS in 3m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/09a702664ead48de8b7051a3264c83e2 : SUCCESS in 21m 13s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3ac11f9c567543e7a969f83be0649dcf : SUCCESS in 4m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6267dbc557b14dfa81dc72e00c4f7cab : SUCCESS in 7m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bb964c805a3243e1b4303c0752dd2198 : SUCCESS in 10m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f04aec91af634dfb8eadaa34d5fe559d : SUCCESS in 53m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/54bd118ef2b646b98ee41c3a716c3116 : SUCCESS in 1h 19m 02s","accounts_in_message":[],"_revision_number":1},{"id":"ad8b98f87acf6270ef5d01de1bc045e50ba9a9f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 19:47:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"66919c03e1b61dfb15dd2b67dbe9b8a871816521","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 21:17:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e59364bd6b9a272b3e163c1a808fc4289d2aecd3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 21:17:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57aef69d97f740d997513a78c4ced01f : SUCCESS in 8m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e67c08261dc4b78b07d0f0ec923fa20 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fac6f23c204147148238882fd40a433d : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/acafee33dd104279b250c58003757da7 : SUCCESS in 4m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f10ce74b429c44e88ad5ec1cc9b3de80 : SUCCESS in 4m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a75d2c4ecd0e436ca16bc6b56974f802 : SUCCESS in 4m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/01d34f8d7a89489f8709ffb448359873 : SUCCESS in 7m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b6a68da423a64cafad9c705e4608915b : SUCCESS in 9m 14s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b5e20aa8751844b2994d33a6c902ff92 : SUCCESS in 10m 36s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b4d1b261572d49198449b12cbb467a27 : SUCCESS in 6m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/48f3073f13054de499bbef77f391ee13 : SUCCESS in 13m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/757fcc2e4c6446cd934837eff37d151f : SUCCESS in 6m 53s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b838c14da479498c91c66c8efef5528a : SUCCESS in 8m 01s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/28dde0b419df4d91b08126fe1bce999d : SUCCESS in 20m 29s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8230450e9e1348f19cfad5c3a4fe26a4 : SUCCESS in 20m 40s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/984c398376424950b4836998c9e22b08 : SUCCESS in 13m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1fc4f1e5d2344202ab64b0fdf6e9c6c6 : SUCCESS in 4m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/05282ca65e3a435f80c0968043db5f21 : SUCCESS in 20m 01s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/12c900d1b52b436b98991a98924f81e6 : SUCCESS in 6m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b7a5f2f9d5a04f72bc743613463f3982 : SUCCESS in 7m 52s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1c4b39f7a56f4ef092b12841004967aa : SUCCESS in 11m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fe3b010704784549a787c20b07465b5d : SUCCESS in 10m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e562e991cbc64a76b7637b43e2269dac : SUCCESS in 1h 00m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/43757a52391942edba92b8d1faef41fd : SUCCESS in 1h 15m 54s","accounts_in_message":[],"_revision_number":1},{"id":"be4d193f1e169f506c864b43790a9113aac63bc9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 21:20:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1488ba6590114130b1c81faa23d25a90 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"091f57c06d68715c8222b4dc394689512f7c57eb","revisions":{"091f57c06d68715c8222b4dc394689512f7c57eb":{"kind":"REWORK","_number":1,"created":"2020-09-01 15:25:08.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/43/749343/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/749343/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/749343/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/749343/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/749343/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/749343/1"}}},"commit":{"parents":[{"commit":"5b2ee12846e05722a45cd7ca616f8254b7ed1ec1","subject":"Reintroduce setuptools \u003c50 cap","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5b2ee12846e05722a45cd7ca616f8254b7ed1ec1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-01 15:25:06.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-01 15:25:06.000000000","tz":0},"subject":"update constraint for castellan to new release 3.5.0","message":"update constraint for castellan to new release 3.5.0\n\nChange-Id: If10d8e96895be088f43960596b9d62f96274e41a\nmeta:version: 3.5.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Change-Id: I634e7d331d55fbbd83b1bb7e5059d4be6ec86aaf\nmeta:release:Code-Review+1: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/091f57c06d68715c8222b4dc394689512f7c57eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/091f57c06d68715c8222b4dc394689512f7c57eb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
