)]}'
{"id":"openstack%2Frequirements~749442","triplet_id":"openstack%2Frequirements~master~I4b33930856be9b05be47fc8599dad30eb8bbb9d6","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I4b33930856be9b05be47fc8599dad30eb8bbb9d6","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2020-09-02 06:29:24.000000000","updated":"2020-09-06 06:19:10.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b96e1683b596a19d4729073b9c73b64361e0c014","_number":749442,"virtual_id_number":749442,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-09-02 14:21:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-09-02 12:53:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"date":"2020-09-02 14:57:06.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-02 12:53:24.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-02 14:21:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-02 14:57:06.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"ed2c7657cf6e24cc95a78fd687a01bbde30926d1","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-02 06:29:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d25d699f1f0673a41fa27b96f7c29a27642e1489","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 06:49:53.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3d1de1d773b54939a838ad62277701a2 : FAILURE in 15m 27s","accounts_in_message":[],"_revision_number":1},{"id":"2648f4f3aa319ba4b6701fdb99e103a0d2c4dc2a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-02 07:38:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1cf8b51ee3ac5d3244e65c6edbdcf39bea23920f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 08:35:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f890f2c5af04888ab59df1ed59a46f8 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10f7c95eff24400f8d9bcbf21cb3afd0 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40560cda68ea4f818ae471fb03b2f110 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5867b107641e4eaf9a22ff190700350e : SUCCESS in 5m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f520e543edf34f1db05cce19dbab1e61 : SUCCESS in 4m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2e508d6d253b4ff28eeac3c0873943c8 : SUCCESS in 4m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/65f83737f707492fbfba0601f294fab5 : SUCCESS in 7m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fb8e0d74422042e1b13f2ce97f0c3a92 : SUCCESS in 9m 38s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/74a3cf01a52b44a29b3ed776879aacab : FAILURE in 9m 00s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0d1b5f153bfd4397858a2e3cfaa17493 : SUCCESS in 5m 49s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b596f549bc1e47afba71ee3c9025e1b9 : SUCCESS in 11m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/768a456d32664024b4dfcaead41eaa66 : SUCCESS in 6m 32s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/70c7be78aafb44dca25052c1dab02955 : SUCCESS in 7m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8fe3e6eacd3944cf9a460a74bb434102 : SUCCESS in 12m 36s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7ff27c0ef608478f98e900ca885c3c9a : SUCCESS in 35m 57s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e83649124b7d498da865bc4bef26cde3 : FAILURE in 11m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/829da4fd370d4999ba8d46b5f4950c39 : SUCCESS in 5m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/712312e3d6084f36b1495920427f8ab8 : SUCCESS in 18m 11s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b975659c19f04ec39b9080ffbe913892 : SUCCESS in 5m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7ddd613549724487a62c2c822b7b4896 : SUCCESS in 7m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/293c868450e9478cb6e9a387dd20905b : SUCCESS in 14m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1681252c68254bb8ae08f9c22098a0a9 : POST_FAILURE in 39m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/518a1714aa324f6691ddf894a5d44a3a : FAILURE in 40m 01s","accounts_in_message":[],"_revision_number":2},{"id":"7da6aa15a434ce44a717842b68dbbce210854dd9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 08:52:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/905560c0b28748a099531d71c53481e1 : SUCCESS in 1h 09m 00s","accounts_in_message":[],"_revision_number":2},{"id":"271b733ffcba573dcee74ecc0dfb0a9d70958c67","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-09-02 12:50:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"21ca48a2a68e8292477c9d1af214247e473f74cc","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-02 12:53:24.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)\n\nWaiting for CI","accounts_in_message":[],"_revision_number":3},{"id":"396f82f9d714b446cd990e8b916f20d85cba05b2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-09-02 13:13:13.000000000","message":"Patch Set 3:\n\noh, sorry, didn\u0027t see you were working this, did think it odd that the othe studd was fixed lol","accounts_in_message":[],"_revision_number":3},{"id":"03284d8ca8a105f2d34c34ef1452576717089c66","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-09-02 13:13:25.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":3},{"id":"ce6c074a71bc53ca84bdfd009b40d81391620326","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 14:04:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/af7e8e7b321b4f85890276788361cff7 : SUCCESS in 1h 06m 06s","accounts_in_message":[],"_revision_number":3},{"id":"88df917707023eb5ed756b8953c770071ac05f2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 14:21:38.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/986625f62256449f8fb8ade4b0d05350 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d849c14a80f94440bb9de2b854ddc8a3 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2169445364f64ad3a021f338eea150aa : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc9bd61ff1584faeb1a947258d972e98 : SUCCESS in 4m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5c95cf82a86d4022a1f9bb3b874b6284 : SUCCESS in 6m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e3f671c4e3714554943a756d61177f66 : SUCCESS in 5m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c9634e954a364bf1b7e1466a78aa851a : SUCCESS in 8m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/32114aaf2fa84b1db0a72608e4d37dff : SUCCESS in 9m 49s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e8d48bc6da1a4bc991bff8be74244f96 : SUCCESS in 15m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/30fe7038bded491dad9f2f196ed06b4f : SUCCESS in 8m 17s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4bfa03e478c941968038bd10cf32aedb : SUCCESS in 10m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7a3806ba59e34f41a7975c1b4d2c9f05 : SUCCESS in 7m 27s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/dc712210b766414c8136e09f10fb1973 : SUCCESS in 11m 01s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/24c467547ce141ed984cef6d45291b66 : SUCCESS in 14m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3ae6ce881fae4c2b90a71730efc16e67 : SUCCESS in 27m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fbb4cc7439464924ac663b78fb77f328 : SUCCESS in 16m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8c0098a86b6c4d72978f0cdce9fdce50 : SUCCESS in 4m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0add706da6f64dc39d4fba84d4d42132 : SUCCESS in 17m 46s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/546d1706de384029b1f6f19d8e9ccebe : SUCCESS in 6m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/be7c999e85b34de292d03769f0c3de7b : SUCCESS in 8m 54s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b9f382e090304848a6aeef8f3ef215f6 : SUCCESS in 20m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e85d4b276a5c434e839d66429a209141 : POST_FAILURE in 23m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6bf518e5c1604d449a3e6a80f2b1263d : FAILURE in 18m 34s","accounts_in_message":[],"_revision_number":3},{"id":"4034f1781b0a0fbcf4ab2c24aa25789e29329b87","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-09-02 14:57:06.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ab2da38e09257ee426179e21aa4dd4ab2ed316f8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-02 15:38:23.000000000","message":"Patch Set 3:\n\n\u003e oh, sorry, didn\u0027t see you were working this, did think it odd that\n \u003e the othe studd was fixed lol\n\nahah no problem, sorry for shattering your dreams ;)","accounts_in_message":[],"_revision_number":3},{"id":"96faedec65fb1ebbeec0cb0541965f56fe136bd5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 16:10:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/84fcf7f294b24899a801038e53539430 : SUCCESS in 1h 08m 43s","accounts_in_message":[],"_revision_number":3},{"id":"a1b528ee93e263944ce115ab5dc8311178067989","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 18:27:33.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/137def76d9754349bea37caec7094f23 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39653246d0954a508b5ebc9fa8c643cc : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea83760e84c54f04be57721d46857b2f : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86ad3ea43f6149f79039d6d76c62ff1c : SUCCESS in 6m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/597a3cf376424ec8bf0d5267eb3a0689 : SUCCESS in 5m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e3ae9958b9e74e5f95ab67b557b6fe0c : SUCCESS in 4m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b5dc7aee175047b1a298b8d89515de82 : SUCCESS in 7m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/833d27f8e3804256a3aa8be420e1d338 : SUCCESS in 9m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7c163441366f4db7be7a89b7e34f147e : SUCCESS in 9m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9018f6a4041244f9b0dae3081694c704 : SUCCESS in 5m 17s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2d900806840d429dab459a7f5e890863 : SUCCESS in 10m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c1def743a521425ba67610ee61a5250a : SUCCESS in 6m 21s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/701c13c4e1cd45bf80632eb0022db196 : SUCCESS in 7m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/15c1243574884eca83cb28980c11d031 : SUCCESS in 14m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8ccc19453174413b9be31ba0f59011c3 : SUCCESS in 26m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e5febe43371e4493bda72312780ba2bb : SUCCESS in 15m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e5bb947d15104e30b2dad470d6f0847d : SUCCESS in 5m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/222de3b50ce046d4933069536a29699b : SUCCESS in 18m 57s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5a0aabcafa464b45b333b77ec4baa05b : SUCCESS in 4m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/37c29dded8f4498cb061d68181a60e93 : SUCCESS in 7m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/65eb0dbe26fa4864b751279c44b0cf1f : SUCCESS in 10m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5e5f8d00a18e4ea6b7e26c4b2162377c : POST_FAILURE in 41m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3a1c4002be314956a304dd9909e1f2a9 : FAILURE in 33m 54s","accounts_in_message":[],"_revision_number":3},{"id":"9b10f48e213f501437205fff691058a73468a96f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-02 18:29:17.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"670bbcd7a21acdc8fcb7f2e88d6a47d0255518f0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 19:43:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d34e36f4ee9946f6b8f275157a55d5cc : SUCCESS in 1h 07m 55s","accounts_in_message":[],"_revision_number":3},{"id":"5742b75b13e177c49cafa9327c3f14712bf40198","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 22:52:04.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3a7e8306c1e4589886bc24e50a8610c : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/265ce34a95694956a7d2da18615e0052 : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b785e8644fa48f8bfd13bf5d60dc003 : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e793640c0ea476199d94ad93d64899f : SUCCESS in 6m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/319687eff0154e1b81311bd53b87de80 : SUCCESS in 4m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a15e2206d72e42c2b43ea2dd5d583ffe : SUCCESS in 4m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/685b30f61d07450a810c30c5243e8495 : SUCCESS in 8m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6d49d83026be42308bd415da6e26ae7a : SUCCESS in 12m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/40e8a1712b894939bf026b8d71f89a52 : SUCCESS in 9m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f99d3faf865e41cc8d7ba27155658753 : SUCCESS in 6m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/397b521970624912b4c30b77bfc2712e : SUCCESS in 12m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e67fc74fabcc461295fcd89fc6a3b574 : SUCCESS in 8m 55s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8e8a8b30e93049d2a7303a0fecd2c0a2 : SUCCESS in 8m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1f2417beec8143a6b85b06eaabc5e7a3 : SUCCESS in 41m 57s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6391070971314e2997267ae831b94ce8 : SUCCESS in 24m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/bedf3bf449d84aa7a6481f35e7693585 : SUCCESS in 14m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3fbb60203849429291d8354d91f56c27 : SUCCESS in 5m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e8fb1785fa804fdf812823ccdd3e387a : SUCCESS in 18m 37s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/43af8fb09ecf4e9bb6473f4dfd39df73 : SUCCESS in 5m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/83f5b0b16daf4901bdc086be5f653a98 : SUCCESS in 7m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/87e97b069e644c5cba4d9f8aea437d87 : SUCCESS in 13m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4203e56915b34bfa8d05d98413126887 : POST_FAILURE in 39m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/66585ea9586e4b578d7fe51c8b9d852e : FAILURE in 31m 44s","accounts_in_message":[],"_revision_number":3},{"id":"492c79a8a623ecf339431d169bf1fec171806c7e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-04 10:58:05.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"c897b9dabb259ccc109e3d6dc5f111353dcf2d39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 11:43:03.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e85e1d41dd1a48529973d1b7bd560125 : SUCCESS in 14m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c411c5cb6fa4cb48afc77c0749234b4 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0324efcdc1d8469e8e9cc8400da16cf1 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f30cff7d8fb24903943b9909dfd2c137 : SUCCESS in 4m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/03351fd848ae44dbbe88e35956959dc7 : SUCCESS in 4m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f3c3b4e5e85347ce8e5fffafb92ceaf2 : SUCCESS in 4m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/014b31f77df04dbbab63a035fe0c1d5e : SUCCESS in 9m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bb1f107d26134ce6a9c1127820722c7c : SUCCESS in 8m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/bd55b2dbb4954159848d8a1ca7028779 : SUCCESS in 8m 22s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7123ee76feac49c1990a2eb94bf6ca53 : SUCCESS in 7m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/51504068c8364c6d9991d3a3607f870e : SUCCESS in 11m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/62b3736f1e634de1bb1504cf66fcf567 : SUCCESS in 7m 26s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d4e97bd3a8194b619f726fe881fedca1 : SUCCESS in 8m 20s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8c175953aead4511ba0f1443acc92895 : SUCCESS in 21m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3781b9afc9c24bfa87dba09aee0e0955 : SUCCESS in 24m 18s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cc878394222d44099feda167c0b5cbd2 : SUCCESS in 15m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0a19835eab754f04a294805aa022e250 : SUCCESS in 4m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/22d74c0c621e45ad9bad75bc76450360 : SUCCESS in 19m 42s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2df095d484c74a118dbc88ac176b70f9 : SUCCESS in 5m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d1fc4fdf0f9749ad8cc3bbeecffe8aa1 : SUCCESS in 6m 42s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/969af34add304b73a01a7146e0b25b0f : SUCCESS in 15m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d6ad0519f252426dbb578dcd93a4c9a6 : POST_FAILURE in 34m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/67036d9090ad497282bc7e4a648d61c4 : FAILURE in 29m 56s","accounts_in_message":[],"_revision_number":3},{"id":"b9f8e98e704803e5b3af4698ca62057f1ece7f37","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 12:11:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/76dd33fe92104a7bb56344f1a45f3d42 : SUCCESS in 1h 08m 27s","accounts_in_message":[],"_revision_number":3},{"id":"b96e1683b596a19d4729073b9c73b64361e0c014","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-09-06 06:19:10.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5dbcf09ddabc40407191a3aa9defdb445db5c760","revisions":{"3a315914974c41be4fac8b0b444dec479ae70f19":{"kind":"REWORK","_number":1,"created":"2020-09-02 06:29:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/42/749442/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/42/749442/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/749442/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/749442/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/749442/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/42/749442/1"}}},"commit":{"parents":[{"commit":"10ebc68a49dfb132f49c1662e16acb5f55090043","subject":"Merge \"update constraint for oslo.policy to new release 3.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/10ebc68a49dfb132f49c1662e16acb5f55090043"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-02 06:29:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-02 06:29:22.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I4b33930856be9b05be47fc8599dad30eb8bbb9d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3a315914974c41be4fac8b0b444dec479ae70f19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3a315914974c41be4fac8b0b444dec479ae70f19"}]},"branch":"refs/heads/master"},"1f69b625b6fd5c84aa0ecfb8c107f7efa6e483f2":{"kind":"REWORK","_number":2,"created":"2020-09-02 07:38:27.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/42/749442/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/42/749442/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/749442/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/749442/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/749442/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/42/749442/2"}}},"commit":{"parents":[{"commit":"10ebc68a49dfb132f49c1662e16acb5f55090043","subject":"Merge \"update constraint for oslo.policy to new release 3.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/10ebc68a49dfb132f49c1662e16acb5f55090043"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-02 06:29:22.000000000","tz":0},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-09-02 07:38:12.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I4b33930856be9b05be47fc8599dad30eb8bbb9d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1f69b625b6fd5c84aa0ecfb8c107f7efa6e483f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1f69b625b6fd5c84aa0ecfb8c107f7efa6e483f2"}]},"branch":"refs/heads/master"},"5dbcf09ddabc40407191a3aa9defdb445db5c760":{"kind":"REWORK","_number":3,"created":"2020-09-02 12:50:56.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/42/749442/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/42/749442/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/749442/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/749442/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/749442/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/42/749442/3"}}},"commit":{"parents":[{"commit":"10ebc68a49dfb132f49c1662e16acb5f55090043","subject":"Merge \"update constraint for oslo.policy to new release 3.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/10ebc68a49dfb132f49c1662e16acb5f55090043"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-02 06:29:22.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-09-02 12:50:40.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I4b33930856be9b05be47fc8599dad30eb8bbb9d6\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5dbcf09ddabc40407191a3aa9defdb445db5c760"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5dbcf09ddabc40407191a3aa9defdb445db5c760"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
