)]}'
{"id":"openstack%2Frequirements~749644","triplet_id":"openstack%2Frequirements~master~I61d1e7170a9df31f68d1feaeef7c5c2d6685336e","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I61d1e7170a9df31f68d1feaeef7c5c2d6685336e","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2020-09-03 06:22:16.000000000","updated":"2020-09-12 17:16:43.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c62af2fe6e06a8f6f438a82b90f6c3e2af0e3297","_number":749644,"virtual_id_number":749644,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-09-05 19:17:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":-1,"date":"2020-09-11 14:29:49.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-05 19:17:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-11 14:29:49.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"9bcf40ea9e22da93f2e6153f5d11a97abcb670ef","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-03 06:22:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a606cfb8f37eb7e1d7469ae439a9dccade43fe2f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 06:44:04.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a861a45bd8db4e75a621c7db1265fb0c : FAILURE in 15m 28s","accounts_in_message":[],"_revision_number":1},{"id":"67654f54cbc899c5cc5f1421be753fc00bcd8a80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 07:01:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b17b2b494a9040c88ed84289742b2884 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3108b207abc64d08a7fc956732cd7425 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d7634114ca24cc58c7e3d1ef27e1d92 : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1054bf77d544dd7b229d8f7bef93de1 : SUCCESS in 6m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9131f57b0e344e6c8dec59abd00eb4b9 : SUCCESS in 5m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2796d6d58689494c8263e5c63fdad3de : SUCCESS in 4m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/48a6a129e2f441d9807a2bc0e4326023 : FAILURE in 8m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c2f3b7f4382a4d46b5afb595c64c84b5 : FAILURE in 8m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/33645d7c1ecd4b54abb6e30d2a654593 : FAILURE in 11m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e8b2ecdf72214721a704fa47c6b769ba : SUCCESS in 6m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b15307c3be1e4e9bb1fcd41146d312d1 : SUCCESS in 12m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/575b476105f6493e8aaf2d32f93d2c7b : SUCCESS in 9m 07s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/97d01012360f484090453655324b27e8 : SUCCESS in 12m 28s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f8b76262644b4bfaa192a74a678f2f07 : SUCCESS in 24m 41s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f35f1b53292c41c4b1f25d97d0e69b51 : SUCCESS in 26m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6e1c9961dae140c691f72f5ae0d278f5 : FAILURE in 16m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/13546a8b59e04d668d995ea3c9ad66cb : SUCCESS in 5m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/94fc64b1540a4a61b933dfcc2b8e00cd : SUCCESS in 18m 03s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e5a6b5afbe784b1986e3bc3a9a00b0e6 : SUCCESS in 7m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/67041b27383049b0930084f85a36764f : SUCCESS in 8m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d421133717044d049054897052fd766c : SUCCESS in 12m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/537d4c041b1743c2b846cf3807a808ee : POST_FAILURE in 36m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e4502b82aa654d109b377d55edb630f5 : FAILURE in 29m 15s","accounts_in_message":[],"_revision_number":1},{"id":"5149d4cd52d7d680b47ac50c0a9c5fd994a2ea56","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-04 06:24:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8547f2cf06d17b250e3384d7bdab14df5249ecdf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 06:44:33.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/611a0b4fe1454b68a804055c175cedff : FAILURE in 15m 33s","accounts_in_message":[],"_revision_number":2},{"id":"a237badaa51e8e1f710c1ec9f01803ce02d7edea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 08:42:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c35a912ba7164aa2aa0c6cf8305adf2d : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de9bb42f65f047108b25ad13c96b7be9 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e11104f3ae7a4e57a933e877d405e45a : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26fabfedc20a4ac88143ad03eb51d26c : SUCCESS in 5m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/62db43cc79634933b625db47df9afdeb : SUCCESS in 4m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7cbc856664a24b899d819ca487b7b046 : SUCCESS in 3m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/310c1d19e1604239a411b84560c547b6 : FAILURE in 5m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6023bd2c23ca4400a6b9f2981cb5e197 : FAILURE in 7m 07s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8225352dc72f41cca868d73ec95527ca : FAILURE in 10m 46s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b459b874a7e24bb0bc52ed927db9b203 : SUCCESS in 6m 11s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e2bb5f7863214049b28331daa6bd0db0 : SUCCESS in 10m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a02611c083e34cfbb71406d2f6f980aa : SUCCESS in 10m 09s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/29266404b7874a8d8303ddeb3476497a : SUCCESS in 14m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1ba556fe93984f9882e50cd03814bdcb : SUCCESS in 14m 00s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/11cc5378b2f44c8b9d26f6e3aef4f079 : SUCCESS in 31m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3c96dcdbff5c4bf19a59d96c701e9911 : FAILURE in 12m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4566a9721c8c47329fee055520015934 : SUCCESS in 4m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3eb3db72b0644efbbc56e4b435e5ef69 : SUCCESS in 21m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c6de7ae7af2e44749ada77772e3560bf : SUCCESS in 5m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/22ae64477c734297ad2b6d6ac81109cc : SUCCESS in 7m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/87d67b96893e488dbdf880df4eacbe49 : SUCCESS in 10m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4742c6f3179946c68fedb765cc569f02 : POST_FAILURE in 39m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ae8ba4cf658f44d89de113054d3df90f : FAILURE in 36m 12s","accounts_in_message":[],"_revision_number":2},{"id":"ba18aebba316bc8397ec6b164b8384f4f1e35e01","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-05 06:18:25.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"391d1608e9d876306a1fe1e1427a5e0f0c89db2c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-09-05 06:24:25.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6a07bed984667c1352b10f74a80f6d941c271126","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-09-05 06:24:26.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":4},{"id":"9fcd1783531d27ea3f9bf9d0e726dbafcb3090c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-05 07:14:50.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e33322021aa54ec5bddc98bc724fa7cc : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63715921d7ad4f87b9952ef814400d54 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0ceb243ba274d72b6d133022bccdc28 : SUCCESS in 7m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71b765a76fd5499ba6b20f703f657876 : SUCCESS in 5m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ccd6e1a384ad49cb8bd2b7b7c2d3f893 : SUCCESS in 5m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/875f3ed675b04d35b8f40aec60c5a2c6 : SUCCESS in 4m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3727a78989984db1bb6716df3f98a724 : FAILURE in 8m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/062e5adaff544be3a206f7dcf0cd4a48 : FAILURE in 12m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/15163e0691ca445b9660123132c78e94 : SUCCESS in 11m 35s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6b78d472d81740808643e7729e054203 : SUCCESS in 6m 23s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/216b136df78042a9a1704d2b08532fc1 : SUCCESS in 11m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/81374dd68d114d16a7afc6006ebe49da : SUCCESS in 7m 12s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1b7a4e039dc14de691bac8773e42a8c7 : SUCCESS in 9m 04s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1c97151aa8284152a2fb9e418eb2a076 : SUCCESS in 13m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9354e0a793ad40d2b5a8653cd2f4fbeb : SUCCESS in 25m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7c228703a8564599b254f8ef20dc329c : SUCCESS in 11m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9d57a82e8f66425996458c4ca10c742c : SUCCESS in 3m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/208124ac8a6144678332f92550fce9c9 : SUCCESS in 16m 43s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/63ba137c34a24232a0e1b4dc7ba1b0c0 : SUCCESS in 5m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bbee1bb5a23f4146a672808b9fd12947 : SUCCESS in 6m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/85ba35ad887a41e1a3ac626ae9e43dc2 : SUCCESS in 15m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/50ae3708f13c4526bf3356895c0749ea : POST_FAILURE in 33m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9fbadcd10ddb4f8b90ee79f6ed41f099 : FAILURE in 32m 25s","accounts_in_message":[],"_revision_number":4},{"id":"49f217074fbcf0a32e49a7503a134382b23255bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-05 07:45:18.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7f3257549e484f8cb6e7db1b59df1bb3 : FAILURE in 1h 14m 11s","accounts_in_message":[],"_revision_number":4},{"id":"8c0b6d8d02504ee966771688180b3b9cf825fc60","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-09-05 17:19:31.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5218181aa87ec6eb0fbf85b9b6ffb9b9c95cdb4e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-09-05 17:20:06.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b91e0453335e704a1000b18c981b4636af0beb67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-05 17:52:14.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cb54d12a9f246f883079e888625d812 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0adee15e28a4b30ae9e478d9068f9ee : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa47ceec41c5439bbc84b7de729cb6de : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c035552d1404bdea29b888585c44032 : SUCCESS in 5m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f9ef4492cd124f6d9bd176e781b32996 : SUCCESS in 4m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/91b0450ba1df45e39d7c7aa725686f9f : SUCCESS in 4m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/df12f3dc219d44e1b46542ea2cb9cc58 : FAILURE in 10m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/eb7b61866b164ac8a9f315e2aab1d7d1 : FAILURE in 10m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fba6bf56f848408cb4cda41d047f5199 : SUCCESS in 10m 16s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f9de15c28d8a4919a3f113501d3f8c50 : SUCCESS in 5m 49s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/93dd0532ebe2430bb51280b2db93a610 : SUCCESS in 10m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f580b87c735c457dbcc79308aa87faf0 : SUCCESS in 8m 05s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3a80afeb237f463e99e9909354b1aece : SUCCESS in 7m 00s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f7ea53dd3b8a45b1a15c42490143dbf3 : SUCCESS in 18m 07s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1c92d4254f494cd3a61e8ad2d44ab6a6 : SUCCESS in 16m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1ea2a2a969b342abbeb666c1d2a9495d : SUCCESS in 15m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/75e1140283f54d948956f7285138a996 : SUCCESS in 4m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0bfc12131129479e8e0f53987c4c9e29 : SUCCESS in 17m 43s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/29b304b4aeb14b57be35335a9044e3a5 : SUCCESS in 5m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b009d22e6e80423e8f703385c0aec32a : SUCCESS in 7m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3297e478314a42f9a2191f5605f88ed3 : SUCCESS in 9m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/171c8386a8c840bf8bb77214aa844103 : POST_FAILURE in 30m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ba057c005ee3428f8f56f0715d05ddf3 : FAILURE in 25m 11s","accounts_in_message":[],"_revision_number":5},{"id":"5858c3b84b195af50bab9ab2360ebfb3406cf96e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-05 18:28:29.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/670c90c260c6481db818191703bc64c8 : FAILURE in 1h 04m 12s","accounts_in_message":[],"_revision_number":5},{"id":"d6deab2cb4506c9808b339c835986cb89d5e31c6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-09-05 18:34:41.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"81a18c5ef1f96f58f645920d76d969d3922b9a30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-05 19:17:15.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09d34502a5f9402f87aaba1fe1da5b74 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc0f37f9bba748c0accebd2f5860f7ec : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5479f3dab93a4eb6b6268a873855903d : SUCCESS in 3m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f466d9c21cdb4ad29199aa81317c0a97 : SUCCESS in 3m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9a974dee555b4d15a9844621de444200 : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/983e2a28fa14413085704c5c243b7244 : SUCCESS in 2m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/30c899410304450d8d16935b6f02907a : SUCCESS in 6m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5f50ee9318084ab3a44633d09bb77e97 : SUCCESS in 7m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a69218d073764495b2d431208d8c1902 : SUCCESS in 8m 12s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/886e484caf0a44beba0c8541b696e246 : SUCCESS in 4m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c4a58912e0a6459cabe064645f2ab8b6 : SUCCESS in 10m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c8a95f180e774a48b64db8387eaec9a0 : SUCCESS in 5m 15s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/54545c98e09b4d2388ff1f9d10c04cb4 : SUCCESS in 5m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/adfc2957c14a435a80808106413c39c9 : SUCCESS in 12m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/25e348fcacd942348413a5fa81690b4d : SUCCESS in 24m 21s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/da2482739ad543a2bdf5715515bfa73c : SUCCESS in 16m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9012a5c3a9fd41fea4633819c9ff7951 : SUCCESS in 2m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e65559017be34416a28f50b8bf042837 : SUCCESS in 17m 40s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d587178fc6694b558ba692251fb6d7b7 : SUCCESS in 3m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/efac399c76814ff7b3220a7751c22de1 : SUCCESS in 5m 36s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ee30344a071f4ea181cb2f7c82aaf842 : SUCCESS in 9m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/be1404e7a38844f2a0dd2cc9cecc72d9 : POST_FAILURE in 36m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a0860f6a12bc4860ae7d8e997e686bcc : FAILURE in 28m 59s","accounts_in_message":[],"_revision_number":6},{"id":"6b19139d25211a97c67b52c6064dd273341eeb8c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-09-05 19:35:31.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"b17ff1adc33a96c493edb6740610b0efb4d736fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-05 19:42:48.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/99b8ae40d5e94194ad10b6b6d82c6fbb : SUCCESS in 1h 04m 40s","accounts_in_message":[],"_revision_number":6},{"id":"8956af247f191a3f12fcf4e71537c0f105816ef5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-05 20:15:41.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8871e112555c4e5aa168d19cc188bd29 : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4ee8170909f4c0582c19375ba0fd6de : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/244e7c37ea644feabb78780f6ae1df84 : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be4cb23d96b24923adb4d0281fc7dd4e : SUCCESS in 3m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7cec8e1082af48ec8fd81efb44e5c854 : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cc48c29d73414cfe8f459e83ef0fa1ce : SUCCESS in 3m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3281495fee9a455d9129f810c0e53789 : SUCCESS in 7m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ccbd42cca0844080a70c28eb3be41eb5 : SUCCESS in 9m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/458febbfae2e42e1b274eb2c2b5a31ad : SUCCESS in 7m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e421b5d521d84b89baff7823ca44bd85 : SUCCESS in 4m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/467295f0af1043a1a7f6d5765f8328df : SUCCESS in 10m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/48adef31ad234ffdb735560e412440d4 : SUCCESS in 5m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3987d3fd3a95416caa45b5132bf0d3f7 : SUCCESS in 6m 29s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a25b65fbd48547cabcbcda161980eff1 : SUCCESS in 15m 37s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0e53ff0d37554605913492caf4bc27c5 : SUCCESS in 29m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ed698d5a02704a35b95ecb9ff70f5fd8 : SUCCESS in 37m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a6e62f08c7f24d77a888e76bc3723ba1 : SUCCESS in 3m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/71d4d2586bd146ca9e85220e40b1d6c3 : SUCCESS in 34m 36s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/bf81bb6b52b4485e845ff1d159482945 : SUCCESS in 4m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ad68c9ec2c344c518f30f237db0f4b7c : SUCCESS in 12m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b13f005b549e439093f764b033946426 : SUCCESS in 8m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6a1fbe5110574da79b5540304b4a30fb : POST_FAILURE in 27m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/df1408d1cdd64a399c4304f6da4b91c2 : FAILURE in 27m 23s","accounts_in_message":[],"_revision_number":6},{"id":"2a1cca54718ba98e0643e1ed85cc694684160f3e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-09-11 14:29:49.000000000","message":"Patch Set 6: Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"c62af2fe6e06a8f6f438a82b90f6c3e2af0e3297","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-09-12 17:16:43.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"7af5eb154331eef712c1b25e9a5d9803ad5d1a51","revisions":{"5e68a49373cd68dd2cc537512b47e286770f242b":{"kind":"REWORK","_number":1,"created":"2020-09-03 06:22:16.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/44/749644/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/44/749644/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/749644/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/749644/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/749644/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/44/749644/1"}}},"commit":{"parents":[{"commit":"10ebc68a49dfb132f49c1662e16acb5f55090043","subject":"Merge \"update constraint for oslo.policy to new release 3.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/10ebc68a49dfb132f49c1662e16acb5f55090043"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-03 06:22:11.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-03 06:22:11.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I61d1e7170a9df31f68d1feaeef7c5c2d6685336e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5e68a49373cd68dd2cc537512b47e286770f242b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5e68a49373cd68dd2cc537512b47e286770f242b"}]},"branch":"refs/heads/master"},"81e67963dac42f5883bdaaf30831c6920aefcb42":{"kind":"REWORK","_number":2,"created":"2020-09-04 06:24:14.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/44/749644/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/44/749644/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/749644/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/749644/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/749644/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/44/749644/2"}}},"commit":{"parents":[{"commit":"943fab85a7947be4f3fc866ca9309c6d5f7fc174","subject":"update constraint for ironic-lib to new release 4.4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/943fab85a7947be4f3fc866ca9309c6d5f7fc174"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-04 06:24:12.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-04 06:24:12.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I61d1e7170a9df31f68d1feaeef7c5c2d6685336e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/81e67963dac42f5883bdaaf30831c6920aefcb42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/81e67963dac42f5883bdaaf30831c6920aefcb42"}]},"branch":"refs/heads/master"},"050d9f85ebc35c8896292c86be13b51276bf737d":{"kind":"REWORK","_number":3,"created":"2020-09-05 06:18:25.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/44/749644/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/44/749644/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/749644/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/749644/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/749644/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/44/749644/3"}}},"commit":{"parents":[{"commit":"d2c8b9d472320a8e916f38df4580a7f7bbab2bda","subject":"Merge \"update constraint for ovsdbapp to new release 1.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d2c8b9d472320a8e916f38df4580a7f7bbab2bda"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-05 06:18:21.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-05 06:18:21.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I61d1e7170a9df31f68d1feaeef7c5c2d6685336e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/050d9f85ebc35c8896292c86be13b51276bf737d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/050d9f85ebc35c8896292c86be13b51276bf737d"}]},"branch":"refs/heads/master"},"889156d40496fb1397fa7af1410b59dd5ceb0020":{"kind":"REWORK","_number":4,"created":"2020-09-05 06:24:25.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/44/749644/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/44/749644/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/749644/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/749644/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/749644/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/44/749644/4"}}},"commit":{"parents":[{"commit":"d2c8b9d472320a8e916f38df4580a7f7bbab2bda","subject":"Merge \"update constraint for ovsdbapp to new release 1.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d2c8b9d472320a8e916f38df4580a7f7bbab2bda"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-05 06:18:21.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-09-05 06:24:01.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I61d1e7170a9df31f68d1feaeef7c5c2d6685336e\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/889156d40496fb1397fa7af1410b59dd5ceb0020"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/889156d40496fb1397fa7af1410b59dd5ceb0020"}]},"branch":"refs/heads/master"},"33d73cc90984e34a25c37222e806df8b71414916":{"kind":"REWORK","_number":5,"created":"2020-09-05 17:20:06.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/44/749644/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/44/749644/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/749644/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/749644/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/749644/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/44/749644/5"}}},"commit":{"parents":[{"commit":"d2c8b9d472320a8e916f38df4580a7f7bbab2bda","subject":"Merge \"update constraint for ovsdbapp to new release 1.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d2c8b9d472320a8e916f38df4580a7f7bbab2bda"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-05 06:18:21.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-09-05 17:19:54.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I61d1e7170a9df31f68d1feaeef7c5c2d6685336e\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/33d73cc90984e34a25c37222e806df8b71414916"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/33d73cc90984e34a25c37222e806df8b71414916"}]},"branch":"refs/heads/master"},"7af5eb154331eef712c1b25e9a5d9803ad5d1a51":{"kind":"REWORK","_number":6,"created":"2020-09-05 18:34:41.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/44/749644/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/44/749644/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/749644/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/749644/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/749644/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/44/749644/6"}}},"commit":{"parents":[{"commit":"d2c8b9d472320a8e916f38df4580a7f7bbab2bda","subject":"Merge \"update constraint for ovsdbapp to new release 1.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d2c8b9d472320a8e916f38df4580a7f7bbab2bda"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-05 06:18:21.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-09-05 18:34:22.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I61d1e7170a9df31f68d1feaeef7c5c2d6685336e\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7af5eb154331eef712c1b25e9a5d9803ad5d1a51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7af5eb154331eef712c1b25e9a5d9803ad5d1a51"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
