)]}'
{"id":"openstack%2Frequirements~749804","triplet_id":"openstack%2Frequirements~master~Id6c0b808efee02571e235651d18aa3322761bce0","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Id6c0b808efee02571e235651d18aa3322761bce0","subject":"update constraint for os_vif to new release 2.2.0","status":"MERGED","created":"2020-09-03 17:27:46.000000000","updated":"2020-09-04 07:37:55.000000000","submitted":"2020-09-04 07:35:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749804-1599204918789-1545f216","meta_rev_id":"af0b314ee6643775c7a88faa9cc82bbd3d76c317","_number":749804,"virtual_id_number":749804,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-04 07:35:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-09-03 23:46:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-09-03 23:46:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-03 23:46:29.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-09-04 07:35:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"db25a627ca5013b9db5b702697310307e2e3bcc0","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-03 17:27:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5d9f7cc4ec7f0f0ea6c47c41f76636c0376f1e3d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 18:35:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/af259343ed0a449f8fe2d31c854bab0e : SUCCESS in 1h 03m 03s","accounts_in_message":[],"_revision_number":1},{"id":"3a1dcbba1772c48c957d14291784fa068c0bd677","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 19:59:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ca1c563bd944c669f39f869ac44dc21 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8b22941ed854360bd3de1dba61bbe48 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b51edba8d1e84e6098bc7dfa6aade5f1 : SUCCESS in 7m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1dd9eb7a17de4fbea5c1839d66415c5c : SUCCESS in 4m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/506c7ccc4361426d839a1d59ed891559 : SUCCESS in 6m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/573383386a6b4d2f8e072f08bf87f260 : SUCCESS in 4m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/82aa72eb7f814ca49ccbcb2e707b05c3 : SUCCESS in 7m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/50ccce8c579c446597870c8c5f3c8608 : SUCCESS in 9m 48s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/831539c329f543a2a1947c2828ed4b1d : SUCCESS in 8m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4c2db29e84c1405cae46cce176770fd3 : SUCCESS in 4m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3edda11cb5e34b4bb5eb950b2b8158b1 : SUCCESS in 7m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1dfde9bafcca468bab7d61ca5785eb26 : SUCCESS in 8m 05s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1497bb7b2c3c41628d791468b8739d3c : SUCCESS in 6m 54s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c08585dc2d0e4057b001571728e60397 : SUCCESS in 42m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6fbb7404ccb449ae86473fb36011e1d2 : SUCCESS in 23m 41s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0192ebfe65ce4ab5900ed9c584014c91 : SUCCESS in 13m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bda1fc388675409182a488ad1b3b9698 : SUCCESS in 4m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3ab328f10cc14638b728d4e8dd6422dd : SUCCESS in 17m 37s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/196a21b21e52438296ee4e72b173e963 : SUCCESS in 6m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3f016507315543ff8985af2a1d9dd6fa : SUCCESS in 7m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ef10cff3a6254330a7d6229aec8471ca : SUCCESS in 10m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8fa607daec1d4f6b9e45302e455b3bf3 : SUCCESS in 59m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/49e819bf9ffc44f7be2fa5525b7e286e : FAILURE in 1h 10m 59s","accounts_in_message":[],"_revision_number":1},{"id":"5f11c4b7b28aad36e56679386cdb9c1d18e3e150","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-09-03 20:29:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9a454c75d8f4d86e657ec888784d05a387810000","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 21:40:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ebb6308dff3b4c54824c3e2457bf3a73 : SUCCESS in 1h 04m 13s","accounts_in_message":[],"_revision_number":1},{"id":"42147d25f93914987f723850edca643deaeaf690","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 23:06:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4461808e54447299485ba0e62b769a8 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7292345b1d8495cb75e362ae59c3ae1 : SUCCESS in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b9d0fd4889441b4be8439472adbe5d3 : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20b4addae499462d873eed8f8ca37767 : SUCCESS in 5m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ee15261f175e448791810bc132efc6c5 : SUCCESS in 6m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6d3f8414f0f44a758e15a20477142d5b : SUCCESS in 6m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f705999e1f684006b51bc55b0ed8d0b0 : SUCCESS in 9m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cc1b2c6fac464c819e03b9fca4f9bafa : SUCCESS in 12m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4e028f08443444c3b41a0779f16000bc : SUCCESS in 7m 40s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/999f91f072ae44e797ecdcfca1f37cb0 : SUCCESS in 8m 16s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/640157ec785d402699ddde36174499f6 : SUCCESS in 12m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3c97d351741442f698f0a549adbbca28 : SUCCESS in 10m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f16c1084b796424b91b282fca5918d8a : SUCCESS in 9m 57s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/68c810f14c6144c2ac6724af122b489b : SUCCESS in 29m 21s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/495095c965ad4541829d5fa877d74ba4 : SUCCESS in 26m 49s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/dfcc6ba08130429f916bbe07ef69c773 : SUCCESS in 12m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9cc9a7e3523d4406b6e71680fde6d537 : SUCCESS in 4m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/627c434a272f4d3c9d9e81ba03dd4dcd : SUCCESS in 24m 01s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3088f57376b840dc8086d5de633bb079 : SUCCESS in 5m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/717eccd5db004a50b60a97f4586d640e : SUCCESS in 7m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8b356f2c04684667a9bdfe9255a72ce3 : SUCCESS in 11m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f9490eaf64774eb1ae1fadf977138e59 : SUCCESS in 54m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bcc2fbc06fed47d9a2a7827dd75f4ce3 : SUCCESS in 1h 14m 33s","accounts_in_message":[],"_revision_number":1},{"id":"b12115f49a34d2377a7b3a4a33e0085b9024ede0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-09-03 23:46:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0a964ef5a9fab7475ec13844303fc0767d3efeec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 23:47:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"526d01c0edd31c4412e7f38278c053f1bc3e9cab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 07:35:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14879c16be8f4611ba6474be891aebd1 : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f751ee8c7a3c47e6a6c9e3fff6a59256 : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7cfcc9a957b240f0894af4709e37b68d : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7195e843d19c4744834e44f12f41ab0a : SUCCESS in 7m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/024a79c1bc7142c4b20335b1a9c7c182 : SUCCESS in 5m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e61633c482c04543901bcac1f94f34ed : SUCCESS in 5m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e9e5b4a0f620441393244fd8ddf06eef : SUCCESS in 9m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a6621c7bf7b94011a8cc407aa0364e5d : SUCCESS in 17m 44s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e34baf1247374f49aa1e5457b3aed06c : SUCCESS in 15m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e3637467470f4bef907697f9a291257f : SUCCESS in 9m 11s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fd6ad1528b5740bd918311ceeb6813ae : SUCCESS in 12m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0f689fe026224a1a8e0b88f54dce6843 : SUCCESS in 9m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e4672292f83b4261979d3d671ff5316e : SUCCESS in 15m 38s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/bede129c9ef742a6879892865bf1b9d4 : SUCCESS in 26m 35s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bb118ac272e04f789e307c0e2536e060 : SUCCESS in 29m 00s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8611f27b0fc942b3972b6925367e9302 : SUCCESS in 23m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d6b92049bedc4d0e8781905135fa7bed : SUCCESS in 6m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fc6842d76a374ac8857d934590a87423 : SUCCESS in 22m 46s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a868f68f9eb7420fa850f8d6501a41a2 : SUCCESS in 7m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f1b22aa41825415a855e2aeb88b6e4de : SUCCESS in 11m 02s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f545f59ddd774482afcb0d65ab3a753a : SUCCESS in 19m 36s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e5999b73b255431f91d8959dab5f6f7e : SUCCESS in 20m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/088d759d80da4fb2bff930b237d4cec5 : SUCCESS in 1h 07m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0fc0b1a163e94dcc898b7dc5ed66aa11 : SUCCESS in 1h 34m 30s","accounts_in_message":[],"_revision_number":1},{"id":"7923f2faa986f56e12d7075b824479054f855334","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 07:35:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"00ec0fb5dc756a0ca503244dfec6c94fcc308dad","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 07:37:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c47e9e3f81d749c2b00e7e9cea004d4e : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e194030f1512deef8667612ed3edea1fd282fb9d","revisions":{"e194030f1512deef8667612ed3edea1fd282fb9d":{"kind":"REWORK","_number":1,"created":"2020-09-03 17:27:46.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/04/749804/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/749804/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/749804/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/749804/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/749804/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/749804/1"}}},"commit":{"parents":[{"commit":"10ebc68a49dfb132f49c1662e16acb5f55090043","subject":"Merge \"update constraint for oslo.policy to new release 3.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/10ebc68a49dfb132f49c1662e16acb5f55090043"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-03 17:27:44.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-03 17:27:44.000000000","tz":0},"subject":"update constraint for os_vif to new release 2.2.0","message":"update constraint for os_vif to new release 2.2.0\n\nChange-Id: Id6c0b808efee02571e235651d18aa3322761bce0\nmeta:version: 2.2.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta:release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta:release:Change-Id: I1eeaa2e28d2b74c596f715c84f89bc91050509e8\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Balazs Gibizer \u003cbalazs.gibizer@est.tech\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e194030f1512deef8667612ed3edea1fd282fb9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e194030f1512deef8667612ed3edea1fd282fb9d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
