)]}'
{"id":"openstack%2Frequirements~749905","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I5586495104c3de42067ab8f6f353314b54a1d135","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"I5586495104c3de42067ab8f6f353314b54a1d135","subject":"update constraint for os-brick to new release 2.10.5","status":"MERGED","created":"2020-09-04 08:59:54.000000000","updated":"2020-09-04 22:04:02.000000000","submitted":"2020-09-04 22:04:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749905-1599257042735-9d651cf1","meta_rev_id":"e484b59652624c7f52a054629f557adf1963b718","_number":749905,"virtual_id_number":749905,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-04 22:04:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-09-04 14:30:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-09-04 09:06:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":2,"date":"2020-09-04 09:04:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-09-04 14:30:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-04 09:04:40.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-04 09:06:34.000000000","updated_by":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"reviewer":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"state":"REVIEWER"},{"updated":"2020-09-04 14:30:40.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-04 22:04:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8bd81e96e341ed915efa0ff8c4135bff64645751","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-04 08:59:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"226631b4af63ffad1424cde7187cb78e3954f22a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-04 09:04:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ac407e6a9c8d4696aa9f6a18c33ef933a6a69a73","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-09-04 09:06:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"efc1c595b5b762fcd728b2f3650cb43ae6cadc67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 10:43:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f259d90465f46e5a3701132d85069be : SUCCESS in 3m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ef7bf3adc9ea48dba1848eb84701ea53 : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c47a1f95b3c42e398d5afae749c1490 : SUCCESS in 4m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1f8e5a1b162b482c9807d2ab521f18de : SUCCESS in 11m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/218846e81cd34c3fbb2f7b382dd33978 : SUCCESS in 4m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dcbe81bc015444d397b8b8ae7e04a5d9 : SUCCESS in 4m 41s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/6e96fc25a657457cb37cb9dfb91227ec : SUCCESS in 7m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/678013b0ff964d04be7585939ab8169c : SUCCESS in 9m 53s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/a482e40ba81f4daeaedd8b9b4136610c : SUCCESS in 9m 20s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4b94ebbf21cf47dda669e32e31333a1c : SUCCESS in 8m 40s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/a8a805ccf56b408a98c2eebb019a9621 : SUCCESS in 5m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c97ef7032948478a8d28e691af6190d0 : SUCCESS in 5m 20s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/572c23ad45bf4102922ca5c055b41a65 : SUCCESS in 10m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0bf6a1cb2fc942e88210a341d64d4b13 : SUCCESS in 9m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8c6de896f1134906af7c439489bb75ad : SUCCESS in 8m 14s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/fa4e039b81f643e38c64c486976255ff : SUCCESS in 12m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5bc6984d55d54f7e976d1742b343092e : SUCCESS in 12m 32s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/be2d1f1b12d94a79bb0d6fa45df7b1ee : SUCCESS in 21m 51s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e6ef3d990a264d339385d2f05fe8fa66 : SUCCESS in 21m 30s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/9c6bc6d59ae54a0fb4db641bbfeb2cb5 : SUCCESS in 12m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a7937669262b4bc4ab4592f3f5f98bbd : SUCCESS in 13m 27s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/1352f4626fc340029ee8047d3b85a5b5 : SUCCESS in 25m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6f3455e085c9494b8ff4ef1d28dc500a : SUCCESS in 19m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cc31271335d646348196d4b17a52da66 : SUCCESS in 5m 46s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9347e9f9b9e14a718e07d1d67762d04a : SUCCESS in 9m 03s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/84703f9cdddf44afab0c663df1134938 : SUCCESS in 9m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1ab6db5e1e0041dab436ed98506c47dd : SUCCESS in 1h 09m 18s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4b6d1c29483e472f91f3d154a2527802 : SUCCESS in 1h 30m 14s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9e345741f3174b05aa0e12b4852ab952 : SUCCESS in 1h 07m 50s","accounts_in_message":[],"_revision_number":1},{"id":"24af952b00f02518be438e6a7f560409776c00de","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-04 14:30:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"93eb435715a8e545a16d94bf6e9025e8a4e86182","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 14:31:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"712832fbaa2c01bb2c0f116d23803b971c9761d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 16:44:31.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88a4bb17c2af48e28111924523d5f7cb : SUCCESS in 5m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ed06d9d2e61c41939b9422d62a7b6867 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ef0a04f2ec64893982fe81c841fe606 : SUCCESS in 5m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/611b6f54b6484c52a2947e5adef96b8f : SUCCESS in 6m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/669d5d8e7f354949b0e33b0263f423fa : SUCCESS in 6m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5f596605986d43ef890301c73c1b671d : SUCCESS in 4m 53s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/999728d5834b4f24acf77477ff66c230 : SUCCESS in 11m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/640284e488524d71bfbb3eb8dbb24f13 : SUCCESS in 8m 06s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/d16d512d72a645bdb88e43cea70efbd2 : SUCCESS in 8m 14s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/20205bc2dca844b186b01dbcde1c145f : FAILURE in 8m 31s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/9cd0bca9767c45098b126aac5a44f5a7 : SUCCESS in 7m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3697ef975a3b4a38a0cad8cab3ca174c : SUCCESS in 6m 54s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/0e67af310924424b8b808858c853d177 : SUCCESS in 10m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c9d41e7561564d06b11295fe5d1c8866 : SUCCESS in 8m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b3a3f637c97144aebd70418bfc94e375 : SUCCESS in 7m 13s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/0464dd5bc91f47b2ad28e8438d9ebf38 : SUCCESS in 13m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/33acde0054fd453e9df7f72bf53c6745 : SUCCESS in 16m 48s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/b7c7a81e3fb9402181ae2b0c20f70088 : SUCCESS in 28m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ae37f134e0184f27b70561a78725a5b8 : SUCCESS in 22m 18s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b071f0591c8f496aadf3826b1fe9a276 : SUCCESS in 14m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/412ae39f18664472864ab4a5ad51cc7f : SUCCESS in 12m 36s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/2cf1adec30f34b94afb2f8d21434e87c : SUCCESS in 13m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f1143f521e4d435692a3cd99d5e6e553 : SUCCESS in 20m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1cbe0199a2ba4225ac83cb5cb69490cf : SUCCESS in 7m 55s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/cf6a4c384ef643c7bb7921920d8fbf59 : SUCCESS in 12m 57s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/620be00a4c484d888f017e36299218e5 : SUCCESS in 11m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ca54513986f144aa90711c9f620a94b3 : SUCCESS in 1h 28m 05s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f1fce7cf81bc4746878d0c79edc3118a : SUCCESS in 1h 41m 01s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4882c1eacdb946d387bbcc4a1fb3b0da : SUCCESS in 51m 56s","accounts_in_message":[],"_revision_number":1},{"id":"2998f0c16d04b46b6520572fd289b2ffcb21dae2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-04 16:48:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"600a801c10828c013f36cb1ca3713e5271447e28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 19:26:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8fa12645a18456f9ba4c721ca52baa1 : SUCCESS in 7m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/56e873e994174d01b5680f7e4edcb5ad : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ca57e0a66e749d5a1b4f66e745336a5 : SUCCESS in 6m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aebf369be54d4264bb682c9c43afee3f : SUCCESS in 5m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/28a4796ad853451bb5e1941afdc2f12d : SUCCESS in 4m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/24305a5594fc45fe908ffabef392d1f4 : SUCCESS in 4m 29s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/70fc7a218f2c4e1c9027175fe69f7010 : SUCCESS in 8m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4d36627a080b437494b010cc63d8badf : SUCCESS in 6m 29s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/181c16b4a7f049c18b2b4a34c1efbcce : SUCCESS in 9m 38s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1b7a3b8561a94051884f3e18a8228498 : SUCCESS in 7m 04s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/a5fcce6eb614495db7a898bfc1fad7fe : SUCCESS in 4m 36s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1debfb6cc28d45c88144ceafe0212806 : SUCCESS in 5m 18s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/f2c9f5c6119b47268e8977fcbd4a82fc : SUCCESS in 9m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/73853361b9e94ec993ecc1d7a21074b9 : SUCCESS in 12m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c2bb5a26b296459e99ffdd1199591a06 : SUCCESS in 7m 49s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/f66f66335100419fa14b5df6ba9e3463 : SUCCESS in 12m 27s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/26c4231fb1b24315b0e08f31f7ddfccc : SUCCESS in 13m 51s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/509a8ac5f9dc4d3b8f1b96afb87d57dc : SUCCESS in 15m 08s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fe7086aad9a54dc797de54e452061cac : SUCCESS in 21m 59s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/2bf70a3994b04f989c468ce06516a137 : SUCCESS in 14m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2cd73197fd9649859f51dc47ff75704b : SUCCESS in 14m 49s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/12f0807a85c9421fbae234fc72b3f080 : SUCCESS in 12m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4cc2c799425345fa8567fcdc49ae215e : SUCCESS in 18m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/28d010b24b8d4e8a986c3a0fcdf8a123 : SUCCESS in 8m 11s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/fd249cc37b6c49abbb055a1165086e39 : SUCCESS in 13m 53s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/dbffda1a2447491a8d263614db7af599 : SUCCESS in 8m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d0e93251a52d4cad9591f00f09edbe2a : SUCCESS in 58m 19s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f54e38dbc21d48d9a4797ea786f91b3b : SUCCESS in 1h 46m 18s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f6fd940ad7fa4775a94f474ca25a4047 : SUCCESS in 55m 19s","accounts_in_message":[],"_revision_number":1},{"id":"6d1e7c03a74dc595fe25ea4d4e82477f9fa6cb98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 19:26:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9ecb1f11c6fc93c6792e480473abc1b08069104a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 22:04:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e19c181a9d55b5d263bed25de96b4d05feb6f5d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 22:04:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56e3a5970df84724a54f7d202989503d : SUCCESS in 3m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aa961a89e74c43439a70658e011acb85 : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37eedf910d96462fb73500d527ad241e : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cc984d1641d24ec8a59d79d6eaf3ff30 : SUCCESS in 4m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ebd64b55b787446492a7931118088704 : SUCCESS in 3m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5225325ca1cb4867968f89d1ee0852c1 : SUCCESS in 3m 53s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/8bbef86971c7491da3e7557180919e1e : SUCCESS in 5m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0c28bbb6e1914a4a9679b075f51bcac8 : SUCCESS in 8m 46s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/dc091ea713fb402d8071aedd81ba425a : SUCCESS in 6m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8355984cfe6541c09b71a3a7af683e20 : SUCCESS in 9m 15s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/19ff2a1589174a24996a1a5a687143bf : SUCCESS in 4m 14s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d40ec194e24a48ed8dabe595ed3bd70a : SUCCESS in 5m 43s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/28f2c39755da46e6891ffb413416ad59 : SUCCESS in 7m 11s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4470ff9c52684ca482e5a3cf93b14ea4 : SUCCESS in 8m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cde69c245ed84e3cb4bc17271395516e : SUCCESS in 6m 36s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/c06f88894a324ac6b6f747eb13ca7797 : SUCCESS in 33m 26s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5c00e35bfd4e414798b20c9084900ef7 : SUCCESS in 10m 34s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/1dec2e3990f549c2bd59e5e0b731b2d1 : SUCCESS in 24m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f3969e8bff5e4dd9842e7f5957ff8dc9 : SUCCESS in 19m 58s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/ef01684912df4990bdbc9bbb0dadc7eb : SUCCESS in 12m 57s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/184be0cd863d482e85d471246edca3be : SUCCESS in 21m 26s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e1db3dcd51624793a073967e40a35715 : SUCCESS in 14m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dfc1f30772ce418fb912335a7e3222dd : SUCCESS in 18m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8bc31838c6494f1393765cc68687b871 : SUCCESS in 5m 20s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/98b4873bfb7b4ea6be4aac6083633590 : SUCCESS in 17m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e004cdbcf7524d7eb9dcffa17f1645e1 : SUCCESS in 8m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b9669057edbb4fb9a1d4635577093480 : SUCCESS in 1h 04m 07s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ec4dcb20a56942e19a71e56edad02644 : SUCCESS in 1h 23m 08s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5a717e829d3a43129c2cdf22568301ea : SUCCESS in 1h 24m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0ffbc751e486c92782f2a06b0684a77c1a8aca15","revisions":{"0ffbc751e486c92782f2a06b0684a77c1a8aca15":{"kind":"REWORK","_number":1,"created":"2020-09-04 08:59:54.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/05/749905/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/05/749905/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/749905/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/749905/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/749905/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/05/749905/1"}}},"commit":{"parents":[{"commit":"f99535d223460dee15c5a8e294ddc6b12c0c0e7e","subject":"Add validations-libs to the global-requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f99535d223460dee15c5a8e294ddc6b12c0c0e7e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-04 08:59:52.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-04 08:59:52.000000000","tz":0},"subject":"update constraint for os-brick to new release 2.10.5","message":"update constraint for os-brick to new release 2.10.5\n\nChange-Id: I5586495104c3de42067ab8f6f353314b54a1d135\nmeta:version: 2.10.5\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Commit: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Change-Id: I336163e4e7fae05ef62ec7cfec59ee127340c93c\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Luigi Toscano \u003cltoscano@redhat.com\u003e\nmeta:release:Code-Review+1: Boyron Sébastien \u003csebastien@boyron.eu\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0ffbc751e486c92782f2a06b0684a77c1a8aca15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0ffbc751e486c92782f2a06b0684a77c1a8aca15"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
