)]}'
{"id":"openstack%2Frequirements~749952","triplet_id":"openstack%2Frequirements~master~I165782d1301afdf2d1edf2391f651519db689d9b","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I165782d1301afdf2d1edf2391f651519db689d9b","subject":"update constraint for mistral-lib to new release 2.3.0","status":"MERGED","created":"2020-09-04 13:20:10.000000000","updated":"2020-09-04 20:31:54.000000000","submitted":"2020-09-04 20:30:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749952-1599251430986-1d2acb16","meta_rev_id":"e12a8ad39c5d4192b6d9d042f60e6e0c0e676077","_number":749952,"virtual_id_number":749952,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-04 20:30:30.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-09-04 14:30:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-09-04 13:21:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":2,"date":"2020-09-04 13:21:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-09-04 14:30:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-04 13:21:26.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-04 13:21:29.000000000","updated_by":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"reviewer":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"state":"REVIEWER"},{"updated":"2020-09-04 14:30:14.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-04 20:30:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fffb998213368a068160d3898a8a5b4920a9ced7","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-04 13:20:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"13b03ab0e53d67f87e824bcf12274787535a9946","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-04 13:21:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a047869c33d455c6aa26b33be1d38f71b471c13d","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-09-04 13:21:29.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"91636743a284c43357ea4e6f16c8a589dde0a475","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-04 14:30:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dcb69d2ed478a24446bc379423f50e3a7e6d8861","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 14:31:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/42c612cc74ed4c12a4a9f9428b5bced9 : SUCCESS in 1h 06m 53s","accounts_in_message":[],"_revision_number":1},{"id":"98c7e02984f49e5c710721b5fcb023927ca82b8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 15:47:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d2655a8559f45b7805fd4d470f87052 : SUCCESS in 8m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b68aa19768c241d785568211302c3747 : SUCCESS in 9m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/107466b5d1924175a9d28716419c887e : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9cd241a7acdc45b0ae1409894fc2ceb9 : SUCCESS in 5m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f9d7ac9cec824f7a83eb213d27909daa : SUCCESS in 4m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/855a769be64d418998d7d0a3589d1f1d : SUCCESS in 4m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/39c04c9f0cc149098867c17ebae8ac91 : SUCCESS in 8m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8bc54a587fea489f82c81eae45691acc : SUCCESS in 11m 02s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/54f116eb23f04856b33a3ec1fa0ca943 : SUCCESS in 13m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f86b5fc3d55b4f889ede59c69e4712dc : SUCCESS in 5m 16s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6bc047e047dc4ff5b1383c1b8a529471 : SUCCESS in 10m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2da7dd01623b4cb3b667c4a6fb58460e : SUCCESS in 7m 10s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b4fea8b2890145a199bac1e4115fbea8 : SUCCESS in 8m 29s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/68b99ce220ec48a5a37f44d490ab1b60 : SUCCESS in 14m 07s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fe0a111db2554a5a84b6c9e5577656f4 : SUCCESS in 25m 57s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e9c6a97a29024aeeb6ccac5001075837 : SUCCESS in 13m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/df9e77a1e2fe47d9ad6e382b724f51c4 : SUCCESS in 4m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4854c6e91d9b479a868f62dc432d478e : SUCCESS in 12m 59s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8e0af9f9ecbb438bb2a5210c0b29a32a : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/281641a12e544f978e4d7aa032373540 : SUCCESS in 6m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e444aa4b418f4799ac3dea1b7d2c82cd : SUCCESS in 11m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/65c68ac704b04a168e582643bbe1c1ca : SUCCESS in 55m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/229845ea066845df88accaaffeb37f57 : SUCCESS in 1h 31m 35s","accounts_in_message":[],"_revision_number":1},{"id":"dafd0ace8fede1de1cef5cc74cb4327665617255","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 15:47:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"55ca14307fefde1fea923c512672f253c0f46e64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 17:18:50.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cd2a02fccc04649a74d41fdd2bda1bb : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6371cb330ab546ba98daa834cb5f14a0 : SUCCESS in 7m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bfd73a1a011146b598e33be8e0506e2d : SUCCESS in 7m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3488405ba031451bbeb4ebd54bd995c8 : SUCCESS in 5m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/02e2185573af4fd9aae57aa4faf6879a : SUCCESS in 4m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/853e75c5dd634d64a944ba3ce6244ea2 : SUCCESS in 4m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c541893d627244d69ba5ba2a0140ba3c : SUCCESS in 8m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a8ecdc66c94b4b5b8e7483c74cfdb289 : SUCCESS in 10m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a5831fc9b4dd46339082cb14468b5310 : SUCCESS in 9m 12s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1f95a3c1f5e24afbabb190e4d7115803 : SUCCESS in 6m 46s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/22a18939e78f41b7a0545c86018ea897 : SUCCESS in 12m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7896312a6d9541d89815bcfa7e05ca5d : SUCCESS in 9m 18s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f57bac5d9b1346fe8ec99e481e2ceb49 : SUCCESS in 11m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/eab615c9bbc740b291af8fc0fa07e780 : SUCCESS in 14m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/eb07732e03834d0ab880420c67a95ab3 : SUCCESS in 26m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/09d5ad5122aa46c7a88370a760765b3b : SUCCESS in 13m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bd67c77c4232486f8df53443e49e0bac : SUCCESS in 6m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/000c2f65e48d456798e54097b41b4b63 : SUCCESS in 18m 01s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0046812e4d244acbba503c162bb21458 : SUCCESS in 6m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/53882fc7568b45e480e32678374916bd : SUCCESS in 10m 00s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2363308571924d6b9c97d10d2b059114 : FAILURE in 12m 27s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/70b62cf3913a4250828af242499223d9 : SUCCESS in 11m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e5c4e4d366f54f838249154a6e623360 : SUCCESS in 57m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d391a76159ae4fbaa67e270a8103a8fe : SUCCESS in 1h 19m 15s","accounts_in_message":[],"_revision_number":1},{"id":"6cc7f61d81f566b2f4e50e4bb0100154fe3872f8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-04 17:26:39.000000000","message":"Patch Set 1:\n\nMay need https://review.opendev.org/750000 to land first.","accounts_in_message":[],"_revision_number":1},{"id":"41e9b201e070e0b25dce4e0fdcebd68ab9cf5507","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 19:00:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b03e0e4db5cb452a9649abee1718cf79 : SUCCESS in 7m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/972f29c2bd9344e4bd0f56f8ec16b35a : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f699f5af9adc4d3fadb0fd2180b903a5 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d03ad33ec4844ae83cf8cd221911d96 : SUCCESS in 7m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0db3956b08b1438894b06b83ee6fcad9 : SUCCESS in 5m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8244ed89473141809dcd701bbf3dbb67 : SUCCESS in 5m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6c26558af2db42e7bf3635b92513d938 : SUCCESS in 6m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/17344a45908044299c1b373165629e80 : SUCCESS in 13m 16s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/74f56d33c7e74d868b01eaeb94ef2ce7 : SUCCESS in 13m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/60cef4fc71da4592b19da8df400e0051 : SUCCESS in 6m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/63d53515c97842a7b5db38ac2ba07de3 : SUCCESS in 11m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/98c938c7067247a2b6773775a24764a5 : SUCCESS in 6m 40s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1e31801a39cc40ed86ba3885368cbe0f : SUCCESS in 8m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7d5f2acfd0c94be095b465c296c3e230 : SUCCESS in 20m 59s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fe32b94d240e4bbcb1e605b4a7bb111b : SUCCESS in 25m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/75b3f2e81cda4d6897618555c77a079c : SUCCESS in 12m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/062eb02e5ef8428ab0072df2d5e7691e : SUCCESS in 4m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7ee6e40e2bda404287dffac6bfa7d4bb : SUCCESS in 17m 48s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/7ec543c7b5f045f0bc81fdcde73325a0 : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/70780e579ca34d5eb1c206764309dbdf : SUCCESS in 6m 24s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6bb44e7b79a7431aade45c5ffd0c4651 : SUCCESS in 18m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/46c4d0ef9f5f4ea19113d35105018128 : SUCCESS in 55m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/551f9dde68c743949e394d1047c52b3c : SUCCESS in 1h 15m 19s","accounts_in_message":[],"_revision_number":1},{"id":"ed931cc08c7e54e0193a45eee81e6e1ef0caecac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 19:01:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b742672e04b4dcb143a5d09a558cd90a0a056212","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 20:30:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/644fecd44a1c43d192ae12ea4e384a2d : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4903d47b958e4d9dacb00937e1d7ea52 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/854c7b24cf7e463ea999f5320a803a27 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6f907f229184487826d96b17145fe9c : SUCCESS in 4m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/758602813f53410cab24a85f5350973c : SUCCESS in 6m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4a5b8cee1de2440a95c9da144ae053f6 : SUCCESS in 3m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/13b76fc6a81a4c2089f6892a100e99b9 : SUCCESS in 18m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5fa6db906ac3452a9a1dc2736629c062 : SUCCESS in 9m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/97db2a7cf25d486fa685864ff8a39f38 : SUCCESS in 8m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8ef9ab9008ff4116b7332498f20bbed8 : SUCCESS in 11m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9e2d7d2ea35f446d828a365183232b3c : SUCCESS in 10m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b5500da6de15474097e7109dd166fb4a : SUCCESS in 7m 59s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7536866aebbc4535aeb61e2b41b24ead : SUCCESS in 20m 01s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/add4c0777731473dbeedfbb83e296251 : SUCCESS in 13m 35s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/33040f56cc0c4ae18fcb2d29b70d1330 : SUCCESS in 34m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4f21aa505abd4d01b2eaaf773af4ed93 : SUCCESS in 12m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9d963e4f142f457593bdc63730cffc5c : SUCCESS in 4m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e8ad302416604a59b54dc9d08d670692 : SUCCESS in 21m 40s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9d72e6174f28489aa14da59449ed0f1d : SUCCESS in 6m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/12ddea5534e046ada2a4120bf2ec562c : SUCCESS in 9m 13s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4b254bf0b5124865b21518512843e600 : SUCCESS in 17m 08s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/41a9c51b1585418791ecc57cdefee281 : SUCCESS in 16m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3dc09807009b4ed1981d6e5cff77690c : SUCCESS in 59m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9182ffb366a448b1af16d67e0410b347 : SUCCESS in 1h 06m 49s","accounts_in_message":[],"_revision_number":1},{"id":"b1fbcae0270766e49aa598e32e54b2f934b9b9e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 20:30:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c8901ae1ba1694829584f456b07520766f225c79","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 20:31:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e548d21b4354e568e35313a259b306d : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c2d833f1e54f17c12d3f82d4401809d495d195f2","revisions":{"c2d833f1e54f17c12d3f82d4401809d495d195f2":{"kind":"REWORK","_number":1,"created":"2020-09-04 13:20:10.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/52/749952/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/749952/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/749952/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/749952/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/749952/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/749952/1"}}},"commit":{"parents":[{"commit":"a924149f94563ca8c84d15bc95a0dfece1384f9e","subject":"Merge \"update constraint for os-brick to new release 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a924149f94563ca8c84d15bc95a0dfece1384f9e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-04 13:20:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-04 13:20:08.000000000","tz":0},"subject":"update constraint for mistral-lib to new release 2.3.0","message":"update constraint for mistral-lib to new release 2.3.0\n\nChange-Id: I165782d1301afdf2d1edf2391f651519db689d9b\nmeta:version: 2.3.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Renat Akhmerov \u003crenat.akhmerov@gmail.com\u003e\nmeta:release:Change-Id: Id45fec0969de11fe6eea4f033484d7b9f4fb074d\nmeta:release:Code-Review+1: Renat Akhmerov \u003crenat.akhmerov@gmail.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: Boyron Sébastien \u003csebastien@boyron.eu\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c2d833f1e54f17c12d3f82d4401809d495d195f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c2d833f1e54f17c12d3f82d4401809d495d195f2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
