)]}'
{"id":"openstack%2Frequirements~749957","triplet_id":"openstack%2Frequirements~master~I02e6802e7f078f3a97fa6f747bf1dcf83734bd3b","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I02e6802e7f078f3a97fa6f747bf1dcf83734bd3b","subject":"update constraint for ovsdbapp to new release 1.5.0","status":"MERGED","created":"2020-09-04 13:24:20.000000000","updated":"2020-09-05 05:49:03.000000000","submitted":"2020-09-05 05:47:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749957-1599284823550-210ac784","meta_rev_id":"e25920edfde67ccfa451744397f8eebe91d7e0d1","_number":749957,"virtual_id_number":749957,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-09-05 03:12:13.000000000","_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-05 05:47:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-09-04 14:29:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-09-04 13:28:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":2,"date":"2020-09-04 13:27:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-09-04 14:29:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-04 13:27:09.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-04 13:28:03.000000000","updated_by":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"reviewer":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"state":"REVIEWER"},{"updated":"2020-09-04 14:29:59.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-05 03:12:13.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-09-05 05:47:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6886fa913cb0f92f3017a283cb4d32573a1fc2a6","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-04 13:24:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6fb136bcd18e346f181a0b8e14f588704ddd0bfa","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-04 13:27:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0120c22c06414abcc83b6d01da8910044ac5bf1e","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-09-04 13:28:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d581e02b529fa45c9f96e7a2edcc28a61dbb43dc","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-04 14:29:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ef9a092fdc740e62b3561975083d3920af71716f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 14:35:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/28a54a783e8043e19477131cdd34c3c5 : SUCCESS in 1h 05m 56s","accounts_in_message":[],"_revision_number":1},{"id":"f154681c287d2a2df721de4d282f65eff76d0c17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 15:48:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/beddae14b8a6430ca21179445f84d19c : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5502279dec7424e9ac4207f0f941a28 : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe66b1e32df24768b11d702c072fbdbe : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dfbf8a15713d49e9a0891717f190fda6 : SUCCESS in 5m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c962efd952374129892577d68d4ea873 : SUCCESS in 5m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/75d6bb2e35ff492f9d672e3201fb940c : SUCCESS in 6m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f62de9b5bb1d4da8a36d8c65e056ff4f : SUCCESS in 7m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fc607666097f45099f3be20aa55a85f3 : SUCCESS in 8m 53s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8862e11ec3844bf9995e9aa29f31cffd : SUCCESS in 8m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/72dc6b1463084e5d989436c92a01d928 : SUCCESS in 5m 20s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4d69ad11f62e49dcb152e3d0cf994f2c : SUCCESS in 11m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8c3fe17f431940dc9b1dab97d718ba4d : SUCCESS in 8m 30s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0b7d31a98a174dabb0770628d3388666 : SUCCESS in 8m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8a92c7d4ff0c4fffac9b9b046baedf98 : SUCCESS in 21m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6dcb97ffa62644dc995ff23262390dcb : SUCCESS in 25m 04s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3d4a0982263c433c890735d1e81f66fe : SUCCESS in 16m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/febebe8834734a5f98cb5beff1283d1b : SUCCESS in 4m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/257841267d244639a7ba80f3e732ffec : SUCCESS in 20m 41s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b9bec1e9604f434f90ba08806ac16a01 : SUCCESS in 5m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3f5affb525714ec5912ea5b0fc9fdcff : POST_FAILURE in 36m 20s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/04623992880c40c7b52ee214810f4ded : SUCCESS in 11m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fcac86e0fcca41b6934d84e6a547007e : SUCCESS in 56m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2ae0b7c7be0245b1866eee73f8dcf1a5 : SUCCESS in 1h 17m 10s","accounts_in_message":[],"_revision_number":1},{"id":"4f7cf376adb36555d4432ae226dc5bd941d3db3f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-09-05 03:12:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"db6788aba06eb692939b4e9e87be71fe48022617","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-05 04:28:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/eb8085e29d944602853e9a0e099e607c : SUCCESS in 1h 11m 57s","accounts_in_message":[],"_revision_number":1},{"id":"73f8411d8da0ddd6345f848fc762c4169f4492e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-05 04:31:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0e7562c47e649429cd808cfb98ec086 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cef35130562e4bd2a9637d074c9a313e : SUCCESS in 2m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c363a4baedb04fa68c8e1881062f7cd3 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d6cfec868dc4f1b9d88c2a05f17e345 : SUCCESS in 3m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a868b5cb4e7f40aaa8d1ab2eed7af547 : SUCCESS in 4m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c73e82b25d6b481b8d3f1b0bf637fcd9 : SUCCESS in 4m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e4eb1f469ace481db442adc42c0ad0a3 : SUCCESS in 6m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/69e69571c1bd4d50a5bef3d3b8ac5b94 : SUCCESS in 10m 05s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2c649950f6a341899b75e7d60a9228e0 : SUCCESS in 8m 00s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fcae0c50f0ad4474962400b50304f098 : SUCCESS in 5m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/24f842aff4b74a14aec54994072b9546 : SUCCESS in 9m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b7663d7070f34447969619c442ee7ca6 : SUCCESS in 5m 34s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/be144551bc2c468ebc2371a7d0f252e6 : SUCCESS in 8m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b1dfef859eed4085a8adff31a5b167d3 : SUCCESS in 16m 21s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/20a58007fe3240b6b161c6b762d0bf6d : SUCCESS in 23m 53s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9419c299adde49dc8d66ac40c28aff4f : SUCCESS in 13m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1e4d52c4142a4cdeb1e3152ce344e61d : SUCCESS in 3m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5394da6b3fef46c2944002ad55cf44c5 : SUCCESS in 25m 33s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/7f6fb3b6e2a44dd4a233291898bbb03c : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/18f16cc85c704d54b896b15726e69245 : SUCCESS in 5m 57s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3e02a7d9e2ab4ac9890540a07476d21b : SUCCESS in 12m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00f4b000ab1d4a0f83928931b089b984 : SUCCESS in 52m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1a75ed02547f439fa6dd63ed93a1a99a : SUCCESS in 1h 13m 41s","accounts_in_message":[],"_revision_number":1},{"id":"e11817fc5069d1087e2574ceba27e7b6bebf9581","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-05 04:31:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5b03b32ff59cc985055675c0b5be366e9db8e8bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-05 05:47:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/759d71c71bfd44bea83a894616805b44 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccf21b89851c4363a9256b7d29c3e124 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e7feee73b734f7ab6f34f263e24dc03 : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d485764c18840bcb5f68aede4bb6d62 : SUCCESS in 4m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d908b6727c304f1ba2ce92cfd2e3fbc8 : SUCCESS in 4m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/31f2174faa52400f84d0085e24034310 : SUCCESS in 3m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2906fed34e1141109ebd02a3d4e1a94c : SUCCESS in 6m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/867171be47d1421e904fc9d6f9beab03 : SUCCESS in 8m 33s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/78f95a3d279c4b7a8bd92eea7b843894 : SUCCESS in 8m 00s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b44fc9a91b1f409bb826ff58f2cbbd44 : SUCCESS in 4m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/64ed6811b1214e48a2a160ed426e68f7 : SUCCESS in 8m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8b73e1b834b746d9a9510f2f4694c79b : SUCCESS in 5m 44s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b4915eeae4e84168af05b6895dddb3d3 : SUCCESS in 6m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/18f6777e3b31405f962d78fcea1a5f69 : SUCCESS in 12m 59s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/52c7d13b02a64b1b9714983d829aaec4 : SUCCESS in 27m 49s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/61ef2d231986472f800034975e00a2b1 : SUCCESS in 14m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e9a251cee21b440bbf542b14d9e9b944 : SUCCESS in 4m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/90b1d1f77b1e4b8a81497570ee64c90f : SUCCESS in 16m 34s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/14b33985ef434b4d8e17393983a5766c : SUCCESS in 4m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9dd04dcad622432987138aea0291078d : SUCCESS in 5m 11s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/96d683ffa9594e5cabee6b5036ad7f87 : SUCCESS in 9m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c6319da099ef419bab8666a067d796d5 : SUCCESS in 57m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5d47c4c918bf4020a89ea229427c3d21 : SUCCESS in 1h 14m 16s","accounts_in_message":[],"_revision_number":1},{"id":"a2ab994c8558ada07ac362b948a4f641a0dd2f85","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-05 05:47:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c9af859b6aba565c95c28a7ca3ba45cb4ac72cb2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-05 05:49:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f070659b28941b0a5890ea15e814b18 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"693167c43fe4bc425ad4aff0d7019cb8771c3c42","revisions":{"693167c43fe4bc425ad4aff0d7019cb8771c3c42":{"kind":"REWORK","_number":1,"created":"2020-09-04 13:24:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/749957/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/57/749957/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/749957/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/749957/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/749957/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/57/749957/1"}}},"commit":{"parents":[{"commit":"a924149f94563ca8c84d15bc95a0dfece1384f9e","subject":"Merge \"update constraint for os-brick to new release 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a924149f94563ca8c84d15bc95a0dfece1384f9e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-04 13:24:15.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-04 13:24:15.000000000","tz":0},"subject":"update constraint for ovsdbapp to new release 1.5.0","message":"update constraint for ovsdbapp to new release 1.5.0\n\nChange-Id: I02e6802e7f078f3a97fa6f747bf1dcf83734bd3b\nmeta:version: 1.5.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta:release:Commit: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta:release:Change-Id: Ifa930c53f404ad233e578a4c37f9e55717a0535e\nmeta:release:Code-Review+1: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/693167c43fe4bc425ad4aff0d7019cb8771c3c42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/693167c43fe4bc425ad4aff0d7019cb8771c3c42"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
