)]}'
{"id":"openstack%2Frequirements~749959","triplet_id":"openstack%2Frequirements~master~I1718fb5671ee2bfc5fe1b319d1d54c735a551442","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I1718fb5671ee2bfc5fe1b319d1d54c735a551442","subject":"update constraint for os-ken to new release 1.2.0","status":"MERGED","created":"2020-09-04 13:27:42.000000000","updated":"2020-09-04 19:31:02.000000000","submitted":"2020-09-04 19:31:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749959-1599247862658-7e3b2230","meta_rev_id":"4803223183104008c53338428dabeaaa8a8ff51a","_number":749959,"virtual_id_number":749959,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-04 19:31:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-09-04 14:29:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-09-04 13:28:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-09-04 14:29:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-04 13:28:23.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-04 14:29:44.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-04 19:31:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"235e13307aba1cd46ab411ec4b75de73fe0103a3","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-04 13:27:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"931f67cd354e1da84145af2437299d725c3404ad","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-04 13:28:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"446ce0ad32550528fe2afa5110b75e78e2824444","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-04 14:29:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1d0600aa8c1fb2dc7a4569b75159a0c668ec4cc3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 14:34:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f8e6d687638e4258a5add35e18f45e33 : SUCCESS in 1h 02m 30s","accounts_in_message":[],"_revision_number":1},{"id":"92ffa63fae5c7b696fc44aed33e51bac1b5fadd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 16:18:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbc596168c444a65bf6d7d1b679ddb30 : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd46365c1e0e4830b1e0ccb446f1e160 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d649c1bbf70a493aa0c197c48006cb6b : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/766010c4c45142fc8e27a9d431615ab9 : SUCCESS in 5m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/080a5db0922b419aac04ba982ccaa38f : SUCCESS in 5m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/058f38c25a2e42e2b513d32c64f3b186 : SUCCESS in 3m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/227063e473854216ac82f742adb14428 : SUCCESS in 6m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6d517295b87b45fda0fcb57f004b72c1 : SUCCESS in 13m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a2cd6a0c623541d18cf9099df41cf5fa : SUCCESS in 8m 19s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9fd2bf0ede07400a984f388a37bba664 : SUCCESS in 7m 57s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7b62e9dd8b44477b99d4d8426e0d7056 : SUCCESS in 11m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f387b9bbd6954fb3a059bc7cf2d91a13 : SUCCESS in 7m 45s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/39ac10a4adfa40beaef4f68a40bd9cbb : SUCCESS in 7m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/bee6081a8ba74bdea3f12392fc5a2ab2 : SUCCESS in 22m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1450d2add4ca4f39914b4e35cddec34e : SUCCESS in 28m 11s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/62f92b09725b4d88973de83b1383b5fc : SUCCESS in 15m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/11dd0abf9f084553b2b7c88fe3d954a0 : SUCCESS in 4m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e463a22f4cbb478f887026b19155d2ee : SUCCESS in 18m 49s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5e8e894c3a6745588729a0bdf56353db : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1b67a2a700424a66939039ad8394a7fc : SUCCESS in 6m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/77309ed4e0174ef4abea4789705c506f : SUCCESS in 14m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0bf883870f94442f818110f6d4b60685 : SUCCESS in 56m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cac186e84fc441d28368ba92744c48b1 : SUCCESS in 1h 15m 03s","accounts_in_message":[],"_revision_number":1},{"id":"8f9f97ac1698489348c7be0ff04e46f779ee5503","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 16:18:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8b2aaa767a4e3e30b80a91d8bccd4094caad712a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 19:31:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5b9c71a58af8cd91221c49bcb0a3f5b5ced3f56f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 19:31:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddfbb1ce137446d680ad495abdcf2183 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a581f770a6bc453b9953ea7630f99da1 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e41e8ded7ccb4aefb0d7ff3d23076201 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38caaebf516b449986f5b80b1e6a67c2 : SUCCESS in 5m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/35fa6098a22547fd941c15e4fcdcd3a5 : SUCCESS in 4m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/718da625f3664457ac153a7109c5afcb : SUCCESS in 4m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8cf7db2306074587acfb6c157642b51a : SUCCESS in 7m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d968777c901e4701b8aba51084dd854d : SUCCESS in 10m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d34efb5d277c40e89dbb07259e859d62 : SUCCESS in 10m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/90c78918ff544a5096e216edff0c8920 : SUCCESS in 6m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4c78494644e34232ab9ebbc50e1ff603 : SUCCESS in 11m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/17f4ff218ade424c8f797627c873ad8d : SUCCESS in 8m 17s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/516add382a024d2b953fb4b33190c29c : SUCCESS in 8m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1a9d49a559c045209ac10f5c32e9233f : SUCCESS in 14m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2e1bfc982c5f47e189566dfad85e6623 : SUCCESS in 26m 00s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2dcc9ad6ca8f41c790b1dc5d567d1c3f : SUCCESS in 14m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ce977f038dba45cf87cae0bd12814cd6 : SUCCESS in 4m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b1d54dc0a8674ed993662f8288937370 : SUCCESS in 17m 10s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b30dca27865d49fb846b57f929dc6313 : SUCCESS in 6m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e93b1e8f08f649c6a71d66223e0acf3a : SUCCESS in 7m 52s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a45b001194df42049645db3413f9e543 : SUCCESS in 10m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8a2f6ca25a0d40408aee0387f3a2a7dd : SUCCESS in 10m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8b10399df2734cf390a43b04acb9d6f9 : SUCCESS in 58m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9e1bffb0692e4da9962d4d8bade2a89b : SUCCESS in 1h 16m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b34a1b148d991e5e7593749cafda94765dde03af","revisions":{"b34a1b148d991e5e7593749cafda94765dde03af":{"kind":"REWORK","_number":1,"created":"2020-09-04 13:27:42.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/59/749959/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/749959/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/749959/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/749959/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/749959/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/749959/1"}}},"commit":{"parents":[{"commit":"a924149f94563ca8c84d15bc95a0dfece1384f9e","subject":"Merge \"update constraint for os-brick to new release 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a924149f94563ca8c84d15bc95a0dfece1384f9e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-04 13:27:40.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-04 13:27:40.000000000","tz":0},"subject":"update constraint for os-ken to new release 1.2.0","message":"update constraint for os-ken to new release 1.2.0\n\nChange-Id: I1718fb5671ee2bfc5fe1b319d1d54c735a551442\nmeta:version: 1.2.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta:release:Commit: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta:release:Change-Id: Ifa930c53f404ad233e578a4c37f9e55717a0535e\nmeta:release:Code-Review+1: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b34a1b148d991e5e7593749cafda94765dde03af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b34a1b148d991e5e7593749cafda94765dde03af"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
