)]}'
{"id":"openstack%2Frequirements~750084","triplet_id":"openstack%2Frequirements~master~I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533","subject":"Updated from generate-constraints","status":"MERGED","created":"2020-09-06 06:16:41.000000000","updated":"2020-10-23 21:57:54.000000000","submitted":"2020-10-23 07:08:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"750084-1603436930773-041edf42","meta_rev_id":"c02557fd3c7f00c996f645ab9da8df1b8bc7062f","_number":750084,"virtual_id_number":750084,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-23 07:08:50.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-10-19 17:21:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-10-19 16:41:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-10-19 17:21:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-11 11:01:11.000000000","updated_by":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"reviewer":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"state":"REVIEWER"},{"updated":"2020-10-19 16:41:01.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-10-19 17:21:53.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-10-23 07:08:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a9d0e7aafe666fa969efcb7be675c105b6341610","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-06 06:16:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b2afaeacae359b67121152a289d785ca15d9ff83","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-06 06:42:52.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3d087d2bbf644c1f9c20158a512f0752 : FAILURE in 15m 47s","accounts_in_message":[],"_revision_number":1},{"id":"f3b3b5aff4cb33065dd155aafe0b36495ffba19f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-06 07:02:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ced02d7a609240c292833e0c72024ecc : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a8fc9edf68e4cb3bebb826f6907c50d : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/298a22599b814ba7b74b40019630da66 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a99ab922583448bc98fc0fd29bd225f9 : SUCCESS in 6m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1eb901c801b44f2aaa1e966afc3b0a5e : SUCCESS in 5m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9a5efa1048e54bf5ae8c0409a7bf0f3f : SUCCESS in 5m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5dce357e53e846b6a09c65ffa2dadfaa : FAILURE in 6m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0672a1ba597d401d8300f40098f7e0d7 : FAILURE in 5m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8e0bca1e5c244e4dbb2f7ce873b9a7a5 : SUCCESS in 10m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c0115ba21bda422180850ae7cbbd66b9 : SUCCESS in 6m 21s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fe48073605274316b69b0f3b8aad3f1e : SUCCESS in 12m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6e41eb1e492b4c25aae9e99142fd57d6 : SUCCESS in 5m 50s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/267d1fca536c48af98bb72eeb464720c : SUCCESS in 9m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/675a887a26d5423d8713165394d83e83 : SUCCESS in 12m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/34f7bf602dd749d2a88c332067049d61 : SUCCESS in 17m 46s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4c91612556b241b89fb0982e937bc9ab : FAILURE in 10m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9248ffdbc1b343888e19bb67894d7d01 : SUCCESS in 5m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e2d727bdfd7e4c4a8498419f3530a880 : SUCCESS in 20m 11s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b57e9e4eda9c4a8695ab8a1c49e02ebf : SUCCESS in 8m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6207a3d754274e06a904540db3e888f1 : SUCCESS in 7m 35s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/709de0e39935457289ca09ba71af6e08 : FAILURE in 11m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a18d1c1641f4c2d957131aa42e03e5d : POST_FAILURE in 33m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2565ae47df14473bbb95c63ef199be07 : FAILURE in 27m 34s","accounts_in_message":[],"_revision_number":1},{"id":"31fc12a66ce97f55747af4d691ae569f74ecaf29","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-07 06:16:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c43db8432c7a0a79d2d39b97aa7cf2c3cef01717","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 06:42:02.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b1e10fd762254d8e925fd7432f3dabda : FAILURE in 15m 10s","accounts_in_message":[],"_revision_number":2},{"id":"0336b6a17df6150c862425331b25aa2e77f9c2a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 07:19:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/760b5c342ef440e391cc0f73289ecc6a : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32cabfd5e55d471fb3edaddd828fc70f : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb46ce8c5ff244cd8d0a39bb34d0d6e7 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8dfd6cd2d94f432d8d6d91cba5691a25 : SUCCESS in 6m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/96d6ff1b569a48ac8dcf9f6b55873620 : SUCCESS in 4m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6bdfc365ffa9407e94f982c453f4dbd2 : SUCCESS in 5m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4d1a10176694481984a902f7c5843315 : FAILURE in 8m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fdfd45beea91487cb9dfbd42fce7be0e : FAILURE in 7m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a7786353dd694451b26d8303cf150486 : SUCCESS in 10m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0f85319086a5488a8f6e7f7e92b4e8ce : SUCCESS in 5m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/680a53a2dd244292811278533be5902c : SUCCESS in 11m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c25cab8153ca4c018551944f689e78ec : SUCCESS in 8m 44s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/945c2e25c8b145dbbfb213c023f8c6ff : SUCCESS in 10m 02s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/df4591bf980942cf9655c75e494570d2 : SUCCESS in 22m 27s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e39f233507d14dc2865599fbb69280b7 : SUCCESS in 25m 21s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/68816902e479473f966ad574f8d9c4c0 : FAILURE in 14m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dfe7c20445fb41f3bd3b7ccc73cfe0e4 : SUCCESS in 4m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f66c2e48d17442408657913c4a91a724 : SUCCESS in 18m 56s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/75c5cee6a4024939b59650fc40e35c00 : SUCCESS in 6m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7154b59a12c4408f99b0fc0591c54e0e : SUCCESS in 7m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/964ab0180ff14e9793488ae095f7cfdf : SUCCESS in 11m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bbf04e2217ab4ddf9923d61e8cde64d3 : POST_FAILURE in 34m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9427d194add14d3a90683c14099c4eb9 : FAILURE in 28m 48s","accounts_in_message":[],"_revision_number":2},{"id":"8a821f047795796a7e0c93b36bfa8f8a53161014","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-08 06:27:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"12279b3fc3dbb8947f7905fdd0de6dbc88144a69","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 06:51:58.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/cd4ea7e6aab84dd5a9b378196fe9d03e : FAILURE in 17m 44s","accounts_in_message":[],"_revision_number":3},{"id":"5cd3ee6cfae6c0fd783f48d296eebca34da936ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 07:49:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5be0bfbaa31f406ab5fcf6d3832fd1a9 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a390a474595940b89e2b5d9ccf2a0764 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44f2d79ae5df4c80a6fdfcd6c7333138 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/617d978fd966473eaf5eae47db17a2bd : SUCCESS in 4m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5a669dcc47d3420f97527c67cc8f6036 : SUCCESS in 5m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a38d54ad350a4c98bf2e60f228d7e9e7 : SUCCESS in 5m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/add2cc0d273440b8b2343b83ad44e0ce : FAILURE in 7m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3bc5a035010d46bd939a64a401bd46a2 : FAILURE in 5m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e16d6bee807d489e8f32afaac49d9d87 : FAILURE in 7m 12s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5ff231f992e6468c9661c3b9c4131882 : SUCCESS in 5m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3a32f094d4a7452ba7ffc0df4f7363c9 : SUCCESS in 12m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8f75cdeb15154fca852b838805b0baf4 : SUCCESS in 7m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/efe903340dc640ca806aa983097c0752 : SUCCESS in 7m 09s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1b5803072c7b4788946c03c01b3b5ece : SUCCESS in 12m 25s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fafb81bccd8c4256961427bd2ade6553 : SUCCESS in 24m 18s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fc2a70e4b473422dacc295476a6f9175 : FAILURE in 12m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/679e52b6a017447fab93f16383a23d49 : SUCCESS in 4m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/467512e6749a43f88d1f9ed364ca9736 : SUCCESS in 15m 36s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6231f5f16e0c49d5bad5a2e1645732a6 : SUCCESS in 5m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/22c153400747483a908a361c41a8c909 : SUCCESS in 6m 53s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f945a59ee4b241aba5f706c363f5eb87 : SUCCESS in 10m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/640d234d5e1441deaabc7ba636e5a8b6 : POST_FAILURE in 34m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0381a1392dc5408a9c2e671c4ba0f378 : FAILURE in 26m 21s","accounts_in_message":[],"_revision_number":3},{"id":"5f281902d118ef5e4f6a45e0482bc65477af5796","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-09 06:31:02.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4b91b857c845c95d33a40042a1c3e43d00263b36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 07:12:38.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebc200d5d60040f9a735d2629125a77e : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22b82786b25243eda2dc4182bd01f8da : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/368fe96810b741f0bd5e5bf5d204d101 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd0626d43dc54f9cb9ecdd8ba8ba41b1 : SUCCESS in 6m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/187a6e2ec85f4fc38a65c1ca54087443 : SUCCESS in 4m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2782c6b134bc4050ac34ed91a7a02050 : SUCCESS in 6m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5fa99f6dd1214339a302f6d8a916b3b7 : FAILURE in 6m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a8a33525c84e460f8934c447d8364b11 : FAILURE in 6m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3e0ad0cdbe7747d29a0d848b51288013 : SUCCESS in 11m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/305d0acc5bdd4b1fb214a2f7835401fc : SUCCESS in 8m 10s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/cf50681b0396476dbbb0b68fd0d356aa : SUCCESS in 10m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1fec2d0538ed4f83aa1f6aafefc85fdc : SUCCESS in 8m 10s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/71bec9cbadc04c55ba86473b9fdcd9af : SUCCESS in 10m 16s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/524ec0f8a18e4137895fb5a306ea05de : SUCCESS in 21m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1ba372b87d5f492caa8a1370c4b5b703 : SUCCESS in 25m 41s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/eb700c15615342eea815a62f31cd243b : FAILURE in 12m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2f8a903be6234231b117f4acf48038b5 : SUCCESS in 5m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/26c51343ebaf4f10962f13893c1cd512 : SUCCESS in 19m 31s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/05263a14f1884d949390a3f28e3be7cb : SUCCESS in 6m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/719356c1ada144ab99ce26eee4469425 : SUCCESS in 7m 08s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/18ecf5bc38cf4e25bf446bd0ea36d984 : SUCCESS in 13m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c11abd37c63545579c67a0cf7626641c : POST_FAILURE in 34m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6c64f6a85afe4623a9136a5937e9fa17 : FAILURE in 29m 40s","accounts_in_message":[],"_revision_number":4},{"id":"9e53a2b22324a289e7f52fb2da4036be56dee046","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 07:14:02.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5c6248e29ee748d693ec4105da4a3cbd : FAILURE in 15m 45s","accounts_in_message":[],"_revision_number":4},{"id":"678c63deb472f4cc258c93059ee60cbcb3e502b4","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-10 06:25:16.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b6cde915e06846fe5d17f85dc4a1b38bb9b7cd3a","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-09-10 06:30:20.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"f919b1b3ddeb61e880a92535d159b15886c40bc1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 06:52:38.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b7371a25c71146cbbfa2c060a4d1ec72 : FAILURE in 21m 23s","accounts_in_message":[],"_revision_number":5},{"id":"0ec78bc3f87a7dd78031a5aff7fedab13faddc80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 08:01:37.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95e56344c93f4aa6a269ba92c4ddf7fd : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80d5a2ce745a417180674d7fa4852099 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec86e44350824ca4b1917cb948b487e4 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b26f19c77fb74ca6a6d6b4889b99ffa5 : SUCCESS in 5m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/33bcf4bf628d418dad05dfb9a885540d : SUCCESS in 5m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f418e562bebd42438f9ba4ae9dddcc3e : SUCCESS in 5m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cb11703f3bfb411793b11f725dbad9f5 : FAILURE in 7m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/89f12d5ab74a4108880b273a30204a7f : FAILURE in 7m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/97b22b4de9d94a42a9cba67d2a5e8fcf : SUCCESS in 9m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bc4285de2673481c91c756678b8c248a : SUCCESS in 5m 08s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4ab3d7a94a224396903e1d939bce0362 : SUCCESS in 11m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/26d0b5abc8c84a5693ea2a3091f81737 : SUCCESS in 7m 38s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/84d3420ef89f465e979cb5ba27d88a6a : SUCCESS in 10m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/93a22f0b271b4a3f96d7b62251ea6a7f : SUCCESS in 14m 10s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ce893b5a6ad84e4d9271be6f27973c14 : SUCCESS in 25m 53s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4b47fdc530f648199056d57179b97bc8 : FAILURE in 15m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8aa482abbcbc4577bdaa3a8fc04a543c : SUCCESS in 5m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/da6cca0cf42c481cad41c7547c55d5f3 : SUCCESS in 19m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8ca3748d1bb0445b8ff1771f04807538 : SUCCESS in 6m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9101ee355cc84716b48eb7426331fec5 : SUCCESS in 8m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e1e5f984df014b0cb90e0ea65d1e9c9b : SUCCESS in 11m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d13569db918b4ba7b3b111676cbe6a48 : POST_FAILURE in 30m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/25bb70c337d04b11816c952e5ea8dfd6 : FAILURE in 28m 31s","accounts_in_message":[],"_revision_number":5},{"id":"080673e8f8763c1d85abaedad5b980f9c62253b5","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-11 06:34:41.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0fff6a3799cf2c874ec13205443ba0f4e2aca929","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 07:24:04.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5ec58bc19cd342ff8eca85ca1ae7755f : FAILURE in 11m 37s","accounts_in_message":[],"_revision_number":6},{"id":"0d098a29737043fca919524c5c5a4b69f452efc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 07:46:04.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3b971cdc0f545fbb15a5583b3a9492c : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21018d32a56a46239047cbb039fead5f : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/148bc79f2aa24220a4b688b3e93e3058 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ebdf714df51e417ba3e3ce1787b0d0e5 : SUCCESS in 5m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/75229ecb83ec4d149e88d5630e0bd7b6 : SUCCESS in 5m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ba785954391d45d58c24ee10b357d56d : SUCCESS in 4m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3898cb27824c44c5b096e56f0821fbb3 : FAILURE in 5m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c25ec92f39e1456e97f3968432b6149c : FAILURE in 5m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/93f802fbbe824e83892ecef26176e81e : SUCCESS in 10m 32s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1a1b24ecb91d43b4bbf0fd79865f0adc : SUCCESS in 6m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2620d83c651b46d8b30fe105425ec5bd : SUCCESS in 11m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/adf34bf53d3045e9a460df45a265b239 : SUCCESS in 6m 32s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/127da5c12088440599cc6f4e0ae4940a : SUCCESS in 6m 56s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ee77b8e721f64f679751f16352d6715a : SUCCESS in 16m 58s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e0d6521f92724fb18d0bfc2551a3b2a5 : SUCCESS in 24m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9d267b007f7d40548c6e60625eee63bb : FAILURE in 12m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/102d84a1b96b4d88aeb1fa72b486da4f : SUCCESS in 4m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6e117d5d215d4bad934e13c0f5d4aa73 : SUCCESS in 18m 19s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/307f5de0bc91479c94e1cce3cc74ddd5 : SUCCESS in 6m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/88b3b4719ab24dc3b71344c0a2763fa3 : SUCCESS in 8m 11s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/324505aeb8b54ef1926f9c7d2280aeb6 : SUCCESS in 10m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/90131cddb0354f88b78fc226e52d2eaa : POST_FAILURE in 22m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b2bcbc6462f34cd6b534f47fe955935d : FAILURE in 24m 12s","accounts_in_message":[],"_revision_number":6},{"id":"dedf80c2f93951b6aab023412f9c67b35dd9643e","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-09-11 08:50:18.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"627b999b0fc5cc1f4ab9b706c2ea52f45b51d8a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 09:33:58.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04340fef53c343aab4e57f77f547ff61 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67eb081121f7492b960d73ceb3b481b3 : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8207f226ce940c79aae128ed7e698c2 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1993c0dc5b574ea7bf26129087925ff0 : SUCCESS in 4m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1f919cdf3b064b77b61a180ba2f84817 : SUCCESS in 4m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/accd41d9609f4d65a77deb054d3531d1 : SUCCESS in 3m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4d05f5f9017c41b1afcef04c01719765 : SUCCESS in 7m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/85db75d4847640ad83f1d5f871436874 : SUCCESS in 8m 05s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c9a1148a875f409e833c5b53d441e0f1 : SUCCESS in 7m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6f9d5885222f468d809fd6a725305dfb : SUCCESS in 5m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5d85e87676da4ae1b617713ebd4aa539 : SUCCESS in 11m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/88235bb1a91a44c295c7320597e05628 : SUCCESS in 8m 05s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8a792a43347c436fbb02dcd4ad19114d : SUCCESS in 7m 38s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4084bca0d8984d87bac29aac2c83b28b : SUCCESS in 22m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/80abb00ed7d24602a5c01329150f5fe0 : SUCCESS in 26m 15s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9b9f96f4e99f4c51a282fdb2b44bc6c4 : SUCCESS in 13m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c020cab912b1407f9d5e3c9e39fd59eb : SUCCESS in 3m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/31581742831b4fe085cb779807195f2d : SUCCESS in 17m 45s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0c9b8bf62ebe4a6882a18b682da0f641 : SUCCESS in 5m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c334f642a0ef45fcb5accf14660970a3 : SUCCESS in 7m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fb56b2e1b4a04688bc6feebc02cbf988 : SUCCESS in 9m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b1b84a27148945c9804c6b70887fef8f : POST_FAILURE in 37m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5328385c13f6458aa542a99abebddaad : FAILURE in 27m 21s","accounts_in_message":[],"_revision_number":7},{"id":"59a4979fa5fd10b16f8eb7daeaa7943ad3cca6ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 10:00:26.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/565bc530ce564c5bbd780bd92d977f42 : SUCCESS in 1h 09m 04s","accounts_in_message":[],"_revision_number":7},{"id":"05ba999ed637de4f214d69d9a5a7ede9ac320273","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-09-11 11:01:11.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"d133180eb21bdab261bf9249e0a41ea071f876c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 11:59:39.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40b65d82f6924debb2859e35c66e95fd : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7772436d0b5d4944b69f7be57b158f1b : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/385f9782aff74f2ab8d3b4fb1fac07f5 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e597072fc1d541d78912cd33d2550abd : SUCCESS in 5m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a687b24c6b0744e2a8f3f43a04925836 : SUCCESS in 5m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/98477c652c6f45908deefea2deb219b6 : SUCCESS in 4m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0d0ea2dca2f148de9f0acae27b5dd51d : SUCCESS in 9m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a83e55444f9149aa8d07ef738a680cac : SUCCESS in 9m 16s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/330a924e2a7242569aed856cfd8ee4f1 : SUCCESS in 8m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9df410144fc2425e90d9443a20e75268 : SUCCESS in 7m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6e51ae8ec2bf459d8c4bcc5f0fa75a54 : SUCCESS in 11m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3fe7f7432ba74917bcaba7db1a8154f3 : SUCCESS in 9m 11s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9bb0ca5fa4164b8aabea6a0c9e85b171 : SUCCESS in 10m 03s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f38ea7318f724ebcb5aa82c9ef2b3a8e : SUCCESS in 18m 47s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d22babd198a245d78bccc2c68cfd3b7d : SUCCESS in 26m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e0dab2bcffcd4ccf8cfadd02bba3758a : SUCCESS in 12m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d35b7d397a534d30a3d2c136512f4084 : SUCCESS in 4m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c5afe1e8ea9e46b6898d4b5a168b9062 : SUCCESS in 21m 49s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2dacd73bd70543e7997cddb32bdd52e3 : SUCCESS in 5m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1c2dc5bc705f4eceae1c5b6e65d8bbe5 : SUCCESS in 8m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6ae7e9356adc4c2eb4be6d7d0eac6814 : SUCCESS in 10m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb2c5fc9691741cd84e2de6fa1155c73 : POST_FAILURE in 33m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/904f85e96e574d509a971b4f43524c98 : FAILURE in 22m 43s","accounts_in_message":[],"_revision_number":7},{"id":"728ad3a37f5926f1a4615882fc0026d14862208d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 12:12:43.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/473c0a034f464bef861374ec7cfa6b61 : SUCCESS in 1h 07m 31s","accounts_in_message":[],"_revision_number":7},{"id":"7ff2294176a3d15b4ffc04d17aba2d98a7e213d6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-09-11 14:29:51.000000000","message":"Patch Set 7: Workflow-1","accounts_in_message":[],"_revision_number":7},{"id":"38daabaee04fd9110527528ef2cd8e6e01bb26fc","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-12 06:26:57.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"40b97e736f487eadc6efea6f9a59d32c9a163a26","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-12 06:51:14.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/40f4c2893d984dd3ac737c1462eabb27 : RETRY_LIMIT in 3m 36s","accounts_in_message":[],"_revision_number":8},{"id":"4b7818c706384c252c94ec78a000972d58b98589","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-12 07:33:50.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22cd03f14ee846e8a2300bb1b87bd771 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d093d64dc444eb7ba549afa4337be4a : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c65557fa30b6485aa705bff689db5253 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8a7f959c1c8465a8dbd953a8de2c28f : SUCCESS in 4m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/93c96bc29cbf4a5180d2643e5b109c8c : SUCCESS in 5m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b8d262a61995437cbdadbd66006dc400 : SUCCESS in 4m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/12ae357d05654d7ea292156c07f80ff5 : FAILURE in 5m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/08969c30c5e44d93bf1950d5c92e37cb : FAILURE in 6m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ab1ea5f5aeeb40ce9d0a4a081c7d2b48 : SUCCESS in 10m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/acb86882e90944eb8c29ffb77aa69691 : SUCCESS in 5m 16s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4b03d068cc46464aa5ff6eab551ed06b : SUCCESS in 11m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/056763b3b3e04998994e23c3efcc553a : SUCCESS in 7m 45s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/42546373cc3045bab94b3c3708dab3ef : SUCCESS in 8m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/86271d85f2394192a4b78a588a94850d : SUCCESS in 13m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f67e813f47b94e588920773e89a009cb : SUCCESS in 23m 37s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9d0c7b75081946cc87ed5219d034f14a : FAILURE in 11m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cef466bf94c6430e933cb0ba47f5d4d8 : SUCCESS in 3m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/18ec0408e9d04ec49e9030477f895005 : SUCCESS in 19m 01s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ce30fb041d9a4c9c9cac5170658855c7 : SUCCESS in 5m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1d35ee98ebf2409ca447f2e1faa49d98 : SUCCESS in 6m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/146bad5c010d4026b1eedd5dc6bc5613 : SUCCESS in 10m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/42afbd58365648c0a19e9e0af69f7410 : POST_FAILURE in 27m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/83e9e5e59af34c5e8171505c218a1cf7 : FAILURE in 18m 39s","accounts_in_message":[],"_revision_number":8},{"id":"c7dc62b47c41ee5d6309e69cf3e9eddb8be1bc23","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-09-13 05:23:14.000000000","message":"Patch Set 8: Code-Review-2","accounts_in_message":[],"_revision_number":8},{"id":"8b789b752ddf4840ec53fa12bb775dd20beff309","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-13 06:17:29.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"e7b06b5ed6d39cfa42a2b1ceb56def8694f5017a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-13 06:41:53.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6b0dcccd62384a8e99fe61634c0069ff : RETRY_LIMIT in 2m 32s","accounts_in_message":[],"_revision_number":9},{"id":"859bb6c148524d7f0d407e35a9a2a204c91d1d29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-13 07:08:33.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df34237afd8243f8a91c8ec826206abd : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8723ee11c11e4957bddc9d3d4f046bd3 : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72a9b12228944f0f93ff26df04cbb14d : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c10c444ae25f42438fe821b952d3cf15 : SUCCESS in 5m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b15d52234a7c417595fa7263d6fee152 : SUCCESS in 5m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d17c8bdfb8b349b2809f1a5245292de9 : SUCCESS in 3m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a73dbf32030445d585ea52f039d5b239 : FAILURE in 5m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/91e06572cd304f9abab67aa476c2b209 : FAILURE in 6m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fae7a055a81e403299a0cbcaabde8d44 : SUCCESS in 8m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c99e668a17614e98a859743282b15060 : SUCCESS in 6m 27s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ff6d44b2de544d56b298eaf032ccb116 : SUCCESS in 12m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b881121ba5d347bd846afcb6ff79eb51 : SUCCESS in 9m 00s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/48fcca8d0e5b4f759e39f8da299893f4 : SUCCESS in 7m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/77f1f4b469c3478faa10a86f903e8852 : SUCCESS in 16m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/591cfbd83cca406cab44c9da2de600ff : SUCCESS in 25m 45s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/69e4a6abec9345e99738a73924571717 : FAILURE in 12m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/42bcbf5f9d984c199180d4546cdb1999 : SUCCESS in 4m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e6ea9e2dd3134219a49602c6d544d018 : SUCCESS in 17m 13s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8145b9963217453db927dc498a7cd1e6 : SUCCESS in 5m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b75147a296414eefb0fc4fbd61fb640a : SUCCESS in 7m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/34c7f903e268461294db68218fc078c3 : SUCCESS in 10m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3e803f4a0556433981fc1ecb78dde414 : POST_FAILURE in 34m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/af1dd807b95e40519ddceb369e24a802 : FAILURE in 24m 51s","accounts_in_message":[],"_revision_number":9},{"id":"949220387e67bc4cc5c2352e330f3150632f54aa","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-14 06:19:36.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"04f61f293cf11401cf874700cbf3aefed739b4c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 07:02:28.000000000","message":"Patch Set 10:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":10},{"id":"aa774666529cbab3b683fefb89aa5273e994b4c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 07:47:27.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44e5799676aa4707bbbc3c532343478f : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81cfaef5a120413ea886a6c2570d4674 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/175f5ef41dc04513a1438e47ccb88fe2 : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a40fdacf27c47499257b84128d6b5a4 : SUCCESS in 8m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7835514e3dbf4a4d95dbbb786037a773 : SUCCESS in 5m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e2a3830961db4684804259f1cc81cec8 : SUCCESS in 5m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/96b64fcb70b840c5a31b9df2ad041f14 : FAILURE in 5m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ed94eb1c36024d9b8be5a0ae65db1e7c : FAILURE in 6m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6d9edf839d7145d0b4138783c93beb79 : SUCCESS in 10m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/99cc12d65ddd4fc881710d974a654538 : SUCCESS in 7m 11s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1c0cf14c25e14230935b9c53aa24e15d : SUCCESS in 11m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6bd028f3df6744d29ce0e3e33fdf4883 : SUCCESS in 8m 56s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/22530ba9baf94ac9b3abfce80061e196 : SUCCESS in 11m 28s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d3e6435361b14f22a9a92fc431820c6b : SUCCESS in 15m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/318a067b74bd41a0bd75adddd87c2e13 : SUCCESS in 25m 40s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c9729260714e474aaf4fb75fbe6b6efc : FAILURE in 13m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f24e0d1131f64440b86318500915b4a7 : SUCCESS in 5m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6569289bc0e94b9c986e563332ebea8e : SUCCESS in 18m 20s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/72b42ec65c72428691d74a9072780301 : SUCCESS in 4m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/150f018d524f45bfa2714f8ccf4e787d : SUCCESS in 7m 57s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/475d9fbefab74afb84ef5bc1bf9e4dca : SUCCESS in 12m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/452bbeee86434a03a6eff244fc1227a9 : POST_FAILURE in 36m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9a0328af70fc47239c91fbfa2d540b7d : FAILURE in 24m 25s","accounts_in_message":[],"_revision_number":10},{"id":"4d85ade32463507ae9bdd448cfa8eb3337726dcc","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-15 06:30:50.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"c59e64b285dc440c0c3735bcc1e4b3a872987289","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 07:27:02.000000000","message":"Patch Set 11:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4a74e1ee2d9642b4abfb01993b02a903 : FAILURE in 10m 29s","accounts_in_message":[],"_revision_number":11},{"id":"b9461972fc5cc5a6235591a808dbd63a0611d240","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 07:55:44.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b24cb259e8ca46ac8de8c1ce0c5561cb : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6db1864d4c0d4a069eacc473103083c1 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ca064ecb16d4e2a8171672d36954ba5 : SUCCESS in 7m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bfea900074314a97bcde64cd1bde2cf0 : SUCCESS in 5m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/543b461787f34f1fad8e1c9284a447bf : SUCCESS in 5m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/03458817989b4d329e0173926f8221ed : SUCCESS in 5m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/649f3d0dde37458881903aa2324c58b7 : FAILURE in 4m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/79e09f8cf60c455089b20c810c838458 : FAILURE in 4m 39s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e44e8f6deb214b99808547d5180f0ea5 : FAILURE in 4m 16s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2f2ecb650e904ca592ea5ac3b5ea86b4 : FAILURE in 3m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/de08661cfbf54b798f033f335eac87eb : FAILURE in 4m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a0bae221889f4b40a093f6d5165492c6 : SUCCESS in 7m 16s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/86fce51e53c6429a85282f901e749072 : SUCCESS in 11m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1eec155ca1e941bd97fb9ff421742e99 : FAILURE in 4m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a1d4eedabc8d4ade9b5a2c7706a59c58 : FAILURE in 4m 45s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d4142e4c10264858b834f045d0001c6a : FAILURE in 4m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a8b458440c0d4a6199859682fb4ab8f4 : SUCCESS in 4m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f09c2dabca3e4b618fdc77af872c466f : FAILURE in 5m 08s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5abac74a08954df1921adc472b9b962d : FAILURE in 3m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e9fce7be428b4063af71a1d55b255de7 : SUCCESS in 7m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/88d56df6eb82491aa75e3afb4c16c844 : SUCCESS in 9m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a41c4a77735d43f48e0d3ec193e8ac51 : POST_FAILURE in 31m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6de0d1e7b6a7499c8af63f7d04bbc9bb : FAILURE in 24m 41s","accounts_in_message":[],"_revision_number":11},{"id":"9b8618f1ac285ddcbf23ce50ac1e95f35c759d26","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-16 06:30:53.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"2b780d2270352fb5b5cab3fcc721e01b208b66c1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 06:52:32.000000000","message":"Patch Set 12:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/72f63027ef9e45b2af09f4d3ead14a0e : FAILURE in 10m 56s","accounts_in_message":[],"_revision_number":12},{"id":"ea0c842167717bc08c04647f803ef43bf809b328","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 08:00:41.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f74afc3bcf6f4356be84956d3cefe72f : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c31c777185cd4d3a9a9f7faf6d166516 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/964a4dd42b9e4e42b82387b32e689d72 : SUCCESS in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/918aab1249ca4b0d93446ef06c10137a : SUCCESS in 6m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1fd78fd845dc40d8a95b1ce7c6d276b3 : SUCCESS in 5m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/549969b048e84eb980a629fbf5ef0b65 : SUCCESS in 4m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fd88e77cfba94356a299f6eca3a7df9e : FAILURE in 4m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/82c52e40ba264d1785e2e3f90ba2009b : FAILURE in 5m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8008b9614e5046b9ab7ab90f1a4cb422 : SUCCESS in 10m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/179a7dd383e14f1fa5c6cc193f2c45cc : SUCCESS in 6m 29s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/35387b3d7d8c41539aaac71779b832cd : SUCCESS in 11m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2e202627698d434bb89aa897d69f4a16 : SUCCESS in 8m 04s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2c9fb01e07fd49e7881585ad85a806ed : SUCCESS in 8m 05s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/87b99dd073fc43df83a82ce830c1d6c2 : SUCCESS in 13m 25s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/601a4c03ed864d8b810c748857297677 : SUCCESS in 24m 03s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fabc91c10fa94c11abe9651b42cd1e96 : FAILURE in 17m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0f71a17b3e6346efb4d691d2492b1d39 : SUCCESS in 4m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/53374d353b0c40dc9fb02aff2b591681 : SUCCESS in 19m 25s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/11e7754da41b43ff9edc87a705b7d37a : SUCCESS in 6m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a13188c361a44daf8578395b1ba936a9 : SUCCESS in 7m 20s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/10ea60a551354a99b31606e85d6bf389 : SUCCESS in 17m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a74e0772ca9e4f03863f4f75ee243235 : POST_FAILURE in 35m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e80a570cd1e24916a0c7ba1bc38b0e87 : FAILURE in 32m 54s","accounts_in_message":[],"_revision_number":12},{"id":"4e4f31350cb8b5e1b5e8be2e95b0a8c9544bf3d7","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-17 06:33:12.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"b437a322527b0f4b6fcf744a376ce56080122650","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 07:03:35.000000000","message":"Patch Set 13:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1816fa797f7e4c62abaecfc239b569b2 : FAILURE in 22m 09s","accounts_in_message":[],"_revision_number":13},{"id":"73eec814c616cfa54ebbf85114f27f2a5314fc51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 08:52:39.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57fda40417264307b3adbecca2c14d1a : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/235e176542dc41788a4a7b64613eded8 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be78b504acea4a29b3587298b14ba186 : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb2e3bd76f2a43bd8e9204103cf8751d : SUCCESS in 6m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5590eeb46bd647d8b785e262b929a605 : SUCCESS in 5m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/baec838d89ed4b7e999638794a564cea : SUCCESS in 5m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/70b96e2485b3497eb2f2ca73ee25626d : FAILURE in 8m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a69680b756114b4e9d3cf85a4072a15a : FAILURE in 8m 14s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3ec1c2b7be684cd2a8b34a3866012c2e : SUCCESS in 10m 27s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/61aeb001889a4cca8c22cb69260a47b5 : SUCCESS in 7m 32s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/91b2f24f4b7545848ea8e1372a155435 : SUCCESS in 12m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6f72c9ea5d254111b9daed25714f25f2 : SUCCESS in 8m 16s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/67a66f43aca84784afc7d45e3643a406 : SUCCESS in 9m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/596fb27348d14dd2b5c84ff286a5138e : SUCCESS in 14m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1b199085ac0b42be8541eecf5276b7b7 : SUCCESS in 27m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/65189d802cb949dfbef0f245c19b79cb : FAILURE in 20m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f8a8db0b91794cf7b3395c8621aab0dc : SUCCESS in 5m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b2d417cd1e9845bfb455f493363a5d88 : SUCCESS in 19m 02s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/01b5364aae324f9e9e18df2310fddfa6 : SUCCESS in 6m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/50a9da7e005c480bbaed8e863600706a : SUCCESS in 8m 44s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2d15d256565b4601a795349d868e8086 : SUCCESS in 11m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c89a85e7f9844790b1016736ed6c93eb : POST_FAILURE in 38m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/531d8e0ffe9f427dbc228ff69d2ebb96 : FAILURE in 29m 10s","accounts_in_message":[],"_revision_number":13},{"id":"7b6acf3a210c0821f5bb436102f36f95a431ec55","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-18 06:26:35.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"deff72449c94e410ec05b4788a8f7949cc1030fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 07:22:06.000000000","message":"Patch Set 14:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":14},{"id":"054ddceed6237481ee344bd7b87ecb2e5b0d6405","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 08:20:01.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3653a4ad2e44b938a972ffa2dd8c31b : SUCCESS in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1177245b7a2943bd8d20a341d306421a : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dfa8982d47dc4abeb56c870dea5a9bb8 : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0045e8e071041a2a5549db9616261fd : SUCCESS in 5m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b5c3c33b67a4446dae9d3aae2138f794 : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c39128e4cd70406a85226c2984fa2c39 : SUCCESS in 4m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/751727f661104688a3b805e45c14355d : FAILURE in 7m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7fbc900bd7a1496485735a48f1ee6bd9 : FAILURE in 6m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/85f895e663b74a4494f1ed1bf0357268 : SUCCESS in 17m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ac5c8d9dd70c46b2a5a44acde9ee04b4 : SUCCESS in 6m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/206f8e066c2447f8b2a668bbc4fcef7b : SUCCESS in 12m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/eedad491885b4c1aac48acc2b1f4bb39 : SUCCESS in 7m 59s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4eee474a696b4e9ba824bea520f7584b : SUCCESS in 9m 03s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8113ff79a725426ea202e9b1d2d3ccc4 : SUCCESS in 18m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5dd8aa88197d4eb4be7e0e5d9d697765 : SUCCESS in 25m 18s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e2bee76393ce49a2ad4fbb7412567022 : FAILURE in 12m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a357798c0566421588b48ef6f7810eb6 : SUCCESS in 5m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f93c66cf38374ba291065ad442d2deed : SUCCESS in 19m 26s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9fad853d566043a197a8e97d1ea807a7 : SUCCESS in 6m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/342699bbbfae47a590e271e6f2e33daf : SUCCESS in 7m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9654d57e2c444b2f8d32352acf929bbc : SUCCESS in 10m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/007b001f0d84460984783fcccced2010 : POST_FAILURE in 37m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f020adb9133d49009b454938e5406cdd : FAILURE in 29m 37s","accounts_in_message":[],"_revision_number":14},{"id":"4218d4f175481b4f1d7ca9fcfcb7e1038c4ff788","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-19 06:15:49.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"86204c41750f02f3510c8908ca49816b44b2073e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-19 06:41:23.000000000","message":"Patch Set 15:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4644c1bca15740b5940594eabc7eb72a : FAILURE in 18m 01s","accounts_in_message":[],"_revision_number":15},{"id":"f1e2cc2bdef5b41b938a17a696e9da90bd57e01a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-19 07:59:01.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90e9cae9cc54466ba7b149e68f5b9585 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e66a804dd5c4a638a0ff746ff2563a9 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4bef9466bef4118892812df624d2a4c : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d86fd34b9a054acbb2c6a2258d769236 : SUCCESS in 4m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9d91bcaccc994606b3f3d5754fb37125 : FAILURE in 4m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ecb0184d82194a0c86aef036c7f6a027 : SUCCESS in 4m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bc665fd7f7ee47449d2d5ad9aaa81f7d : FAILURE in 5m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/119e46b7e514401898feefaea127bf6f : FAILURE in 6m 55s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/07a5f1f6cf3c46be973a68b3f679d436 : SUCCESS in 8m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/089914a058d24e61be117e97bd1d7c27 : SUCCESS in 6m 29s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a24b2119b0774670bfd68549e8a629e5 : SUCCESS in 9m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6ccaf5b05e0340a3ab3d47709abbe8df : SUCCESS in 7m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/87b72ff30a184c86804b8e8bbce46fb4 : SUCCESS in 7m 35s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/54296a1d230a4f24aaf56688e3b53314 : SUCCESS in 1h 01m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fa6b1fa37f16447288d7c660a43e4d84 : SUCCESS in 22m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ac9c01f6b16d4f228581195c8ce68c51 : FAILURE in 12m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0ff0b26a6dec46249a85d20da4af8629 : SUCCESS in 4m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9b29ab343264427c908b06a6d8e278e6 : SUCCESS in 16m 30s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1124ee4dfb044020a460b92835abb502 : SUCCESS in 6m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/808b4bf07a5c4f5cbaccefb742899111 : SUCCESS in 7m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0c9101ffbd4342ba8b251fafde5f2efb : SUCCESS in 12m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5f47aebfc7034c1799b91fe3b4fefb5b : POST_FAILURE in 46m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/545a9e37e2cc4fa0babcbf028c8d5efd : FAILURE in 23m 30s","accounts_in_message":[],"_revision_number":15},{"id":"4b96b839fd5bb4a8a8d7b8d40269be54d7a55da2","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-20 06:33:37.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"1d8b306819e4fc40d8cda466be1501dc2c0216cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-20 07:22:46.000000000","message":"Patch Set 16:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8d472d3d9d934b4688cf53131eeb2b82 : FAILURE in 16m 25s","accounts_in_message":[],"_revision_number":16},{"id":"98cf583308a67a15ebc5d38a31809c2e34b5ba0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-20 07:25:11.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56ac4e410a844049a4380c65c2c906c8 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/555e8c72a2f941b7abbbd30b198eee05 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e4cb6ac52554262bb8e98330fef1eb9 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b8d974debcf409a8aaae3b8204fdec8 : SUCCESS in 4m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/29accd0718f64493ad3e9d48fd3b1539 : FAILURE in 3m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/128ce7dbc7614f1b97365d7a5e56a52b : SUCCESS in 3m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e682e2b467634289af2926068027fdd5 : FAILURE in 5m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/65463c0d0aec463f843a243d04b6dc40 : FAILURE in 5m 54s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7b3cc866a60a44c88509c6aaa6e79e75 : SUCCESS in 9m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/eac9f4e263834792bfdb52d6137b1c0f : SUCCESS in 5m 45s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/45828ac583434004984d3c5de3a16c59 : SUCCESS in 11m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/36c120016f3648be90eb005891118804 : SUCCESS in 6m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9579605b03ca40fb9dc1c839de264d8b : SUCCESS in 7m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/70c6db2f2a3c43458e9ee9fe2788e8cc : SUCCESS in 12m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e7225388913a4cf8b39923e3fc030ff2 : SUCCESS in 27m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/55ff1fe71090411c9816217a4d818b86 : FAILURE in 15m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4b43c0b85c824c0599777e8a0e102d60 : SUCCESS in 5m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c4501eb3908d43b7bc29e081e3f5fe3e : SUCCESS in 15m 39s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6fe82d972f6046baa71a2a323c80f2b5 : SUCCESS in 5m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0ee2537d8761491d830d1bad5238aa12 : SUCCESS in 7m 38s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/06acac4c5ec74a8181fd7acf23546f40 : SUCCESS in 9m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb9ad679c56442b0b95d3a971390d869 : POST_FAILURE in 24m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d21d18c10e274613ab09a2aa886bfbdc : FAILURE in 25m 18s","accounts_in_message":[],"_revision_number":16},{"id":"f5329e6d6525ad766025c0ac88c4eb0ddc9ca1dc","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-21 06:24:44.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"7161acf9582920c5616ac1385e9e5595ea804bf0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 07:44:35.000000000","message":"Patch Set 17:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":17},{"id":"626debb7fc9b5ca80abb858a4cb517643dde05b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 08:48:33.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90076cb750ba409baf70c36a5f6bd67f : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9387806de9304b0bbb59b593654899fc : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8c6a2ba99534d6882540cc978c1fced : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cfa95db822d14f37af430149dd09115e : SUCCESS in 8m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/27f2df670b764497b7aa5cf7eb7cc347 : FAILURE in 5m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fc3a9d73aa824ae18951c07c34fbd845 : SUCCESS in 5m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0bc89e9bead647cf85742a49d8ce2a45 : FAILURE in 5m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fdee475cfe554fb9ad8329ccffde31ec : FAILURE in 6m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1d5fea8378e24a89a32695476973651e : SUCCESS in 10m 18s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/14bb482564b3410fac4211a6cef556f9 : SUCCESS in 6m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b137cf7a920a41cab728215ab222d2d0 : SUCCESS in 10m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/eb09cd157a5a4dafbe6b0882d9fc4dbf : SUCCESS in 7m 43s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9b3b3cabcac444f1911b39b5b45e5dc9 : SUCCESS in 7m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/de993e3703d74cd9b8a2c05ea9030dc6 : SUCCESS in 51m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/70bb103787644f57a57ce189ff93bac8 : SUCCESS in 25m 21s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5f6f7e2e06a04d178ccab374cedf4ff6 : FAILURE in 21m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/700413093fb64ac58b9314d610cd8206 : SUCCESS in 5m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ce4bd052fdfc4d7fac030955a8336b93 : SUCCESS in 20m 18s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ca8811d2a00c4e93b04d1bb41d60b670 : SUCCESS in 5m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ecd3842abb1248c2a82db6a1847f00a7 : SUCCESS in 6m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cdc60ef0af0b4f0795bde8f00645d3da : SUCCESS in 11m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/99b1e8ed54ab4fdaa34e5d79948299a8 : POST_FAILURE in 37m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2c476c2bc932422d8bc5be59919df780 : FAILURE in 27m 39s","accounts_in_message":[],"_revision_number":17},{"id":"6da7311444173335d7255b371dd10aef1d940b9b","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-22 06:20:13.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"b7540fc2d8febda3e7b933618e3ee32a0e836cab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 06:41:07.000000000","message":"Patch Set 18:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/88edb6dc7bb247eb849f125e1e7023d2 : FAILURE in 15m 52s","accounts_in_message":[],"_revision_number":18},{"id":"364fb8d607a9c0162304fe2cd57d95ae88d40bf5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 08:05:26.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6ee7225b22f494b9affef748d1a8729 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e0d2f1916bd46028e0cb4a90b571a99 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc3e499ac5af45ea969b40a52f020f72 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9484722a8b249c6bae6e2ee97b6a4a5 : SUCCESS in 5m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1a0c1b654f83459ba6571390201b7f6e : FAILURE in 4m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c8ff1e37c7c9449bb28d09f95da4eaf8 : SUCCESS in 4m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1b60b5ccf19442e0af6c1056694daff3 : FAILURE in 5m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4159171219d74dfda3db535269306698 : FAILURE in 5m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1c659ae03b6e42b6a4b2a4d6c3a77f3f : SUCCESS in 9m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/57b513f522524eb7b694d116954e6628 : SUCCESS in 6m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/97cc7f8c5f574e31b0401056406c4b05 : SUCCESS in 9m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/42ed4c77c84e45fc9a6988e526f38f67 : SUCCESS in 7m 59s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8957b7861a42440c82e6d763b7ebffc1 : SUCCESS in 6m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3eea95f0245e4f48be556b9cbc15d874 : SUCCESS in 20m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/92a03ab521ab45fe8232903e167c5b82 : SUCCESS in 19m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/475e884249ff42e5b22c8d62effe9155 : FAILURE in 16m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/eee13e4e12834d33b0d280c6b1ef6964 : SUCCESS in 4m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/69f17b167a6c4825a66e0b94e5a5c914 : SUCCESS in 17m 25s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/12fc35ba36f648cc83d87346c190bff0 : SUCCESS in 5m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d634951a8d0b4dd9babded1445598523 : SUCCESS in 6m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8cb1a953b1c54c73998ae004a3dcaa25 : SUCCESS in 10m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/51412eca67c64b66953d4c3d56e01fff : POST_FAILURE in 37m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9efb35044bc541a3a7f481496769a523 : FAILURE in 23m 23s","accounts_in_message":[],"_revision_number":18},{"id":"5b9b17f0040b131cf3289bc8e76ec57998d27f43","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-23 06:15:19.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"38be9f823022d798cc95b7c3dd48da60d210e914","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 07:52:00.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a19a304ec6d34ad5a2a0c93350bbd666 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc6a4a5cb9834cd1930f75cc666f01c5 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/edb14722214748e292044479490837a1 : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c1556aa4f924c24aebac69226af684c : SUCCESS in 5m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d882d9f0658f4ffabc8ba05f54519188 : FAILURE in 4m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3f209114a1c24b8383009d0275ae32a9 : SUCCESS in 5m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/859d8a81ac704dcab9b2d06bbc930563 : FAILURE in 5m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6e2c0c10d2274028b58c8bff71f0feae : FAILURE in 6m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/df0e398588de48849db7cd08412758b8 : SUCCESS in 9m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/300d64ee33c74c5f8b6bd057dc4bd34c : SUCCESS in 6m 06s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/07b86541efa4451bac34c3a4b91581db : SUCCESS in 11m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dc9a43bb1d254d85b2e12edd98316aad : SUCCESS in 7m 36s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/bce5159080bb44fbb7a19c18995867b3 : SUCCESS in 8m 46s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d213d9a2eadd4871866ca74e227be2b5 : SUCCESS in 12m 53s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f3c2a8a46f7a4eaa8a41fe1c78b00678 : SUCCESS in 25m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b41a6c5335c44b7b8c06729e0e5f6d50 : FAILURE in 15m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/54ea518e229a458e9671f520eab234c5 : SUCCESS in 4m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c9001d08daf145a999f4b454ff725ce5 : SUCCESS in 17m 02s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d775cd61230d4eb3a6e6b1038c5678d7 : SUCCESS in 5m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/970f3e700e2242a3a82113979c60ba08 : SUCCESS in 6m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9c5b9f4f5f02401aafc49f12f7502516 : SUCCESS in 10m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/464410fd96e04b389e37d7ffef26041a : POST_FAILURE in 26m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e6c52f3a13e2408e9cfe4a96b39f0740 : FAILURE in 21m 25s","accounts_in_message":[],"_revision_number":19},{"id":"e64cf8313b11406b631a9b636726ee929aee89da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 08:09:07.000000000","message":"Patch Set 19:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8a0d25b2af704938a5d4203303f5688c : FAILURE in 13m 56s","accounts_in_message":[],"_revision_number":19},{"id":"1d99a829a38ecefbc76ce7a6a3d0b143c487cef8","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-24 06:22:27.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"39d16c01fc23824eecd184bef1851c93e9d5bdbe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 07:15:12.000000000","message":"Patch Set 20:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/13a7432cd4424149be25575234bb0237 : FAILURE in 14m 29s","accounts_in_message":[],"_revision_number":20},{"id":"ca5db9ab3dc8b5ff6fe4baa41d3bad75e1f61691","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 10:18:37.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32a53421ba7b4b8ab12e81971fb4b33b : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cec43923fcdd4beea48fbe3c3c503776 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b01c96bd95b94f37bdc29aa390a29ffe : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7483c899a16745e0abb1180d9ff26188 : SUCCESS in 4m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bcd1c10babb24c309f019a4aea705d3c : FAILURE in 3m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4bce856ec1404d82b2cdd75b3589d1ef : SUCCESS in 3m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5663d14e1a5045929473d68bd14e6a96 : FAILURE in 4m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/41c35ffc5f9246edaf6ae9f9479095d2 : FAILURE in 4m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/563aaa40bcf04978bd2fdc321df8b14b : SUCCESS in 9m 11s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f6e267382e0c411da8090eb1887f4adf : SUCCESS in 5m 03s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/86258a8b090e4e26aa89aba01397d826 : SUCCESS in 11m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5a0198b9e28b4bb381f371b0ed1699db : SUCCESS in 6m 35s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/40d9ef26799c40eea903dcfb2976ec04 : SUCCESS in 7m 26s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/cd6882b5a2f24a5bbccb49c9ff14d3d7 : SUCCESS in 13m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/94c4d8d377a5406291dc5ab0ffca16dc : SUCCESS in 30m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/30f887e18ce3497da12eecf528ba81e9 : FAILURE in 14m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cbdeaa4a650c479aa9d65fe811ce272e : SUCCESS in 3m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2b080501d56647bd82c72e1c0f99fd89 : SUCCESS in 19m 17s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/96ec5e55831f4d3f935f33257d11d736 : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bddb70ac013740f3a1c60ab2e42687aa : SUCCESS in 6m 24s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4ff952da0f3b400787a7c3e56440178c : SUCCESS in 11m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0c9ce0fea88f465c8868f18053c4e6d7 : SUCCESS in 56m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c217c30e305f4273a4bc1b30cbd8723e : SUCCESS in 1h 15m 29s","accounts_in_message":[],"_revision_number":20},{"id":"ed86d4fa467449b658ea79196f76ab3242a1d4eb","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-25 06:35:37.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"9f801bc2523317755f924f7549a4b12683fee110","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 09:23:42.000000000","message":"Patch Set 21:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/85a83f8be8ac41b2af582f76f6f7b6a9 : FAILURE in 16m 03s","accounts_in_message":[],"_revision_number":21},{"id":"14c0ae7a92878e2fbc322e2d3d870a35453f9fe5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 09:42:54.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/590d92746099413598957e24607797a7 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f293f5d239aa49ce8cbb861cc8f64802 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/70d5f58945274817b38351c4faeafa77 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9b9d35055df46268db8a540124fe37b : SUCCESS in 4m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/812b92e5772a4ed9880b6b91dc088284 : FAILURE in 4m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/645b69acae1444ad8c66536c2651e35e : SUCCESS in 4m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ebca01f424a042a7b853ea804421531b : FAILURE in 5m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/980bf709ba43427f84a085d279a1782a : FAILURE in 6m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/979da3c552ef401c8efac846426d90ed : SUCCESS in 9m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a679bc1877424d29b02004d5b02a715b : SUCCESS in 6m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ba126819984e48f3aae4674bf2acb1b4 : SUCCESS in 9m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a5cf28637d5d458aa392831b7dcb1604 : SUCCESS in 21m 40s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1a420e25087a4ee48831458adc2e52fb : SUCCESS in 7m 10s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8925a1d0e0e047de9b3970220ad517e8 : SUCCESS in 15m 10s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/540667be3cff44f689687a28c83f29ab : SUCCESS in 25m 46s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8adb29730207446bb351da4ee2f43013 : FAILURE in 12m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/612868a0c28d4155972b818efb467ea4 : SUCCESS in 3m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f21f601f221b46d0bf01e21fc916671d : SUCCESS in 19m 04s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c62b5f599b394928928d33dc18851046 : SUCCESS in 6m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/aa86c70a53164787b803845820bc169e : SUCCESS in 7m 24s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bbe475ba55fa40d7acc1945aa6baffb2 : SUCCESS in 10m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f9bd2941c815452cbe37c79fda3c39a7 : SUCCESS in 55m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/077099499cc84735b1cf2acd1066ff9b : SUCCESS in 1h 19m 12s","accounts_in_message":[],"_revision_number":21},{"id":"88f7064d9e6aa0e2583160f189cb92b1b027000e","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-26 06:13:52.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"feed9ad4b4e2fa8120f0562b4ae4005235a25b37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-26 07:50:05.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/005a2936f5ae44e1ae50f483879788af : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/606b29f6401e4998a43e2855dfb8c708 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d3f0053596c49898176569884f08be3 : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5288427dfe34be39412b690a3c4f84c : SUCCESS in 6m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c2748734d42749d998d50322d4c1f33a : FAILURE in 5m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/aec316a5e13242c6a0afdadff90830ae : SUCCESS in 6m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c04611fd792f404287d8be35652f7dca : FAILURE in 7m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5d038f58d9034c468e00ece245daaf51 : FAILURE in 6m 33s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fb10e8b065874a0aaf8d98b2306977e1 : SUCCESS in 11m 25s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9a39227f411f4d6088275adcf4b39255 : SUCCESS in 6m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/017fad7f150747fe8a3b2c4ad3702ce0 : SUCCESS in 12m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/621f1b0ed7ca4838acf5984e9e3fb5a4 : SUCCESS in 9m 50s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9457c41951664b089406b399f1bf7509 : SUCCESS in 9m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e17d94ebf6f04e80a11803272e12c531 : SUCCESS in 14m 51s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/71517eba0ace457c837d1944d847668b : SUCCESS in 24m 59s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/195cbb3e18814659a87c7db25bd41950 : FAILURE in 15m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f413b1b630cd4f4da0e6f72d1b918147 : SUCCESS in 5m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2ddb9f33d5ec46658f8c9974546b20de : SUCCESS in 19m 58s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/29be7e2a1253448f94b10c94a97d208b : SUCCESS in 6m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8dcd46cf6dfc46a9a82cdd21dd1d3b77 : SUCCESS in 8m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/65f4f7648b3648f784b273f2b386f35a : SUCCESS in 10m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/37be853ad1a148daa0b1620e95bb8220 : SUCCESS in 1h 01m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1b41ad2a10844bfd81b84bca3c7f3a2e : SUCCESS in 1h 22m 41s","accounts_in_message":[],"_revision_number":22},{"id":"03b4d2e0672eac8ddd66e193bb9ccae4f06cf6f0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-26 07:57:05.000000000","message":"Patch Set 22:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d91862cb57ce4adcb81fcbf5c2a198c0 : FAILURE in 13m 50s","accounts_in_message":[],"_revision_number":22},{"id":"46d848868f3c4108613f904e7e1402eb2e7bc8c9","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-27 06:31:11.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"4b61892866fc5b07aaa5178ba60e4919b32bea36","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-27 07:52:31.000000000","message":"Patch Set 23:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/855ad99fba674d58b2e851819dca0c5c : FAILURE in 15m 15s","accounts_in_message":[],"_revision_number":23},{"id":"e739b5d8f01ac5fc0cb9acfa54f9a70ce95447ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-27 08:20:18.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41beec16b3bd4475beb3f3024319c2e4 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fb541a7c9344d2d9d4101a1e0195d6a : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d931b03b2c14f279aa2dab568582bd9 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a2cb83be01d497ab3a844c51ca832de : SUCCESS in 4m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/12ef88fd38a04a1b85227335665895b8 : FAILURE in 4m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/30e71717ff05437e846ff23b10096725 : SUCCESS in 4m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/882e0ee9c9764f38a64e778cf9e6ec4c : FAILURE in 8m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d775c9a7d3be4efea42b96d848e5ff66 : FAILURE in 9m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4935532f5dbc4dcebe6ebe17fb0303c4 : SUCCESS in 10m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ab8b3142700d4507ac383689a85e6a50 : SUCCESS in 6m 38s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/445b0a49d7544962b2ba3e74bbea6d69 : SUCCESS in 12m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e75e15744e714f5eaf62300bf562edb4 : SUCCESS in 6m 54s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0ccc42cea8f345d2957a06d7561c3fa5 : SUCCESS in 12m 28s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/67d25e47dc584897b3fe9f1c9d5d3404 : SUCCESS in 22m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/745c58c02f2145cf89d50b62b9ddb82d : SUCCESS in 25m 53s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ec56b6d817bb4c12b843bad74fda6eed : FAILURE in 13m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5a74787223d84952a348a96bda85740f : SUCCESS in 4m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7bc6869adfd24345a0d6861af855ad47 : SUCCESS in 19m 37s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/402dce4ac63f49029a93b08a2e3c3de3 : SUCCESS in 5m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/76d056fed7104785b309865138890e89 : SUCCESS in 7m 22s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b114d631d5e64f74bb7b526c42af1abd : SUCCESS in 10m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/03d961c2f1a64bf49ef94fef3e855cf0 : SUCCESS in 42m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/55952a7c1a43406b88f5dc14aa8ce645 : SUCCESS in 1h 22m 37s","accounts_in_message":[],"_revision_number":23},{"id":"493ff0ff637a057ed3a8dc31a452131378f82547","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-29 06:33:29.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"e493b1299aa61898ecd2e2b20aa08b57b96e7cdb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 06:58:47.000000000","message":"Patch Set 24:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/0ca08937bd9346e2b24511931057bcea : FAILURE in 18m 26s","accounts_in_message":[],"_revision_number":24},{"id":"c5f30c958afdcf60a90993840fadab1396bce99a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 08:32:56.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bdac6ca7e994c5cb20030b1b650d691 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c45ccd0d84049a2a8a1d91a0430ea27 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8585cd4de28a4ce088a545472c4800f7 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/14cc3a62aea049a19c7f548a55f96e4c : SUCCESS in 4m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fa82e97837c84ed9a76c8441a09a36bb : FAILURE in 6m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d586d8ae4d284fecbe24efa047c7f147 : SUCCESS in 5m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e6ae3f50b9474a078e029ecabc2ee017 : FAILURE in 5m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9c76b5f84bf74dd09ebbaf211edb8862 : FAILURE in 5m 41s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a622aa5384cb4c329fbdc3017c514eed : SUCCESS in 8m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/19636730b3e545fa8f45dc0276fcb1b9 : SUCCESS in 5m 46s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bd85cb3b9283480186f513d8cea4931e : SUCCESS in 10m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6945a72f79d64224afc19ded04aa5b1f : SUCCESS in 9m 07s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8c23d96fc6f74ca9b0f5f62ea5fc9a6c : SUCCESS in 8m 09s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/dbe110619b33452989aaf57757ef5c46 : SUCCESS in 13m 37s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/24d8839ab8b64a3ebb6bca090c679cdd : SUCCESS in 17m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6f41730e8cf8487dadf1d107a3998b36 : FAILURE in 11m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/48b0def340284e6e8cb2cca89c870c1c : SUCCESS in 5m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/51414428ee8c49708228c26a8ad3f04d : SUCCESS in 18m 28s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3883e280806b4f7cbbf2577be7194d43 : SUCCESS in 5m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c1724bb1736549e18b7b76e9271cb45b : SUCCESS in 7m 11s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8b90d85321fc460590a208144579bc28 : SUCCESS in 10m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/56e5934c342a4f53af1e9692129c200b : SUCCESS in 53m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3e1bd46e1ce04517ac442ff3faca922c : SUCCESS in 1h 11m 09s","accounts_in_message":[],"_revision_number":24},{"id":"f467e74f537f8e6319f76cc022485765ae8cb52b","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-30 06:30:18.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"fd1607d1bb0e9d47d75d04be5787f358c0a79ae5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 06:56:50.000000000","message":"Patch Set 25:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/631de2aa76924081b512f852210b896a : FAILURE in 13m 33s","accounts_in_message":[],"_revision_number":25},{"id":"6cedea4e7999bc2ee1fdca6a234cf05e022d802d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 09:44:54.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54303656b3084f309fe6426005b7e4fa : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f4d5cde90444bc488ddc8111151dbb7 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc6388700c4d4306bd7828c35f75b0be : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a300353604d94c6f845b848ea0b9d719 : SUCCESS in 4m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/481be1d2b3734b5796c51d53ef06401d : FAILURE in 3m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/06f7ccaf300f45369ad917523ffb5542 : SUCCESS in 4m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a4809142046e4c709234b1293028c2f1 : FAILURE in 5m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/39dee294700e4d17903363c6887e6e4d : FAILURE in 4m 54s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f3ca7e75853e452e8b25c97aba50db6c : SUCCESS in 9m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/21d4e7613c974c35a086a43de0008029 : SUCCESS in 5m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ff58cb6d57bf4c49a2bdd827c2d82dee : SUCCESS in 9m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ff461a236b7742ed81247836f1d06601 : SUCCESS in 7m 10s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9c5c23fcf2f1407a945b5379edcb3909 : SUCCESS in 6m 54s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7384f7bc673442788e2b8b099c7f6ee5 : SUCCESS in 13m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4c05842a95154c6695b37852a94f9c53 : SUCCESS in 31m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4e0c91658c8c4940a8d445ac8fc5f97d : FAILURE in 11m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/53c06f16f15a470497556d1cbab4d96c : SUCCESS in 3m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dd7562424f134eb4a1eb61db7d5849d7 : SUCCESS in 17m 32s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/686fbd51f3dd4c4aa71bbef016678966 : SUCCESS in 4m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0da9783723264404b04d6ae625498843 : SUCCESS in 8m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c2134e3e23fe413c9c6bf7947df09953 : SUCCESS in 10m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7e9558459609457590cfd4482503318d : SUCCESS in 55m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4dccf9d7df4840e2b9efc054d3d5ac97 : SUCCESS in 1h 14m 42s","accounts_in_message":[],"_revision_number":25},{"id":"378bc657c64f484be62551e33a81e10dbafa3146","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-01 06:32:07.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"a4be1f06ee16b1a4b78226328bac1b0cfc22cc3e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 07:06:25.000000000","message":"Patch Set 26:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a6009716e24548738e07307d659804a6 : FAILURE in 18m 55s","accounts_in_message":[],"_revision_number":26},{"id":"ecc06182278e9fd473a749e47ea3fc5ae623b9ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 08:27:42.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fe14ca0aa724257a5802e3a9aae9cf6 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3153957431848a8a77fd047663385b0 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f005c01fb1b54bc4a373c4bee31892f9 : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f73e481065b4086b404c2d922876f2f : SUCCESS in 5m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/17d3a5895137472a88888df840ee775e : FAILURE in 5m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/11dd54d5eb5a4c95b0f0b93e33a945cb : SUCCESS in 5m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/177092c8887d4959b0eaa12a48334457 : FAILURE in 6m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/740dda059ad0465e80d10020a157a671 : FAILURE in 5m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ca2fb5a6330941cdae7cd1e586d407e5 : SUCCESS in 11m 18s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8cc61d1e17544eedb40ce4d44d4b5d2e : SUCCESS in 6m 31s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ad3a2a5025a34d158b9661ab27413e9a : SUCCESS in 10m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/04561f12166c48eb9624f2ea2c947064 : SUCCESS in 6m 56s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0f047c0220f048e1bb830f0fd068e456 : SUCCESS in 8m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/937b41dddaa64b77ba67696d5e873878 : SUCCESS in 25m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/377ca6338c3a4ab8af3f41a9ba720d29 : SUCCESS in 24m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9f98cd8242294032854f367bd95c9715 : FAILURE in 12m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c8e3f8148c9a46838725765c7365f720 : SUCCESS in 4m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d6d5a94f02e54a82bf373b7daa582c5f : SUCCESS in 18m 11s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/03f652393a9741d4bf5831e6544b96b7 : SUCCESS in 6m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/216424d55825461d90e4239bf40c21e2 : SUCCESS in 7m 53s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1664a7c4a498426a86e73f03176332af : SUCCESS in 12m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/717f00dc6c184612adfad3f34ac7579a : SUCCESS in 57m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5e1d30bb15a74a66bf4c03cff27c1122 : SUCCESS in 1h 19m 00s","accounts_in_message":[],"_revision_number":26},{"id":"47ab293141bc2b5e81d839155ca6dc5feaa3075b","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-02 06:27:25.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"ed2e8e34615f8cf231d298bc7aae0bdf3f12876b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 07:06:29.000000000","message":"Patch Set 27:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/96d38d2aaacb4022b4e50f208efc0e8a : FAILURE in 13m 53s","accounts_in_message":[],"_revision_number":27},{"id":"0d316c519323505d908d82ad98e6d15f29d02871","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 09:51:56.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdc61a1a55a4450b81488379a5c060b5 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f70bbb423d34b91b757c3d93c3f47e2 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b38a2d33fb54416f9384b74260120263 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9215e928707b4188bf22eb1d319532aa : SUCCESS in 4m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e448b26555ef406c8a9b1d53a3ac3c5d : FAILURE in 4m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/df78d4bc75e74639b5f8eeb1b30de9b8 : SUCCESS in 6m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9290bcb5747f4d829718acf452e61f0d : FAILURE in 6m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c0e2562ee93b4903984c935cad003665 : FAILURE in 4m 16s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d04b9eb4b2324bdca9c252f537b811ce : SUCCESS in 8m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/26a1f1b0862a42009a715b44305174ca : SUCCESS in 6m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d83ceb8b310448f1b75fbed6cff39579 : SUCCESS in 10m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8a1da785a14d414fa0d666ab36e10ea4 : SUCCESS in 7m 33s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a677b616885a4de2a5d5b4784d0edc05 : SUCCESS in 7m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/bd8533da825b400094e7636c63796a09 : SUCCESS in 15m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fb4aaec6b3b64285a8a887aafdd76ce1 : SUCCESS in 31m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b859ab4dc5c046ffb183cacb2fc65921 : FAILURE in 12m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/33b02a392f3a4132a9fac43428bc8bd2 : SUCCESS in 3m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/16fbad63a6984cf3894d6de7a358e517 : SUCCESS in 17m 24s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/78dbb33ced58435ab2f2458ca8549ed9 : SUCCESS in 5m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8d5331e317684dff8de55c757a6808ba : SUCCESS in 7m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/42d82055b2764576beb07005c413e1cb : SUCCESS in 9m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2107d920e5fc4ae79bf1516237e3f514 : SUCCESS in 54m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c5b9b0c00c5141bea3e6fc8114d8bffd : POST_FAILURE in 1h 06m 57s","accounts_in_message":[],"_revision_number":27},{"id":"76a30ba4e4603cab7e9673c46f8716ce4ac142d2","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-03 06:19:13.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"38320a10981ef14e080875278d7605f4c4ff6184","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-03 06:43:50.000000000","message":"Patch Set 28:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/08dab57a449442d8b81b3c0509e2e4ee : FAILURE in 19m 20s","accounts_in_message":[],"_revision_number":28},{"id":"af90e518e2c675d13ec1757461d78764d332e9b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-03 08:02:56.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4fc48d285ed464a99cb3f0dfa48ae1c : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66c0871c155b4e6498f5b27e8a1f43c7 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a787f7a086524be09046ce2774475ebe : RETRY_LIMIT in 2m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f6c86fbcb65462faab5daa24f9052e3 : SUCCESS in 5m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f549d3a39a3747a0a06bfcd3c10c11a7 : FAILURE in 4m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/859fcfa5215549feaf5ca3b5cc7fa045 : SUCCESS in 4m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/43d99b8eec8c4037878bd13226ead687 : RETRY_LIMIT in 3m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1a27fe6268a247f0a34a61a69e85771d : FAILURE in 6m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/303ba9d5d97b4b9185e335f66ea1e572 : SUCCESS in 9m 57s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5344b975d6a9473995ceccd7d5ce7d35 : SUCCESS in 5m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0e77a6f2d65f44b2a5161c0434ee8ca6 : SUCCESS in 11m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1f3885240b9b43508daa469f89f51e33 : SUCCESS in 7m 43s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e871d528e8604247888973b82c0b77d2 : SUCCESS in 8m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9e0f8a03a78840ef8c1abd1ed427a96c : SUCCESS in 13m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1c943ca4ef8a48b0847cbedf42ee2d02 : SUCCESS in 27m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/886bda37f4ed4e5f8197114c6269298d : FAILURE in 13m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5f53da518c9e48a49d7fbc491c8b433f : SUCCESS in 7m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4d4cd640b0774e5fb461d72ec0f14852 : SUCCESS in 18m 59s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/32a221efd8084cc99d27bc38358495f1 : SUCCESS in 4m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/245df30291f049eba7946b6a2bc308d8 : SUCCESS in 7m 46s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bd0d1827fa8b4d16ab8d2a3d11d449ae : SUCCESS in 12m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/54c9223e57a141cb916b23ca1449363a : SUCCESS in 57m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/919854528b5547a39a35acc857cee09f : FAILURE in 1h 10m 51s","accounts_in_message":[],"_revision_number":28},{"id":"ea2f9d5b5b976e2a17e7be9a12b850b9c816af52","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-04 06:22:04.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"860da70c00d877650565ac8943847d801fd1ce39","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-04 06:43:35.000000000","message":"Patch Set 29:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d3f114f9d9174cf3bc0fa2719c529d3e : FAILURE in 15m 43s","accounts_in_message":[],"_revision_number":29},{"id":"6ea50ea1b586e09392f059edd7a67c38facd06dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-04 08:19:31.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7e5e99d0b8448b3835d942b2df729f6 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3c042deff7c4275b015a557a749ad13 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/087862cea6e8469184ee83d628cab743 : RETRY_LIMIT in 2m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5765f296aab4000b6f8ccee3e88c2e5 : SUCCESS in 4m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/806811b73e1c42f980f866fdab50c98c : FAILURE in 4m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2812e11626aa4e35982081890b6c2e96 : SUCCESS in 3m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5a61be661bf1492b91995db49603507b : RETRY_LIMIT in 2m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0950ff38624c4c84b0e6179182e3b6b3 : FAILURE in 5m 55s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/38c762105e7c46deb21e140340b7b568 : SUCCESS in 8m 01s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bdebff4ad1294ddbafb2e88947db5f34 : SUCCESS in 6m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fff0b920b065465ea541403a4b7d20ed : SUCCESS in 10m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/40b0188a4ca2482eb2d5c4fda53a48e8 : SUCCESS in 5m 55s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/07156504df8b4355b05ddb202877834b : SUCCESS in 6m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9acd132fafca438c86ce032d9f019d13 : SUCCESS in 13m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a096ccd8e56344d89d7fc4d81318adf3 : SUCCESS in 24m 44s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6b2b4d2339c6443dbae21f326bef6a48 : FAILURE in 14m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/efb49d69bc174343b7fcf16f2bd17677 : SUCCESS in 5m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/38a6307f2962400db8e7fa4205bf73d0 : SUCCESS in 17m 01s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/af2fe804b4dc4673a458b093ff99a4a9 : SUCCESS in 6m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/54b6bdb2b72445b6916b34b386c2d6a8 : SUCCESS in 7m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/aab2de6305c2483aa2cd5d293326a163 : SUCCESS in 9m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bac9c0c410e04a1cb226c4f27ce2be9f : SUCCESS in 58m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/01969b7d61234ff587600bed4735f0ad : SUCCESS in 1h 20m 14s","accounts_in_message":[],"_revision_number":29},{"id":"1b6ee2aff78f98f8a7b8457b954ebf83905e0d40","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-05 06:16:51.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"75351433aa0214b9b1c9e1cfe6bd4aae80ef508e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 06:43:41.000000000","message":"Patch Set 30:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1e6f28f4a644497ba0caaa5c01ff8a26 : FAILURE in 22m 18s","accounts_in_message":[],"_revision_number":30},{"id":"ed1cb93eb54c0c8045653e1d4dab5a4346080208","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 08:50:49.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9ccdee222e04684a5b1cb98cde6b46f : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ea20f23ec784a6caa076b6abc7b3785 : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13b953fab3be4d3ca0157381dee97124 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0ce0f7ed85c46d2a842d2aaa9e6b124 : SUCCESS in 4m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/24f2eb4664174cf3b0e937a09d8cd694 : FAILURE in 4m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/143255fcf9984fed8ceb4af4c043606c : SUCCESS in 3m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cc0bd98c5d52436f9be86a8331025c8f : FAILURE in 4m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d235c2e1dbe34084bd317c9fb4e13aaa : FAILURE in 4m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0b9553f049644f0b9682a15b8687c017 : SUCCESS in 12m 06s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0a24ec66e1e74a03ab6689ee1663e290 : SUCCESS in 5m 19s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bd9d66f08b8e4db4b8a17dd2de91f754 : SUCCESS in 11m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c5e31afc120846a691812fb0f0dfd5cc : SUCCESS in 6m 59s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/725e2d8ddb504013823b3ff6088263d4 : SUCCESS in 8m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/68aa311f16ae4323af340fc9a60abffb : SUCCESS in 13m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c068a909345d49eda07e4cb75edb94a0 : SUCCESS in 23m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/efbb95d4663d45a084e8b033485ffd3f : FAILURE in 12m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cba440990086440bbbe7945bcb5e93bf : SUCCESS in 4m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4ea677c6e9b04e87a845c46a1a0db630 : SUCCESS in 18m 56s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cfc22347c8754ab5a8f08a14957d6f25 : SUCCESS in 5m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/acf511ef6655420c993fad1c27bde678 : SUCCESS in 9m 01s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/23f424ff1b63404b8829d560f5c36410 : SUCCESS in 12m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b7d8b9fe1794a8aae04a809ea4f9d75 : SUCCESS in 57m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3a525f547b5241f7a2b659e817b7e9ab : SUCCESS in 1h 26m 28s","accounts_in_message":[],"_revision_number":30},{"id":"0452aecc4ec7b33d75bb826c3e456729b37c709f","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-06 06:22:33.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"fb0aaf1d9dbc3d9702b3dd0af35f8fcf83ed0a71","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 06:48:04.000000000","message":"Patch Set 31:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f5f1270a42744e6890e505f471e39de9 : FAILURE in 21m 31s","accounts_in_message":[],"_revision_number":31},{"id":"64d4709425734b6e921800ad0d5b22ef9ba4086d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 09:49:32.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15235c5ff7a64875a1e5a29b077611b0 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c3097635ec24db4a7f07f99cbb1be90 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/516cf238d4f84b5cbf85885ee43be19f : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0462d3e5c97841ff83584017b2e28669 : SUCCESS in 4m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c04983c3bf3f4ba7af4b1bc3d70f5e83 : FAILURE in 3m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c9f8d8055f3f43499428dd6a849f1bbd : SUCCESS in 4m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f7699de07f3b4ca095aac6f8ee953aa2 : FAILURE in 5m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/02ffe43298d24a788c0ec9234fec588f : FAILURE in 4m 48s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/89b8f5a57a7e45a9908ff853c835d7fb : SUCCESS in 8m 43s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0692e9a2532d4cf9a4bd55228558331a : SUCCESS in 5m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e67512b3d8af4ad1bc8571af09fcda1c : SUCCESS in 11m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b6b6b8e2c20b4fddaf95acf199501357 : SUCCESS in 7m 50s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9f43d90bc0cc44c7be7e71fc435ef41e : SUCCESS in 8m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0acdeeda21ed4e7b9485183eccc2ccf9 : SUCCESS in 17m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/100beaa0457144f1b5ed6ac41dcf49ed : SUCCESS in 25m 15s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7a5189ccd54149a9a66b90ebe15cad69 : FAILURE in 13m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a90afce4806446a986fa96b43a83a18a : SUCCESS in 4m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/47b4bce28aae4fee9c6d13179a6ad6aa : SUCCESS in 17m 59s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/700bdef4ca07481b820f0d392677b51e : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/915ec81fe4c64c8c8198993128816bcc : SUCCESS in 8m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/eba3d39ed6354dfa9f16921b06b9bb52 : SUCCESS in 10m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/98c2f367a46648cb9ede30b0c1342e22 : SUCCESS in 1h 03m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/25d86960944641bb8f1c991d6babdfeb : FAILURE in 1h 06m 25s","accounts_in_message":[],"_revision_number":31},{"id":"62a23922b38c9abefa1b3e98c809f0900f150523","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-07 06:18:29.000000000","message":"Uploaded patch set 32.","accounts_in_message":[],"_revision_number":32},{"id":"d1ad09e70ca1f1d1c3b800a56fdb3e7875b359a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 06:43:18.000000000","message":"Patch Set 32:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d2e8fe1847154758b7084a6fec16dab0 : FAILURE in 18m 25s","accounts_in_message":[],"_revision_number":32},{"id":"1240b1591ff74c6ab7dc4e291477481af7d3174e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 08:41:00.000000000","message":"Patch Set 32: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2743bba3e6f489c92eed52fa89ef306 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85f2ff7123c24eb1808107192181ef04 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f19b64151d9645ac9df454d6f88fc8c3 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f8824e2a02b4cadaa21798e8f400e17 : SUCCESS in 4m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/be62a3e0c0014c4da5c14f33b2bbde88 : FAILURE in 3m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b2d985d73b804b0880a41e4adc169b1f : SUCCESS in 3m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0579f2723d3348e6a9c1e1f8311e6f83 : FAILURE in 5m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2e86a46cbbb24a4d9489253a03f0e959 : FAILURE in 4m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/831e7c005bf444fd9389bbf9fbdedad6 : SUCCESS in 9m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/112852cb64e94bc4b11335a559c2fcc4 : SUCCESS in 5m 06s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e9c03c011a11410fb9396e44af28d8a9 : SUCCESS in 11m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/673b0a8f301b4dc988b52a00e318398a : SUCCESS in 6m 28s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d8699e468c75447e82cad36649cb1883 : SUCCESS in 8m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1d68265c828246c29ad2ebb87b5ab801 : SUCCESS in 19m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ed9c41381c03487fb5f5fbb6011ccfb1 : SUCCESS in 24m 39s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/df2ac0cf03cf428e8d4ad4997ce04a1a : FAILURE in 12m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/478c00313f164e54b26c0d96cbb9f1d4 : SUCCESS in 4m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a2da6b928fc94aa7b60f8c21dd678fe7 : SUCCESS in 17m 25s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/7a2de8f2b33c4f56bc99223e36a5ecfd : SUCCESS in 6m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/80202f62d3934a24bb9879940c1cac34 : SUCCESS in 7m 20s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/562d6f79f5ee43b79112567492c7e01f : SUCCESS in 10m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/905a551acf674069a58374abce55fefe : SUCCESS in 1h 00m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/df03912fd7db462190f5840fba415e89 : SUCCESS in 1h 27m 38s","accounts_in_message":[],"_revision_number":32},{"id":"af87572a817718b77c9cde4fbcf43ba81f037264","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-08 06:21:07.000000000","message":"Uploaded patch set 33.","accounts_in_message":[],"_revision_number":33},{"id":"62be637533e262ca7749f3e9eadf3ee2a88cae69","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 06:47:22.000000000","message":"Patch Set 33:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/abacac3002524b8e9df8ee7a36ebe067 : FAILURE in 12m 14s","accounts_in_message":[],"_revision_number":33},{"id":"a0c908899f6f44b8b371a77661342ee7d2c6099a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 08:35:02.000000000","message":"Patch Set 33: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95ce2159c8e34f9da77e0db19fd732c2 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acf226ad928644e5bb21abe4167bfb8e : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53ca59cc536846c69d888f8588400b85 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c6f87d8d15d4615a63bc48219dcac44 : SUCCESS in 6m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/19f8a4c5acb6400fa3b666970f77bd1f : FAILURE in 4m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f089bc7ef4414a268a4ae2b4fd9c6075 : SUCCESS in 4m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/90e84e9cb74b45fcac31607ef798448b : FAILURE in 7m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/88fa3d572db84e6599e2e856a42b6b45 : FAILURE in 5m 37s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7bdd3cebdb614e5fb738c3bbef868184 : SUCCESS in 9m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fc132eca2d7147e4b077524cc64bb586 : SUCCESS in 5m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/301982620cbd406ba54aa4dd6101e0f1 : SUCCESS in 11m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/74db17d9735e43418856f4806587f71d : SUCCESS in 6m 22s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2631500aa49049869e0c61463792efe6 : SUCCESS in 7m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e50561ceb69d4ec8b8392a678b275878 : SUCCESS in 15m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c83f83ad8db9481d9d5bb8d094e27260 : SUCCESS in 25m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f896a551cdc046f0a893082004687e48 : FAILURE in 12m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/065d1fbc62794689a5c2c5750756a809 : SUCCESS in 4m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c4e54b086b364a3e989cf5a696c042fa : SUCCESS in 17m 10s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3482970f53ac4e44bf079e12c724dd7f : SUCCESS in 5m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/732407aa1cd947c491a3ba38e7984f12 : SUCCESS in 7m 28s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/75ece448dda24f0d9a435fb9f6afbead : SUCCESS in 10m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8e55c175d03e419ebe95e1c0b40a61f3 : SUCCESS in 1h 10m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/76640e84952e420d81a806eeba8fe25e : SUCCESS in 1h 14m 09s","accounts_in_message":[],"_revision_number":33},{"id":"a174d7fd308c11be29dbb9f4714ec0ad443cea2a","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-09 06:26:22.000000000","message":"Uploaded patch set 34.","accounts_in_message":[],"_revision_number":34},{"id":"b2d223f33ba567247a87d9bb54337090c99ff1e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 06:43:26.000000000","message":"Patch Set 34:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6703cf47933644ba9aa8148059daec36 : FAILURE in 12m 26s","accounts_in_message":[],"_revision_number":34},{"id":"0d5f42f2b6114d2439b90fe055494587c48b6cde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 09:53:19.000000000","message":"Patch Set 34: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/835582bf561c47bbb64400b0587bda17 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/805b6f8274c34205b5695e7e585593c0 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d01a74d23a2e44fcbd82595b0f85b222 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/621f5c6e7f634780bb3c05306c183f7b : SUCCESS in 3m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/25585371474148fcaa62be111ff5b823 : FAILURE in 3m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/43977f6cba7f488aa21135a19369d05f : SUCCESS in 4m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fc998b60a22946bfbca78d00e5c60f9f : FAILURE in 7m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6ae74389e56e4d5cb8aea9da58a99113 : FAILURE in 6m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d9b8c53076f14b0591d19bdc3003205d : SUCCESS in 9m 05s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2a51fd4152e94ac5bff3afcf1c0d4733 : SUCCESS in 6m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2edba503497f417ab176e7cd1c09fbd9 : SUCCESS in 10m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9c159cf8cf604f5aaf7e3c172de47cff : SUCCESS in 8m 58s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/74ebd139d9f0428fb7a40cc9af6e8306 : SUCCESS in 7m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/bedd0af281aa418e8864139c90a2be67 : SUCCESS in 13m 44s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0ef2ce17d3f14249b4ac74b2e99d15e2 : SUCCESS in 25m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e2c3c439e0da46e9b8a4bfe96112aa70 : FAILURE in 11m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f5a4309d10d4455d9d038f7baf7a1a95 : SUCCESS in 4m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/68fab1c51be54f279d9b9ad1e42f265f : SUCCESS in 17m 18s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4563efe6fde54cf4ac9689d2c12b60de : SUCCESS in 5m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/63ae316a1a0842e5951e614d68c978b3 : SUCCESS in 8m 50s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0fd6529caff6428ba09c864fe25ff7ac : SUCCESS in 10m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bfc2cf8450674a5fa3b65a6e800de379 : SUCCESS in 59m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c1b1e2d9913749d5aa8c2c0582ab7a8f : SUCCESS in 1h 34m 25s","accounts_in_message":[],"_revision_number":34},{"id":"d5ec950b62b20fe1db5a713e99606759ec48431f","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-10 06:43:08.000000000","message":"Uploaded patch set 35.","accounts_in_message":[],"_revision_number":35},{"id":"2bba7d954db5b919568a7b591760773a19b78bc6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-10 06:57:26.000000000","message":"Patch Set 35:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e684bf3e4e8747dfb6e3e82d66866861 : FAILURE in 7m 56s","accounts_in_message":[],"_revision_number":35},{"id":"94e493ab8e84e217fcb92acc5cb40881674c412f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-10 11:39:47.000000000","message":"Patch Set 35: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad60d1882c954d71bcd6418ac414f42f : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8dd602f9d21f48ab9775f0b386667d87 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e581f23f526d435baa70d4b194a56191 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8bc150203e68438992de5c20eaebe1ad : SUCCESS in 3m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/77f1c6c843bc4c608ab6a8b5eab636fe : FAILURE in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6b6095022b924b5bb27bad3a8fefecb7 : SUCCESS in 3m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/91a4fa575fbe4aa6b3e880b706bbcef1 : FAILURE in 4m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3c87a2721ef849cfae180c800d2557e9 : FAILURE in 4m 51s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/28241fe1fd0b49e6be7a7a43a9632248 : SUCCESS in 9m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6e553762367446a08e06839465ff9bbe : SUCCESS in 5m 20s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5bcaa9b9039c470eb8913f6c40782aff : SUCCESS in 9m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/94f7d453fab34d2a8a9699475a38490e : SUCCESS in 6m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5cda4671ba62432bb6be10f1a576d351 : SUCCESS in 8m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/64c505cd7462403dbc3885bc62e7dc1e : SUCCESS in 24m 32s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/48c59a4f40ee4cd885535761ffd8af53 : SUCCESS in 23m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6e935b322b75465d98ef35483182a9ba : FAILURE in 12m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/88b90f435dbf4a81b96bab84c2011cab : SUCCESS in 3m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9d6eb2372d304fccaffd64c146c685d2 : FAILURE in 21m 31s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9844d4bfb4f3485a9b8d8df9241b0f71 : SUCCESS in 5m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e8eb65c2de754b858d9736cf5764660c : SUCCESS in 6m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ee6c9068eb9349f4b4e0815cae9e1256 : SUCCESS in 9m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb60d99c4e5041ef9b0a58b72b4330fe : SUCCESS in 56m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fd822ece06314f598ca32af5b2af00bc : SUCCESS in 1h 16m 25s","accounts_in_message":[],"_revision_number":35},{"id":"c827746bd08e211871c50f439d7ab0451bbfebd2","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-11 06:13:06.000000000","message":"Uploaded patch set 36.","accounts_in_message":[],"_revision_number":36},{"id":"082763883d82deefa2d394c75668657ed7eb96ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-11 06:36:38.000000000","message":"Patch Set 36:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a5be53435c494973aa3c5c47aa6c7c7d : FAILURE in 8m 13s","accounts_in_message":[],"_revision_number":36},{"id":"619f28466c320e6e24d04339b69ff3597b58a583","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-11 07:52:48.000000000","message":"Patch Set 36: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa57848bba674d9fb34a04445bf88b82 : SUCCESS in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11041fc4e0054e22a375b55560e6f9e6 : SUCCESS in 11m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cce4faa786844da0b46c918b1f9317dc : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8d90e23d2cd415690ca73b78b46963f : SUCCESS in 5m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/38febd6e97234486affac4d9bcdc31b0 : FAILURE in 4m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fc61f9d283a446cd886ad19d7fb4040e : SUCCESS in 3m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b1d323020eb8490e9f59d67927e42767 : FAILURE in 5m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f09a1d1760db42f6b725f7b21d358105 : FAILURE in 7m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9369e5da19ea42d4a818332000a7f0a8 : SUCCESS in 22m 33s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7484a95eecf24a9a81f4902d7478a385 : SUCCESS in 6m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5592f7fd0a744c5abe2fe54bd4efcebd : SUCCESS in 10m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/981e962964e5465a8e43fe87fd877f43 : SUCCESS in 6m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3f94be4b71a645c88bc93c3b138d8924 : SUCCESS in 8m 10s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5cf4fa227aa649af9803adb762aecfcf : SUCCESS in 13m 30s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6199271eb765471aa31f4f9acb9828ec : SUCCESS in 27m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/85a3a10a9d4f49a18bd05f2bcba17f91 : FAILURE in 11m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/626e628dc6da4151aa7370c4728908c2 : SUCCESS in 4m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e55558406279456cb8096e5c91405d7a : SUCCESS in 18m 03s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/41bf13af6d2b492781d7b2a8f373797f : SUCCESS in 6m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/912ed122c29f4348b7b4d8ef5d194095 : SUCCESS in 6m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2a263efcad3a441cb5d7ee135b0c5962 : SUCCESS in 11m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/da3a39ec0f5d46578179a8d2b694af92 : SUCCESS in 52m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a08f1b0bbefb497b87f77a621d4952b4 : SUCCESS in 1h 17m 54s","accounts_in_message":[],"_revision_number":36},{"id":"9b5e6736c90bc2a0b25e509ff07c8f3aa8209f8e","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-12 06:27:41.000000000","message":"Uploaded patch set 37.","accounts_in_message":[],"_revision_number":37},{"id":"4d6e3ea3c111325be30e6eb06afa40d1381dd54f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 06:42:47.000000000","message":"Patch Set 37:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8ea920bb9d8b4bbe8358e8e3b3e38fed : FAILURE in 10m 42s","accounts_in_message":[],"_revision_number":37},{"id":"fe3d0ce02c5b90077f795095c80924d0f0fcd97a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 10:25:38.000000000","message":"Patch Set 37: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc4d0da217444f7fa1cb802daefbd1d3 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c6bfb323b5e4e1dbca42080c9e9d186 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/532fc702e07b4829b755afc4b7285e8a : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/10aad800c37f482586cbebd4d0032ba0 : SUCCESS in 4m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3b91ae673dfe4c9e9ad2728d9078b3c2 : FAILURE in 6m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/115ca2f49e8747cc85ea675a0406bdaa : SUCCESS in 5m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/15602ed56c734d3892539ba7d9197972 : FAILURE in 4m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a6329256db1941ce8c5846cff1f1f996 : FAILURE in 4m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fd4af0072aa24af182806727fb838384 : SUCCESS in 11m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7a2c3c2ad35c448bbd459373be65c3fe : SUCCESS in 5m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/76faef2b099a473b8a38074bb9bd26c3 : SUCCESS in 10m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4e1e6af57f43470eb0696a078ce00dc2 : SUCCESS in 7m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6c284fcc25eb49968959ff4fa1fb7501 : SUCCESS in 10m 58s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/23d93c88ca8e4edab37889f233f6c9dc : SUCCESS in 12m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/da01b5bf5b974588856e4e403c662ca9 : SUCCESS in 17m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5d57d130d6684cf4b0d498dbb6ec12f2 : FAILURE in 14m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/34c6fbab8ae7405f98cacc998032843e : SUCCESS in 3m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4964bb5e9c864e408f14bc20cfd740fc : SUCCESS in 19m 05s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8bad9dc547054a79b51f9d9d2444b19f : SUCCESS in 5m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/33faad70c25443aabf5a9cc68779395d : SUCCESS in 6m 39s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d8a3cd659cb543949ff373481ecbf5c9 : SUCCESS in 10m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/31963186207c42b0a02b353b5030f32a : SUCCESS in 1h 06m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f3a6da49ebdd46aabab8825a72311c49 : SUCCESS in 1h 26m 35s","accounts_in_message":[],"_revision_number":37},{"id":"0fe7bad3e4d83b6ce01c0523a5bb52280cdadcf3","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-13 06:16:14.000000000","message":"Uploaded patch set 38.","accounts_in_message":[],"_revision_number":38},{"id":"fcc5a745edfe64fea320d64b8fb99abbb3befd02","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 06:30:22.000000000","message":"Patch Set 38:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/629d55e532aa487a8175d575c4d46385 : FAILURE in 9m 58s","accounts_in_message":[],"_revision_number":38},{"id":"71c00c74a430d203d21010582c7c6441e3d3f0e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 09:56:33.000000000","message":"Patch Set 38: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/933507ea389e42318be3aa5cad7e7d79 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2640e748929848bb8faf3fc46e6d6e97 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c8290df344c45bf9b2641c9ad20dfe0 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b0d0928a6cf4538aa4e7790414bf09f : SUCCESS in 4m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ee5eb251731444ba8493c087bbf1177e : FAILURE in 5m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/eb5ac437b87e4a6f87aaa0b2802287c5 : SUCCESS in 4m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ece3b4c46d7241d687251ed3713555e0 : FAILURE in 3m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0e98d90528064010acc5ad341d034e00 : FAILURE in 3m 44s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2706eaf1a7a94663b4c99c17be8437a8 : SUCCESS in 11m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0cd431ee80294e258540141bb31aa4d2 : SUCCESS in 5m 54s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/efff7141a2864af9a6237541a369ea73 : SUCCESS in 11m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c3102aa55a3d46b697cfa1dd89243ec5 : SUCCESS in 10m 26s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/964da32f8b3a4a79a8e93780f5dbcff2 : SUCCESS in 7m 01s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8e4984afc772499dbab0bbe984b1200f : SUCCESS in 15m 01s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/13ddb3ac549440dd8554955187886765 : SUCCESS in 25m 15s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1caa1e14178c4b349f2c9646a3388caa : FAILURE in 14m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/79778f3209264c0fa74b27afcbf10f3a : SUCCESS in 4m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b37f585d9b24463081fac72546931e29 : SUCCESS in 19m 15s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a268f5a492cc46c68b960478f9ad9f7b : SUCCESS in 7m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/84e82c9d3ba34944ad2635df3a21e0a1 : SUCCESS in 7m 01s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/439f1d17334f4f4a9849012ef3f6142e : SUCCESS in 12m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a0864ce6872c465bb201a521ccdb19c3 : SUCCESS in 54m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/18a22a0ba55c47cea8eda0675a73e7f5 : SUCCESS in 1h 12m 55s","accounts_in_message":[],"_revision_number":38},{"id":"2f7f468af031ad2a3ca8f77b7f5a298cd3ef35e6","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-14 06:13:30.000000000","message":"Uploaded patch set 39.","accounts_in_message":[],"_revision_number":39},{"id":"fa2aa88ef3d73cef9482f530caf0da2fe5d53f7b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 06:29:20.000000000","message":"Patch Set 39:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d9714bfa6f734dfa9ea241c5760d14c8 : FAILURE in 9m 05s","accounts_in_message":[],"_revision_number":39},{"id":"73caf9204c73600b1c29181bf437a3ae8a30330b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 09:47:37.000000000","message":"Patch Set 39: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86fd116c65c14e738c630e6aee75dd72 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8401658ac08410b92b48e177ef86a65 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0201587cd3114a82b1409fccf6c4f601 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/527fc3c5119e4fc8aa729ddecd306f20 : SUCCESS in 5m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/46944231139b44fa923eef266789e0bb : FAILURE in 4m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1c3d15e1ba924bf29aa1d8fb1850fe4b : SUCCESS in 5m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9c6379475af04f429e0b39ff55f28fce : FAILURE in 5m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e4ee074339334261a7b7974b0f0f0123 : FAILURE in 5m 05s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a542f2a739ee4f3d958415b3a6377ac8 : SUCCESS in 9m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fe1ec152aa4849be97c7e85ae07a43d4 : SUCCESS in 6m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a40e313c74f341d597bb73b7c7ecc01d : SUCCESS in 11m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/25a5245eb9994201bbd05c42de9c6d54 : SUCCESS in 7m 30s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2ddbeef85f4645e28b6bfdff2dd48f2a : SUCCESS in 8m 31s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2703a789e9c8482fb2cf266fe7922502 : SUCCESS in 13m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a53d0dba18a14ee8af95c5c882ec8dcf : SUCCESS in 26m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/46d0d8d377ca4dcb95a893f87c13a1b7 : FAILURE in 12m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a2c40ce1562347feae6b8cd2862bde34 : SUCCESS in 4m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/24860e7a33eb4dfeab58cd44585c46d6 : SUCCESS in 17m 37s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b6df61323c794dd79bc146f9d0093ef3 : SUCCESS in 5m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/50767c65240a48089003324457e2767e : SUCCESS in 6m 11s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/976aa20ab1a048ed8f14bd9cb0530f5e : SUCCESS in 10m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3fa431136cd0439b8d8d867ffa1f1b85 : SUCCESS in 56m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/25443cbce1b24a68a5923a20146106f8 : SUCCESS in 1h 17m 59s","accounts_in_message":[],"_revision_number":39},{"id":"2e03ceefb680e6f7831bfdda61d2d12b8205b81e","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-15 06:20:41.000000000","message":"Uploaded patch set 40.","accounts_in_message":[],"_revision_number":40},{"id":"5acef5e9e2b9155e047a317be3f7f186dd0ed960","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 06:41:59.000000000","message":"Patch Set 40:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/465de988b5a74ecab8c5b471af492333 : RETRY_LIMIT in 2m 34s","accounts_in_message":[],"_revision_number":40},{"id":"4fba0a7c068c89b77e77eb779f73709638d1709b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 08:50:59.000000000","message":"Patch Set 40: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a887f5c019c348488a08507740212106 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7ce7780ef47430bbfa995093620b2a4 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f10dccc0f1c143869acf2cd73d359514 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee841364fca547309da31d4ad75e7d3d : SUCCESS in 6m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e2659739ac764beb802c3e200d5986ae : FAILURE in 4m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3d1e2e5652c444ea97d1f312e4fc4c31 : SUCCESS in 4m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/07fafb80cce64b9092eb1db9be389dc3 : FAILURE in 5m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/08967baa23cb4e88b6c7ada9e43196ab : FAILURE in 5m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/51ad924b02eb44e09472bab63b42b4ab : SUCCESS in 9m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bf6b18e045434004bc3639d500b95753 : SUCCESS in 7m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d95ae322744d4445a5bf6cee3bed68fd : SUCCESS in 10m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/07c1990103c540c9967178ed3c9eb3cf : SUCCESS in 8m 21s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6ac74febf3fd4f9d8b21df353e8f2ebe : SUCCESS in 8m 05s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d6a482e24b3e4e79aeb858a322659ea2 : SUCCESS in 13m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e41e0ccd1368483c89889b46db8a538e : SUCCESS in 28m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/81ac12efc74846a4bc1b8288a5faa7b7 : FAILURE in 13m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4fea6aa67c0c465386a5eebcb7fb703c : SUCCESS in 5m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d44d21b2e6cc4480b18c13d8db68573d : SUCCESS in 16m 09s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e804a1f4ed4c4a35a32cd3b194f8e391 : SUCCESS in 6m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/442bc291415d4e65a573b7ef831d56cf : SUCCESS in 7m 22s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/511f967adb3748e59d1a50d70696d2c0 : SUCCESS in 10m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d7e5899fcf4442888d2a9dff9190005b : SUCCESS in 1h 08m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e6828d48967e4524b78bde10fe86294d : SUCCESS in 1h 20m 21s","accounts_in_message":[],"_revision_number":40},{"id":"2efecd12c1d2bef710c4c0d2edea7b8e63eb6845","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-16 06:32:35.000000000","message":"Uploaded patch set 41.","accounts_in_message":[],"_revision_number":41},{"id":"37c17937c3dc58718e1eb809453416e5719b9b2b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 06:47:29.000000000","message":"Patch Set 41:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/bf997e6a74864e8c9fb737096c98603d : FAILURE in 10m 09s","accounts_in_message":[],"_revision_number":41},{"id":"8c6ab141c2ea30a43a6a807a2d1351645311a2d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 08:39:11.000000000","message":"Patch Set 41: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ef177927b814b1a956dd033dfaf6f53 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a567cbfa659643e687b71b071b24c3ba : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c51c73cc02a4410d86d972442a12f829 : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/845cdbc55d2a424c834ff044d93b384c : SUCCESS in 5m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d29fd8e9fec54dfcbce0a182f55b8881 : FAILURE in 4m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f6dfc23ebda3438093bdd14fe845d662 : SUCCESS in 5m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f2a0e18d63564ee8a848f4589741d7cf : FAILURE in 6m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f805754eecbf4f52919549ffb7904508 : FAILURE in 6m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7912d40541f34368a87dc304aa401041 : SUCCESS in 11m 52s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c736be2e53d748359c3d3376c6def7a1 : SUCCESS in 7m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/68a9c5310a3e4e5fac395d1f47c00147 : SUCCESS in 11m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d22cffa43f764c58a875dd446892ecf7 : SUCCESS in 9m 24s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/76b47e01977648ae9a09c3c032e63fd4 : SUCCESS in 8m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e38214e2a3874c23a695d27a3f6f545c : SUCCESS in 14m 54s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6ea8f8a894b44519a525a699dc386063 : SUCCESS in 27m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e9471bd7f31346529f3d2d10fd329bfb : FAILURE in 13m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/afb6b3cfb78f4545b9a6c29d5710cf6a : SUCCESS in 5m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5368ee74e8094c2fa1d48ee30f57a3f3 : SUCCESS in 19m 29s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0c2d8e2da9b949a2a53c94f48b384f51 : SUCCESS in 5m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7f286adb95b74932ab06bd342a1ef56b : SUCCESS in 8m 26s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2a44df3d153545fd84e29546eb60b930 : SUCCESS in 11m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0ecc71588c8d4497948d3db7cd1b2ffb : SUCCESS in 58m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e89b6cd9fd3a4e8093d5b4c389e458b6 : SUCCESS in 1h 17m 10s","accounts_in_message":[],"_revision_number":41},{"id":"b91c30469dfb6ab4674b5e37f5dee57423136445","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-17 00:51:47.000000000","message":"Patch Set 41:\n\nShould be good to remove -2 now. Though we still need to make some edits.","accounts_in_message":[],"_revision_number":41},{"id":"a92984cd7903ac8b2ebd7e2d625a53a996dba648","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-17 00:52:59.000000000","message":"Patch Set 41: -Code-Review","accounts_in_message":[],"_revision_number":41},{"id":"6101881eb5d0299cab39e9b58f332d7bc82012f0","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-17 06:22:03.000000000","message":"Uploaded patch set 42.","accounts_in_message":[],"_revision_number":42},{"id":"6a566535a81c72fbf7bc3582320b2d9cfce16c6e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-17 07:03:16.000000000","message":"Patch Set 42:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":42},{"id":"3023b747c535fd43e34e6f826583eaf1358bdd63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-17 08:04:54.000000000","message":"Patch Set 42: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f43fa9635ea64f138a60e480b0079978 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da9640626d064029996dc98878258b73 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/778242f723234017850cb9c4c0149d22 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16f6429431d24652979e0e398de2b515 : SUCCESS in 5m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4abfcecb5048499f97c9227060464afc : FAILURE in 5m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ef9014bd7fbd4f69ba233be72d177461 : SUCCESS in 5m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d325779272af484eb49989bbeaa1c432 : FAILURE in 6m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/91187ea50d314c888c21d70ab4ce2f9d : FAILURE in 11m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2d28e172f3d642d08dbf519c75e3a7f8 : SUCCESS in 9m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/83ac945123234741844648bb73bbcbe7 : SUCCESS in 6m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8ba14a75ecef427babf002690d75ad00 : SUCCESS in 12m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d9ad965a737a4d0a858b38bbf1d41d91 : SUCCESS in 10m 24s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3265a7df5ce048d196819e5c5214df75 : SUCCESS in 19m 36s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/92a429448796440ea386e3fb6e56fb73 : SUCCESS in 14m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fe501488b2264566911a5aa6dd304aee : SUCCESS in 30m 46s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5528fbd766114d438cbd9f6ff279a189 : FAILURE in 13m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4f82c343a5934c189a93b21a40f0b1e6 : SUCCESS in 5m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/918f63f007dd4dbc9063339dc8b82407 : SUCCESS in 19m 47s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8cba82fed1ad4fe6a490f57ff99fe66d : SUCCESS in 6m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b2bf555a5a6742f28795be138d520cc8 : SUCCESS in 8m 03s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bffcd4c865fd4f44b07a883e0da8eff1 : SUCCESS in 12m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5f423617b6a840b1b1bc598ad8aaed98 : SUCCESS in 1h 00m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9ebac51d3a5f4a9bb45764f4bc688a46 : SUCCESS in 1h 29m 36s","accounts_in_message":[],"_revision_number":42},{"id":"1110bdd5ffe7b1ae56db4e1224d992e20194aaa8","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-18 06:20:52.000000000","message":"Uploaded patch set 43.","accounts_in_message":[],"_revision_number":43},{"id":"a10bd59f2bac17c39b9f65fe7b3490bf5bb62a35","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-18 06:39:47.000000000","message":"Patch Set 43:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/355e4153847a4cfb9e95e84b328fea21 : FAILURE in 14m 08s","accounts_in_message":[],"_revision_number":43},{"id":"46c4b5b7163e19a2bee0f07fe4b176c7ed13ea00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-18 07:49:03.000000000","message":"Patch Set 43: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85f55280a47a4e07bcdfb55b22ddbaac : SUCCESS in 7m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/871b0be7dfe0447d9bc1a4ffa07415b0 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/afd8c9e3e6774a4995778044588535c2 : SUCCESS in 7m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77dd909be9f141298226ad54cfffeac5 : SUCCESS in 7m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d53016ef1a7f46d98c55e6e16d91e2e7 : FAILURE in 6m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/51fc57753eb2487a96bba2b51e6e2a82 : SUCCESS in 6m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/81b6789113044b2683e19cbee30174a1 : FAILURE in 9m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6983a7e2c8064b42993069d672aff8c7 : FAILURE in 10m 53s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/867dd47f808040e19e08b7e408b11f39 : SUCCESS in 11m 50s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4addf5ea59f54c84b4a409ca186c9f52 : SUCCESS in 7m 57s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e3abe887ea524a22b782038a9eee8019 : SUCCESS in 13m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/10d69c438b0e4dbdad79bcf7c41c60ca : SUCCESS in 10m 22s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7e0a0af9b431482e84c62e3195883152 : SUCCESS in 9m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/80f6a5e045694972b4760219492fb13b : SUCCESS in 19m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1892c2615a55432ea80f5df512fb6155 : SUCCESS in 26m 45s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/69f9453866664ea096e93d3fc3a9f51e : FAILURE in 15m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/df06987a0c6b4c6bbc3db934d73eab35 : SUCCESS in 6m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cfae544928ec486fb2b551bdf7122b51 : SUCCESS in 22m 58s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/89a41d07a41d484bb0c7b40aa5327342 : SUCCESS in 7m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ce195284dc4d4f198f2f511962fcb278 : SUCCESS in 9m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/071f9bfc9cb24bc3b4d406a14b66ff59 : SUCCESS in 13m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dbf23053020f4e76952e1dbde85ae63c : SUCCESS in 1h 01m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/26da10528c234d9ab1ba353ad52fa410 : SUCCESS in 1h 20m 36s","accounts_in_message":[],"_revision_number":43},{"id":"5f7a4b60380ccabe219171fbaf88aae3949d05b7","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-19 06:26:02.000000000","message":"Uploaded patch set 44.","accounts_in_message":[],"_revision_number":44},{"id":"3c7012ec8a2120f762173f0375784e17d143e7a3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-19 06:39:51.000000000","message":"Uploaded patch set 45.","accounts_in_message":[],"_revision_number":45},{"id":"1a2b64bb8d35147790d9239844cb3e9b75c80f02","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-19 06:40:03.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":45},{"id":"0e2bfaaac28b84591ac452b1b781393122ff7e8f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 07:05:22.000000000","message":"Patch Set 45:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b8a86547f267401b857cefdfdaf5986c : FAILURE in 19m 57s","accounts_in_message":[],"_revision_number":45},{"id":"413d4f991fbe6a1d5e3aa67fc2adc66fe6f1210f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 08:18:49.000000000","message":"Patch Set 45: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e1eee1995fe4e948a09af3cc7bc0988 : SUCCESS in 8m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/913f3ad61e6e4bbeb49625dd86f334a4 : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6739c68f9a2747dbb6a399ddd47210bb : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a53853400666478fb8d69800dd129b9f : SUCCESS in 7m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/35fd47f256e84b26b3fc99ba78977b43 : FAILURE in 6m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fec0b2b82dfc4e60b748ec04c53aeaa9 : SUCCESS in 6m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/992d2cf15f08406d986e6d75f0a9ad41 : FAILURE in 9m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5716614fa2604f77b82f9c93cfd71213 : FAILURE in 11m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4a472990f62f418896f3c484691c0831 : SUCCESS in 11m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/107551e426e745ad898fabbb97391afb : SUCCESS in 6m 54s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6aadf8d5613d457aa08bf1e57cb64a15 : SUCCESS in 14m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ac9cc1f758b74ea7b2ed1d44742efed7 : SUCCESS in 9m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b4b93020b06e40fca88a463c1980ec6f : SUCCESS in 9m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7d2c022d02204c05b8c924af85dcf7b0 : SUCCESS in 14m 53s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2499e7d83d7d47a4a30aa11237f7001d : SUCCESS in 30m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/148349a131854b4794011488ade6c89c : SUCCESS in 15m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b19d04536e6f4eca95eb09155916fb85 : SUCCESS in 5m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/858b25e90c064f2ea82480f2d24ae83d : SUCCESS in 20m 28s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6cf5ba2feb8744dca010d7fbb200d719 : SUCCESS in 7m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/68d3d5b64eba45baa2b6243ae5b6baa0 : SUCCESS in 10m 30s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6edca55152134da6a343607dbd9c261e : SUCCESS in 14m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4010d49a52f34a9bb2dda92315d813c5 : SUCCESS in 56m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cbef0f5612e04d89a6c1e279916acf52 : SUCCESS in 1h 10m 03s","accounts_in_message":[],"_revision_number":45},{"id":"833becc93f7b749d2905d60614d432ae5fc604a2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-19 08:49:01.000000000","message":"Uploaded patch set 46.","accounts_in_message":[],"_revision_number":46},{"id":"345936470846133b5f015c6d1522de179ea84b6a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 09:13:15.000000000","message":"Patch Set 46:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ad56b069f0a0451ab0c2eabe104aa381 : SUCCESS in 18m 25s","accounts_in_message":[],"_revision_number":46},{"id":"8bb67b83b8818cdfff91f903649f20dc4bcda3e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 10:20:05.000000000","message":"Patch Set 46: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4347ae5fe7d14e9f9001427a3241d705 : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e83c41c6445a464d9cbc94bddd8e78ce : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94d9eb2a12394c4fa74cc5667a518e1a : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00f0c878670c4f03862d21194774563f : SUCCESS in 6m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f421ad3b91fb4f37ae804b5ebb04fb9e : FAILURE in 6m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ebe78d964b1d4c20bf7b2099140b71b0 : SUCCESS in 6m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6ac6c701586c44b2a57e366d3ad1294e : SUCCESS in 9m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cfc8e06aebba47a78f74f6269f5d3179 : SUCCESS in 9m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/51c866db52d54ea0bf0b5fe4916e080a : SUCCESS in 10m 46s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d74464e0a91f4f6bbdbb33ebc3cb1184 : SUCCESS in 8m 35s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4815a94eb83e43f3a377ed2f0541a9bb : POST_FAILURE in 20m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dd7925df07694a629171e8418004ce77 : SUCCESS in 9m 38s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a0c1dee2ac1847d5ad5b9dd3314686fd : SUCCESS in 9m 19s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4f7a957c0e61403c9705e2e84bd245fb : SUCCESS in 15m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/54d471fcdfc1444281a1d7e997d8df76 : SUCCESS in 31m 11s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ec577cbbdb0f4d42988a4552f08f3568 : SUCCESS in 14m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0808857c301b4f42a11119132c6bde7c : SUCCESS in 5m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1917a97170024464a29b589126671d8e : SUCCESS in 20m 55s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c546578b4f1948fa9e2ec95f056a5f22 : SUCCESS in 8m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f81f77e7896c43bbad4097d3966899b7 : SUCCESS in 7m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d12e28d2a93d4aa08731e81552157362 : FAILURE in 12m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/695f44fdbe8d45a2a471fcc91ecaceac : SUCCESS in 1h 00m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6d93683474404abab5c901f2b9826ca4 : SUCCESS in 1h 23m 22s","accounts_in_message":[],"_revision_number":46},{"id":"6e51cf19a9f5e75a05701c0390b675e4cb98a200","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-19 13:36:08.000000000","message":"Uploaded patch set 47.","accounts_in_message":[],"_revision_number":47},{"id":"e628f3843b8cefa27483cb1a89a0711e63edf0e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 14:00:43.000000000","message":"Patch Set 47:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2f054e27d52a462e9668b6a59bfb38d6 : SUCCESS in 17m 25s","accounts_in_message":[],"_revision_number":47},{"id":"5bc1112b45e94b9706934eb4221461659687df83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 16:32:10.000000000","message":"Patch Set 47: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c0d9c727c9744efb0b9eef967684a3c : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e6396afd1f844d799f574823a0332be : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96e3b46a07d24b63adb54f0d68fe31a5 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c32e17b4956e4fd4a4e236fa748442a9 : SUCCESS in 5m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/da4fe8d6f6f6489bbca6d1cf5924745d : SUCCESS in 5m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9fbc698843e84d5e83ca47f52309bc57 : SUCCESS in 4m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4f9fa62a8fbc46f192fbceeb78f95aca : SUCCESS in 9m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/59dc2c8954744e62889d96db289ad92e : SUCCESS in 10m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/92f10b64198a4e099ddfd4ca9b62d8c1 : SUCCESS in 11m 50s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fac3bd50788345cbbb2814ed2b9add05 : SUCCESS in 9m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/98339d0e7d384b57804384b0586b50f6 : SUCCESS in 10m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3817ffdca51447a0939c512a9d7660a7 : SUCCESS in 10m 15s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e0c10665111b4744b4e10be5bd270ba0 : SUCCESS in 8m 49s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/83cf91f8ae6849a0990642f3bfe20b7d : SUCCESS in 16m 36s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ec42a8b89fe9496ca8f3b826a71dda8c : SUCCESS in 30m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/16a59eade82f49daaa020a0461770b88 : SUCCESS in 12m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/af1b95361a4d485ebc43a989e20609ae : SUCCESS in 5m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5a741a7ccdcb4f26826d907fdc05b677 : SUCCESS in 20m 02s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/df3ee7319ac44103b6d600894e0e7001 : SUCCESS in 7m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bfce990323a8434db825468a9bffb21a : SUCCESS in 8m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ebaccf8fdc1942c9bc2b743048681ed2 : SUCCESS in 17m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/96b50cd3590645a3beb807da59393cee : SUCCESS in 1h 07m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dd39b6d36dc7406cb38cd5c94cee5cad : SUCCESS in 1h 18m 29s","accounts_in_message":[],"_revision_number":47},{"id":"ddab46d0d7407e6444bf3a955257ed862fe5583a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-19 16:41:01.000000000","message":"Patch Set 47: Code-Review+2","accounts_in_message":[],"_revision_number":47},{"id":"61717e37c7ab69d966e9da846c28e3829be2d30b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-19 17:21:53.000000000","message":"Patch Set 47: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":47},{"id":"ce6cac088b2bfc3bc96bd2b056c6e909969ce213","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 17:22:18.000000000","message":"Patch Set 47: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":47},{"id":"ad21b0c14c74614edb384d47910d2f16214e56ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 21:07:27.000000000","message":"Patch Set 47: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0216b1d6fbb43828339ba7d898c9b87 : SUCCESS in 8m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b14a026178cc4d4c8ef75f583e4f6caf : SUCCESS in 8m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e18b66364a984cbc854e2d3629acf4e9 : SUCCESS in 8m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3dcbe5a4dfa4c8a802ff1b0a9106936 : SUCCESS in 10m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cf7f1d049cae413d96f976cfa6ea0200 : SUCCESS in 6m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3dd7cde89c7c41f4a6c2529226c6041f : SUCCESS in 6m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ae8e42fabf4d471194115013d0cf9d5b : SUCCESS in 11m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/049278692a0c4ef699ff6e80c6c16deb : SUCCESS in 10m 22s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fc9613d955144080ac55dbeb04c67492 : SUCCESS in 11m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/43cc584ac21d46daac1c14b050402eea : SUCCESS in 8m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8779e8ea752b48a2a9c67cda0b14b13e : SUCCESS in 13m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/86953ff8e4b341cba14d1b873188f5ea : SUCCESS in 10m 32s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/dfebd1371afd40509ae86f6e91d990cf : SUCCESS in 9m 57s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f0cdd282fe144cd3b40bc5b68c0d0995 : SUCCESS in 15m 29s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c1fde005a2744dec86342f50c7f71448 : FAILURE in 30m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0f6e098644dd4302be4d56367e98230a : SUCCESS in 15m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8f870802099f48689bf24b876a7e881e : SUCCESS in 7m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c0276280271e4c3e9adeb1e10e919e1d : SUCCESS in 21m 08s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/67856f3c6de1472cb4bcdce00f3338a7 : SUCCESS in 7m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6eb076e501324ab0bd60f39e4452e3be : SUCCESS in 12m 54s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/756c2c725d8d4d458a03f93efa21952e : SUCCESS in 13m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2326b35a5d794a4e8c9cadfe811b1b63 : SUCCESS in 57m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0339fb3ded5d48728e50a38cb2a322f8 : SUCCESS in 1h 33m 13s","accounts_in_message":[],"_revision_number":47},{"id":"bc095339e3aae7f470230cbfb25296c509e5c2d0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-19 21:10:18.000000000","message":"Patch Set 47:\n\nrecheck","accounts_in_message":[],"_revision_number":47},{"id":"c9fd2da8d28a4f083401b1c2def12801cd960481","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 21:31:00.000000000","message":"Patch Set 47:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f4032f305cac4c52a495d102ddb87f3e : SUCCESS in 16m 48s","accounts_in_message":[],"_revision_number":47},{"id":"bedf5d3b9497c95e61cfca380b9b1ac14b1f04bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 22:47:13.000000000","message":"Patch Set 47: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d0bf7a7ca36487790db68cfa9e7a135 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5aa1fd7ce49f4db1bf0f5884f0abf7a8 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/785acc0ebf6c4ceab340e2cb8396d9e8 : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f549264501b948d998a57e17b4f2e32b : SUCCESS in 5m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/735a052079354725b553138372044b73 : SUCCESS in 5m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e0bcc4faa9b24574b7ab39a14848e1c9 : SUCCESS in 5m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/286cdc665312426580bf7ac26b084cfd : SUCCESS in 9m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3a87bc23784942d29bc7a01a98e81b3a : SUCCESS in 8m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/98aa5251c1c14347b67b139c6369d6bb : SUCCESS in 15m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e4f821a1a6c34ba4a07832364fc2f159 : SUCCESS in 7m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c790e6b2a9b7462dbe58f509e42add42 : SUCCESS in 11m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b18bdb19d85549899a6c5778dd5520e6 : SUCCESS in 9m 06s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7a31d38ac484404fae8fd5dedcd6dc36 : SUCCESS in 10m 31s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7182c06860a340e69c2ff354b8e67670 : SUCCESS in 14m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e6af2f76d6d947c08a4fa910d8580c08 : SUCCESS in 27m 31s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1ff3f682d23a4eb3842cb8d430a1753c : SUCCESS in 14m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/887f7c89ed83452e9ec976a63b9dd9e3 : SUCCESS in 5m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ef1cb7e8d2eb4bf9a13ef22dfdbfb605 : SUCCESS in 19m 19s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/91c9c32eb4a949179fc6bcb97e1abb09 : SUCCESS in 6m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/78bacc5c955d4204ba938bad6af634c4 : SUCCESS in 8m 24s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b9bdb821726f4fb7a87459b93a53ab2f : SUCCESS in 18m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/622a0d3068ab4a36b161d77ada86965b : SUCCESS in 1h 13m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/768454c0bc69411581abfba7e72e144c : SUCCESS in 1h 25m 28s","accounts_in_message":[],"_revision_number":47},{"id":"c581b645110e975250a0cd7388a534c3e3a4dfeb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 22:47:31.000000000","message":"Patch Set 47: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":47},{"id":"6e3a5cf245a93183dc0ee0b101948bdfdff03d37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-20 01:42:01.000000000","message":"Patch Set 47: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0876baa5e4ac427085cc201282985e45 : SUCCESS in 9m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/814c8418f096403e9d9abc6109e79e9d : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd364ccad4304608afc86d6cc018999b : SUCCESS in 8m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8cad8d538e5f4a96a3031bfc3694a01a : SUCCESS in 6m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/da121390fb854dfe84040c0c53daddf5 : SUCCESS in 6m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/334a8d59aa2740c2bf077265e03d35e8 : SUCCESS in 6m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c33bda6f75c5423ba43ae7424c18d35d : SUCCESS in 10m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/094b74a0aea442f4bb911dcf05905428 : SUCCESS in 13m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/72b9465a8e9a43aea0c937e8235ee97f : FAILURE in 10m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/597750d8b1a34495a651ff91307e558f : SUCCESS in 8m 06s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7653b5c239f645e091092a68d99e7952 : SUCCESS in 9m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/196661499a7a437a9bf54af0e7601292 : SUCCESS in 11m 45s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/18f6cfada39f49bba4a98b33d78926f5 : SUCCESS in 15m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ae04e309870d47a39addbe57bfd2245c : SUCCESS in 32m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/198ca29b09b54637924194ac6120bf3a : SUCCESS in 27m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e2e02e74367c44cda308cf2c841101d4 : SUCCESS in 19m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6534314830594cd18f164f676307c5ca : SUCCESS in 6m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/888b1cf110a24c26a21112b0beb9bc81 : SUCCESS in 19m 23s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c80ea569f7f4420a90fe381ed06b2278 : SUCCESS in 7m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0af04d548b0a406ebae6fe01ecb8baab : SUCCESS in 8m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/66eb980d8c8b42f28fe7b37577160bb3 : SUCCESS in 16m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/43de07bc9baf412d9d1a63ae677fccf9 : SUCCESS in 1h 00m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3fabd9335a44437ea8491fbc73955b82 : SUCCESS in 1h 23m 47s","accounts_in_message":[],"_revision_number":47},{"id":"af8318b2b4c82e08154a6dfcd81deca2558ac9fc","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-20 01:57:38.000000000","message":"Patch Set 47:\n\nrecheck","accounts_in_message":[],"_revision_number":47},{"id":"58a78d01114882de9e6f691db0603e47826d3998","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-20 02:20:10.000000000","message":"Patch Set 47:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/738cb78944514d47aae7311ae39b9a2a : SUCCESS in 16m 35s","accounts_in_message":[],"_revision_number":47},{"id":"916f43d7b7097d1dc66e0a76749aa5420437b14d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-21 14:39:29.000000000","message":"Patch Set 47:\n\nrecheck","accounts_in_message":[],"_revision_number":47},{"id":"22af52447bdd5935d9033a3ccccafffe952d4ab6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 14:43:15.000000000","message":"Patch Set 47:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":47},{"id":"529ea98fe8a6a9fab55ad74bcca3699d6f28f772","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 16:06:16.000000000","message":"Patch Set 47: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b25c5f1f537457cac1af592fbf41a27 : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd2ce1071782471ca6f75f5d81e03f23 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53ea76767722475ca5e5fff7caedddce : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5c080b2fbb74ac69065cf1125ef6c88 : SUCCESS in 6m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3af8d098ce804a55b7f5fbba0a619e6d : SUCCESS in 6m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/aa9f5f75c0274696baf813fbcb3ad104 : SUCCESS in 5m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d0fe320e681046afb4fb9adf1f5f7f4a : SUCCESS in 9m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/94703852de544ca58e01eecfe0d0f475 : SUCCESS in 9m 43s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d6a0a7667d584fc3a1df411b5a013ba1 : SUCCESS in 10m 32s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a7c714a095624ff996af520258c99262 : SUCCESS in 7m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/64f52618d8774a1c8d883920893b4bc7 : SUCCESS in 12m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d8a7dacc848d402f9c1f0959faae105c : SUCCESS in 8m 40s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/141f0b25198b40cf9e279fce93e34488 : SUCCESS in 10m 00s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/56538c6b59e9421fa53a1773699229d6 : SUCCESS in 15m 57s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/aded28bb103f42599195e5ff3f2e104a : SUCCESS in 27m 41s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/464d8c8c93eb43689cdbd27c310a5e8b : SUCCESS in 14m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4595bc9186c249fc9c571238723b730f : SUCCESS in 5m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7b2c28bbdebd4343b3b82b8fc92ffd1c : SUCCESS in 19m 07s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/304305a8dd0e406b9c1ae5c57d7c4ee4 : SUCCESS in 6m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dd0f74ed1ca7491b8e272ac4b93fa4be : SUCCESS in 8m 11s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e83b1f7a4a4c4807add3f7e255a47c96 : SUCCESS in 11m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/59830d1454164c1795a468fa4e975b72 : SUCCESS in 58m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c9103155159d4bd3846dccb18d05e2bd : SUCCESS in 1h 20m 08s","accounts_in_message":[],"_revision_number":47},{"id":"582d17ca4182722f3ea3dfabc234b3432d202c10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 16:07:06.000000000","message":"Patch Set 47: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":47},{"id":"5282d620e0a2c4189121eb001ee8a507767594df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 21:21:30.000000000","message":"Patch Set 47: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b14a23f310b462d80b6869043110c2f : SUCCESS in 8m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69bdcfdc688f49ca96e567375062cb57 : SUCCESS in 8m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6eaff46624b24b5f9703e58d1c1133ac : SUCCESS in 7m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8df3f2a08da647a0927639165681c5e5 : SUCCESS in 8m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c514f318451d42c0868e1e91543a5b8a : SUCCESS in 7m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/68f0b21f23b94cdbb20bf3b06216637d : SUCCESS in 7m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2c6439e3d7194e3485d820a4e561bc0c : SUCCESS in 10m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5b904c2fd5a04cff88cea251c2be720f : SUCCESS in 11m 54s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/12f2891b78424afb81894ba454cc487c : SUCCESS in 13m 25s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9560a1a215174b3babed21bfe5e55031 : SUCCESS in 8m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/52c09077f63d4b39b443fa90ecb86f09 : SUCCESS in 12m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/523ce57abc6248ea9efbc967f8460287 : SUCCESS in 13m 17s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/05d4c84431e64cf9a2b88022437a1fc4 : SUCCESS in 11m 30s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/833fcc5a0a8b4f28a17b64e0b4ac17bb : SUCCESS in 17m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d5f2f92eb66b497195a7657025f8178b : SUCCESS in 31m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2f07b213eb834afb868b27eccb0e96cd : SUCCESS in 17m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d837a513e0eb48979edc3376d6ffe1fb : SUCCESS in 7m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/306d40e1a1c24df080dea453d1bee24f : SUCCESS in 20m 03s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4c01d25c4c12430780655dda8be5ddd7 : SUCCESS in 8m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f78a850f1b6440078f0fb305f11c9dbd : SUCCESS in 9m 54s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4f4c4b2f6e2f48eeba5f77302b01b6ae : SUCCESS in 14m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/738ed145962b4eeaa1f954a3961c7fe1 : SUCCESS in 1h 03m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e5e13b10ab0a44a0a1bf070a2d044f0d : FAILURE in 1h 20m 51s","accounts_in_message":[],"_revision_number":47},{"id":"d53d7564dfcff6f3e193c538aa751f99e42bec71","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-22 20:07:18.000000000","message":"Patch Set 47:\n\nrecheck","accounts_in_message":[],"_revision_number":47},{"id":"0d09000c9294909d40c4ad34bec9c639208a3162","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 21:04:49.000000000","message":"Patch Set 47:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":47},{"id":"5951c25e727169b62f5edfe200134d721a5c8fb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 23:05:41.000000000","message":"Patch Set 47: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e7d4a327c414aa1a3d846cd642b96b4 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebb71e7d717b442f90815aa69b238460 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/174f01f30d71446b8242d18aca2cf601 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c9eb1cf10f04d9bb573ec727f5742b6 : SUCCESS in 6m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4f424428bfc14d409d4247c5e4ba570c : SUCCESS in 5m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/00004cbd5083478c9f29bc72d21c1d6b : SUCCESS in 5m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/721f5a0192bb4be9962d580de007bf33 : SUCCESS in 8m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4c88849af6854ada8d168d6a81a19273 : SUCCESS in 9m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5eb29dfd5ce040239ac9f1ba36e46847 : SUCCESS in 12m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f3a7cd8dd71d4476a907ad3150add8f8 : SUCCESS in 6m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/65efc5e5a2ae4922bf73905ac67f5591 : SUCCESS in 12m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b87cb9e26c7a4b8c95b77da0f443d32e : SUCCESS in 9m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/18dba5730f9e4de1bb648ed0af6ed7a5 : SUCCESS in 9m 16s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/896e522ad99d433cbedd0b8a2f02681c : SUCCESS in 15m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/efbdc3638bda4c969e80d76520fac4f5 : SUCCESS in 27m 58s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f1feeeca989d47f6abdefb103468f5fa : SUCCESS in 13m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9178a738647d46e89c39d1d96aafbc6a : SUCCESS in 4m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0198a144439b47aeb3010aa930fdea72 : SUCCESS in 22m 10s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/66040741db614b3a9cd82dfed4014f61 : SUCCESS in 7m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8d53f8d0e0a54c94827c20d7cd9fdde0 : SUCCESS in 7m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/39c0f21b8b5e4fadb39dfa4b8c118c76 : SUCCESS in 12m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/23b11bffff934045ab98b67138156117 : SUCCESS in 57m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1c4cfbf63a294b9cb9a51311b0c0cfa0 : SUCCESS in 1h 21m 17s","accounts_in_message":[],"_revision_number":47},{"id":"2e0cf9e71d5684da3e714f11588abd4ad1a51bf7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 23:06:42.000000000","message":"Patch Set 47: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":47},{"id":"1796445a20c8be452bb5f9cabc7ed4a03c9b381d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 07:08:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":47},{"id":"513432a568535f3927bdf93596efb1340a6662c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 07:08:50.000000000","message":"Patch Set 47: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e3a8399b6f546acb15febefba8a8f51 : SUCCESS in 7m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81222d3da1184f48909ddab9e6a9808b : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b89c9f587474216b3142cf20fe356b4 : SUCCESS in 7m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76d8bcb69eea4bb9b7643d0e2815b715 : SUCCESS in 6m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3f3078491d9c4ab68fa7e98a7bf93c53 : SUCCESS in 6m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/39359603e7a042f2a29bd48b5b30f026 : SUCCESS in 6m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/36bc9a71568646889cb67e5f596c6554 : SUCCESS in 11m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ad363428e9cf499bb06560d0c5103f15 : SUCCESS in 10m 23s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6ef5996ba2bd48b2990b05f791e39440 : SUCCESS in 11m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2910c9d11e474efdb472d7b6570e5aac : SUCCESS in 8m 06s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/092770ccec6d4e52a0bb2ab4ef367399 : SUCCESS in 13m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7910224db4584e07ab03e5a1231d6eab : SUCCESS in 9m 27s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b2d30d31b1d4447cac0fd63b59d6bc75 : SUCCESS in 10m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d8e2b3fbe8a74ed7b14ca52168eba844 : SUCCESS in 15m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1fee85ce33394c3ba1a84a88e66780f4 : SUCCESS in 26m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0cc2204b32ef40329fce420afba61081 : SUCCESS in 14m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7aaa874f1d6e43588f73e2855fc3b067 : SUCCESS in 6m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/401b6023c96646b1968c8b18d966ce93 : SUCCESS in 20m 42s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e8f8ff34e818450f9e23971c6d895e1a : SUCCESS in 7m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f69d8533de434c4d844b76fd61b1ec83 : SUCCESS in 10m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/af0553a5041443d2a1c007fd5bc71e4a : SUCCESS in 13m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/163844c14dff4504bc74a4ba3b0c4b17 : SUCCESS in 56m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/81f1f400b43b4e22bdae3ce1c2de92a8 : SUCCESS in 1h 29m 23s","accounts_in_message":[],"_revision_number":47},{"id":"f8b579515501a59d49952fa0f739387fb728842e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 07:11:04.000000000","message":"Patch Set 47:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67dff955ea6342fabee786b9da95875a : SUCCESS in 1m 46s","accounts_in_message":[],"_revision_number":47},{"id":"c02557fd3c7f00c996f645ab9da8df1b8bc7062f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-23 21:57:54.000000000","message":"Patch Set 47:\n\nrecheck","accounts_in_message":[],"_revision_number":47}],"current_revision_number":47,"current_revision":"d8d2e848ec057c47963d33dbd44479fc28e6f38b","revisions":{"d149e15ac5245817c6e2a8a5572718fb7ddcbc4f":{"kind":"REWORK","_number":1,"created":"2020-09-06 06:16:41.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/1"}}},"commit":{"parents":[{"commit":"d2c8b9d472320a8e916f38df4580a7f7bbab2bda","subject":"Merge \"update constraint for ovsdbapp to new release 1.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d2c8b9d472320a8e916f38df4580a7f7bbab2bda"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-06 06:16:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-06 06:16:36.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d149e15ac5245817c6e2a8a5572718fb7ddcbc4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d149e15ac5245817c6e2a8a5572718fb7ddcbc4f"}]},"branch":"refs/heads/master"},"9af2a2295a2499eb4df728f5768f600ef49357f8":{"kind":"REWORK","_number":2,"created":"2020-09-07 06:16:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/2"}}},"commit":{"parents":[{"commit":"d2c8b9d472320a8e916f38df4580a7f7bbab2bda","subject":"Merge \"update constraint for ovsdbapp to new release 1.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d2c8b9d472320a8e916f38df4580a7f7bbab2bda"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-07 06:16:13.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-07 06:16:13.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9af2a2295a2499eb4df728f5768f600ef49357f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9af2a2295a2499eb4df728f5768f600ef49357f8"}]},"branch":"refs/heads/master"},"33f8193f9b54c1a700e555fb70b4def970c71d3a":{"kind":"REWORK","_number":3,"created":"2020-09-08 06:27:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/3"}}},"commit":{"parents":[{"commit":"d2c8b9d472320a8e916f38df4580a7f7bbab2bda","subject":"Merge \"update constraint for ovsdbapp to new release 1.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d2c8b9d472320a8e916f38df4580a7f7bbab2bda"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-08 06:27:23.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-08 06:27:23.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/33f8193f9b54c1a700e555fb70b4def970c71d3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/33f8193f9b54c1a700e555fb70b4def970c71d3a"}]},"branch":"refs/heads/master"},"efafd0e7779821a2a8a201cb641189e479102916":{"kind":"REWORK","_number":4,"created":"2020-09-09 06:31:02.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/4"}}},"commit":{"parents":[{"commit":"85d17024d4296ffdc9d6b8301fc41ff773d73764","subject":"update constraint for python-cloudkittyclient to new release 4.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/85d17024d4296ffdc9d6b8301fc41ff773d73764"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-09 06:30:58.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-09 06:30:58.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/efafd0e7779821a2a8a201cb641189e479102916"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/efafd0e7779821a2a8a201cb641189e479102916"}]},"branch":"refs/heads/master"},"7251385a083e7f62a6d911f6b2cdd6a52b6efe36":{"kind":"REWORK","_number":5,"created":"2020-09-10 06:25:16.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/5"}}},"commit":{"parents":[{"commit":"c55c411970f0f8fc450d943dc6f4c1ea07f8dc1a","subject":"Merge \"update constraint for os-brick to new release 4.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c55c411970f0f8fc450d943dc6f4c1ea07f8dc1a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-10 06:25:14.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-10 06:25:14.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7251385a083e7f62a6d911f6b2cdd6a52b6efe36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7251385a083e7f62a6d911f6b2cdd6a52b6efe36"}]},"branch":"refs/heads/master"},"998f3b0b85877620c0d195e8eb394bbfa67c5777":{"kind":"REWORK","_number":6,"created":"2020-09-11 06:34:41.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/6"}}},"commit":{"parents":[{"commit":"69f74c548ae4dbb7f980c15b199121f67d725005","subject":"update constraint for python-ironic-inspector-client to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/69f74c548ae4dbb7f980c15b199121f67d725005"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-11 06:34:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-11 06:34:39.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/998f3b0b85877620c0d195e8eb394bbfa67c5777"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/998f3b0b85877620c0d195e8eb394bbfa67c5777"}]},"branch":"refs/heads/master"},"aa67ae1769037c9d4602dc28bcb27368130fa1bb":{"kind":"REWORK","_number":7,"created":"2020-09-11 08:50:18.000000000","uploader":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"ref":"refs/changes/84/750084/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/7"}}},"commit":{"parents":[{"commit":"69f74c548ae4dbb7f980c15b199121f67d725005","subject":"update constraint for python-ironic-inspector-client to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/69f74c548ae4dbb7f980c15b199121f67d725005"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-11 06:34:39.000000000","tz":0},"committer":{"name":"Sébastien Boyron","email":"sboyron@redhat.com","date":"2020-09-11 08:50:01.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aa67ae1769037c9d4602dc28bcb27368130fa1bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aa67ae1769037c9d4602dc28bcb27368130fa1bb"}]},"branch":"refs/heads/master"},"6d24a91a2d20270cb6a6936c592a8a95bfb2f7b0":{"kind":"REWORK","_number":8,"created":"2020-09-12 06:26:57.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/8"}}},"commit":{"parents":[{"commit":"9d7cf2ced23eeaf1a8690f9cefa22c6423c476a4","subject":"Merge \"update constraint for python-manilaclient to new release 2.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9d7cf2ced23eeaf1a8690f9cefa22c6423c476a4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-12 06:26:55.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-12 06:26:55.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6d24a91a2d20270cb6a6936c592a8a95bfb2f7b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6d24a91a2d20270cb6a6936c592a8a95bfb2f7b0"}]},"branch":"refs/heads/master"},"b861b2ecb02129547ee89271775144973624541f":{"kind":"REWORK","_number":9,"created":"2020-09-13 06:17:29.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/9"}}},"commit":{"parents":[{"commit":"77bfaa5ae499e37528010f47af112007663a9156","subject":"Merge \"update constraint for oslo.versionedobjects to new release 2.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/77bfaa5ae499e37528010f47af112007663a9156"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-13 06:17:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-13 06:17:24.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b861b2ecb02129547ee89271775144973624541f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b861b2ecb02129547ee89271775144973624541f"}]},"branch":"refs/heads/master"},"69aa3e22743ac99c5c5792d852ca1f67aa2d14e8":{"kind":"REWORK","_number":10,"created":"2020-09-14 06:19:36.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/10"}}},"commit":{"parents":[{"commit":"77bfaa5ae499e37528010f47af112007663a9156","subject":"Merge \"update constraint for oslo.versionedobjects to new release 2.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/77bfaa5ae499e37528010f47af112007663a9156"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-14 06:19:33.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-14 06:19:33.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/69aa3e22743ac99c5c5792d852ca1f67aa2d14e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/69aa3e22743ac99c5c5792d852ca1f67aa2d14e8"}]},"branch":"refs/heads/master"},"dc8b0da6264da93b1317f6f90a9f9046e843f3ed":{"kind":"REWORK","_number":11,"created":"2020-09-15 06:30:50.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/11"}}},"commit":{"parents":[{"commit":"f38aa325e5fa08c24a0a6307a74588ec972bf9c1","subject":"Merge \"update constraint for python-senlinclient to new release 2.1.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f38aa325e5fa08c24a0a6307a74588ec972bf9c1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-15 06:30:45.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-15 06:30:45.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dc8b0da6264da93b1317f6f90a9f9046e843f3ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dc8b0da6264da93b1317f6f90a9f9046e843f3ed"}]},"branch":"refs/heads/master"},"b2448d819270a106d6c50b8889b602a94f100fec":{"kind":"REWORK","_number":12,"created":"2020-09-16 06:30:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/12"}}},"commit":{"parents":[{"commit":"f38aa325e5fa08c24a0a6307a74588ec972bf9c1","subject":"Merge \"update constraint for python-senlinclient to new release 2.1.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f38aa325e5fa08c24a0a6307a74588ec972bf9c1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-16 06:30:51.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-16 06:30:51.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b2448d819270a106d6c50b8889b602a94f100fec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b2448d819270a106d6c50b8889b602a94f100fec"}]},"branch":"refs/heads/master"},"3d8598f49fa5c4dd8ec84b2e16288512e6554da3":{"kind":"REWORK","_number":13,"created":"2020-09-17 06:33:12.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/13"}}},"commit":{"parents":[{"commit":"b120224d74f31de54c46ae7fdf5541c415889e9b","subject":"Merge \"update constraint for python-tackerclient to new release 1.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b120224d74f31de54c46ae7fdf5541c415889e9b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-17 06:33:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-17 06:33:08.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3d8598f49fa5c4dd8ec84b2e16288512e6554da3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3d8598f49fa5c4dd8ec84b2e16288512e6554da3"}]},"branch":"refs/heads/master"},"c19adfca289d31de1c25405c9e3b00ced8f7ca49":{"kind":"REWORK","_number":14,"created":"2020-09-18 06:26:35.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/14"}}},"commit":{"parents":[{"commit":"7482d3a1f111ca0ae813ffb9648b220583146017","subject":"Merge \"update constraint for tripleo-common to new release 12.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7482d3a1f111ca0ae813ffb9648b220583146017"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-18 06:26:32.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-18 06:26:32.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c19adfca289d31de1c25405c9e3b00ced8f7ca49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c19adfca289d31de1c25405c9e3b00ced8f7ca49"}]},"branch":"refs/heads/master"},"17e1e43e5772f3dae4cb763914a676c93361b741":{"kind":"REWORK","_number":15,"created":"2020-09-19 06:15:49.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/15"}}},"commit":{"parents":[{"commit":"41d9437e24935d18d811e5640573187cde3aa2cf","subject":"Merge \"update constraint for tempest to new release 25.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/41d9437e24935d18d811e5640573187cde3aa2cf"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-19 06:15:47.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-19 06:15:47.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/17e1e43e5772f3dae4cb763914a676c93361b741"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/17e1e43e5772f3dae4cb763914a676c93361b741"}]},"branch":"refs/heads/master"},"63b08c6935796ffcbe3bbc6a30b320f49ea33d40":{"kind":"REWORK","_number":16,"created":"2020-09-20 06:33:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/16"}}},"commit":{"parents":[{"commit":"41d9437e24935d18d811e5640573187cde3aa2cf","subject":"Merge \"update constraint for tempest to new release 25.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/41d9437e24935d18d811e5640573187cde3aa2cf"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-20 06:33:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-20 06:33:35.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/63b08c6935796ffcbe3bbc6a30b320f49ea33d40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/63b08c6935796ffcbe3bbc6a30b320f49ea33d40"}]},"branch":"refs/heads/master"},"24c5874c4a064cd7cf5cb44d9ce1a57bfdcef2de":{"kind":"REWORK","_number":17,"created":"2020-09-21 06:24:44.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/17"}}},"commit":{"parents":[{"commit":"41d9437e24935d18d811e5640573187cde3aa2cf","subject":"Merge \"update constraint for tempest to new release 25.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/41d9437e24935d18d811e5640573187cde3aa2cf"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-21 06:24:42.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-21 06:24:42.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/24c5874c4a064cd7cf5cb44d9ce1a57bfdcef2de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/24c5874c4a064cd7cf5cb44d9ce1a57bfdcef2de"}]},"branch":"refs/heads/master"},"84b6393f198bf97053bbb7c9fe4e44a301fb51fd":{"kind":"REWORK","_number":18,"created":"2020-09-22 06:20:13.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/18"}}},"commit":{"parents":[{"commit":"41d9437e24935d18d811e5640573187cde3aa2cf","subject":"Merge \"update constraint for tempest to new release 25.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/41d9437e24935d18d811e5640573187cde3aa2cf"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-22 06:20:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-22 06:20:08.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/84b6393f198bf97053bbb7c9fe4e44a301fb51fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/84b6393f198bf97053bbb7c9fe4e44a301fb51fd"}]},"branch":"refs/heads/master"},"3929aff534af266f756a9551fbd2b309fdbf073b":{"kind":"REWORK","_number":19,"created":"2020-09-23 06:15:19.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/19"}}},"commit":{"parents":[{"commit":"41d9437e24935d18d811e5640573187cde3aa2cf","subject":"Merge \"update constraint for tempest to new release 25.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/41d9437e24935d18d811e5640573187cde3aa2cf"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-23 06:15:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-23 06:15:17.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3929aff534af266f756a9551fbd2b309fdbf073b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3929aff534af266f756a9551fbd2b309fdbf073b"}]},"branch":"refs/heads/master"},"d6575af33fbf8fbc070f1b1ec73ea10f7dc4fe09":{"kind":"REWORK","_number":20,"created":"2020-09-24 06:22:27.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/20"}}},"commit":{"parents":[{"commit":"41d9437e24935d18d811e5640573187cde3aa2cf","subject":"Merge \"update constraint for tempest to new release 25.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/41d9437e24935d18d811e5640573187cde3aa2cf"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-24 06:22:25.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-24 06:22:25.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d6575af33fbf8fbc070f1b1ec73ea10f7dc4fe09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d6575af33fbf8fbc070f1b1ec73ea10f7dc4fe09"}]},"branch":"refs/heads/master"},"09277494a6f7dfc3dd7e9bcb78fc21cd5ba323ca":{"kind":"REWORK","_number":21,"created":"2020-09-25 06:35:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/21"}}},"commit":{"parents":[{"commit":"854bbb43d8d605fdb0f472ee9412bee4f7fb7686","subject":"Merge \"update constraint for openstacksdk to new release 0.50.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/854bbb43d8d605fdb0f472ee9412bee4f7fb7686"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-25 06:35:34.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-25 06:35:34.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/09277494a6f7dfc3dd7e9bcb78fc21cd5ba323ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/09277494a6f7dfc3dd7e9bcb78fc21cd5ba323ca"}]},"branch":"refs/heads/master"},"0e4412b02f2994f8ab3ef084aa55dcf49c930df4":{"kind":"REWORK","_number":22,"created":"2020-09-26 06:13:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/22"}}},"commit":{"parents":[{"commit":"854bbb43d8d605fdb0f472ee9412bee4f7fb7686","subject":"Merge \"update constraint for openstacksdk to new release 0.50.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/854bbb43d8d605fdb0f472ee9412bee4f7fb7686"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-26 06:13:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-26 06:13:50.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e4412b02f2994f8ab3ef084aa55dcf49c930df4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e4412b02f2994f8ab3ef084aa55dcf49c930df4"}]},"branch":"refs/heads/master"},"5f7dac5ab11a051b0618c245e700ef28a8f26635":{"kind":"REWORK","_number":23,"created":"2020-09-27 06:31:11.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/23"}}},"commit":{"parents":[{"commit":"854bbb43d8d605fdb0f472ee9412bee4f7fb7686","subject":"Merge \"update constraint for openstacksdk to new release 0.50.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/854bbb43d8d605fdb0f472ee9412bee4f7fb7686"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-27 06:31:09.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-27 06:31:09.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5f7dac5ab11a051b0618c245e700ef28a8f26635"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5f7dac5ab11a051b0618c245e700ef28a8f26635"}]},"branch":"refs/heads/master"},"20d25360407163df10f69f5b3a828f829886eec4":{"kind":"REWORK","_number":24,"created":"2020-09-29 06:33:29.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/24"}}},"commit":{"parents":[{"commit":"854bbb43d8d605fdb0f472ee9412bee4f7fb7686","subject":"Merge \"update constraint for openstacksdk to new release 0.50.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/854bbb43d8d605fdb0f472ee9412bee4f7fb7686"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-29 06:33:26.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-29 06:33:26.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/20d25360407163df10f69f5b3a828f829886eec4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/20d25360407163df10f69f5b3a828f829886eec4"}]},"branch":"refs/heads/master"},"6cdd3f9c08fe0e8836318ae2d3b099bc998098ea":{"kind":"REWORK","_number":25,"created":"2020-09-30 06:30:18.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/25"}}},"commit":{"parents":[{"commit":"346f659eec09403fcfa1897417347f99bf41b606","subject":"update constraint for tripleo-common to new release 12.7.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/346f659eec09403fcfa1897417347f99bf41b606"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-30 06:30:16.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-30 06:30:16.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6cdd3f9c08fe0e8836318ae2d3b099bc998098ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6cdd3f9c08fe0e8836318ae2d3b099bc998098ea"}]},"branch":"refs/heads/master"},"00440198e310331105fd6032f1951e8dd3c054f3":{"kind":"REWORK","_number":26,"created":"2020-10-01 06:32:07.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/26"}}},"commit":{"parents":[{"commit":"346f659eec09403fcfa1897417347f99bf41b606","subject":"update constraint for tripleo-common to new release 12.7.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/346f659eec09403fcfa1897417347f99bf41b606"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-01 06:32:05.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-01 06:32:05.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/00440198e310331105fd6032f1951e8dd3c054f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/00440198e310331105fd6032f1951e8dd3c054f3"}]},"branch":"refs/heads/master"},"376cc833d96a1d8255342c90a23e130f90d7d897":{"kind":"REWORK","_number":27,"created":"2020-10-02 06:27:25.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/27"}}},"commit":{"parents":[{"commit":"346f659eec09403fcfa1897417347f99bf41b606","subject":"update constraint for tripleo-common to new release 12.7.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/346f659eec09403fcfa1897417347f99bf41b606"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-02 06:27:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-02 06:27:24.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/376cc833d96a1d8255342c90a23e130f90d7d897"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/376cc833d96a1d8255342c90a23e130f90d7d897"}]},"branch":"refs/heads/master"},"f5b8f58182e9e661eddca748b065110f3b8b6281":{"kind":"REWORK","_number":28,"created":"2020-10-03 06:19:13.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/28"}}},"commit":{"parents":[{"commit":"346f659eec09403fcfa1897417347f99bf41b606","subject":"update constraint for tripleo-common to new release 12.7.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/346f659eec09403fcfa1897417347f99bf41b606"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-03 06:19:11.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-03 06:19:11.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f5b8f58182e9e661eddca748b065110f3b8b6281"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f5b8f58182e9e661eddca748b065110f3b8b6281"}]},"branch":"refs/heads/master"},"d8697e7bee24218fd8fe14fec85540e76de7b5e5":{"kind":"REWORK","_number":29,"created":"2020-10-04 06:22:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/29"}}},"commit":{"parents":[{"commit":"346f659eec09403fcfa1897417347f99bf41b606","subject":"update constraint for tripleo-common to new release 12.7.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/346f659eec09403fcfa1897417347f99bf41b606"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-04 06:22:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-04 06:22:02.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d8697e7bee24218fd8fe14fec85540e76de7b5e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d8697e7bee24218fd8fe14fec85540e76de7b5e5"}]},"branch":"refs/heads/master"},"1498ef502b1c15386c9b409dbaabf225d1eb7e09":{"kind":"REWORK","_number":30,"created":"2020-10-05 06:16:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/30"}}},"commit":{"parents":[{"commit":"fc679c617443a8f99601c67454842dd51d02b94c","subject":"Merge \"update constraint for ceilometer to new release 15.0.0.0rc1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fc679c617443a8f99601c67454842dd51d02b94c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-05 06:16:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-05 06:16:48.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1498ef502b1c15386c9b409dbaabf225d1eb7e09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1498ef502b1c15386c9b409dbaabf225d1eb7e09"}]},"branch":"refs/heads/master"},"344f2b7c09898c56cbfcd8ae3d04d04914f32dbc":{"kind":"REWORK","_number":31,"created":"2020-10-06 06:22:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/31"}}},"commit":{"parents":[{"commit":"fc679c617443a8f99601c67454842dd51d02b94c","subject":"Merge \"update constraint for ceilometer to new release 15.0.0.0rc1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fc679c617443a8f99601c67454842dd51d02b94c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-06 06:22:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-06 06:22:31.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/344f2b7c09898c56cbfcd8ae3d04d04914f32dbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/344f2b7c09898c56cbfcd8ae3d04d04914f32dbc"}]},"branch":"refs/heads/master"},"1e6fadf1ae5b09855ef00c11689f26ffff7ea14b":{"kind":"REWORK","_number":32,"created":"2020-10-07 06:18:29.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/32"}}},"commit":{"parents":[{"commit":"486666ac75ac13160a47b7b6262fc62f8594fe2d","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/486666ac75ac13160a47b7b6262fc62f8594fe2d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-07 06:18:25.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-07 06:18:25.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1e6fadf1ae5b09855ef00c11689f26ffff7ea14b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1e6fadf1ae5b09855ef00c11689f26ffff7ea14b"}]},"branch":"refs/heads/master"},"290726ac93a9bf0e091bc399211f6a4381309e97":{"kind":"REWORK","_number":33,"created":"2020-10-08 06:21:07.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/33"}}},"commit":{"parents":[{"commit":"486666ac75ac13160a47b7b6262fc62f8594fe2d","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/486666ac75ac13160a47b7b6262fc62f8594fe2d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-08 06:21:05.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-08 06:21:05.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/290726ac93a9bf0e091bc399211f6a4381309e97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/290726ac93a9bf0e091bc399211f6a4381309e97"}]},"branch":"refs/heads/master"},"a5087301f3d08fa1df9c76bfb4db8ba8ca64f934":{"kind":"REWORK","_number":34,"created":"2020-10-09 06:26:22.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/34"}}},"commit":{"parents":[{"commit":"486666ac75ac13160a47b7b6262fc62f8594fe2d","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/486666ac75ac13160a47b7b6262fc62f8594fe2d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-09 06:26:20.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-09 06:26:20.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a5087301f3d08fa1df9c76bfb4db8ba8ca64f934"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a5087301f3d08fa1df9c76bfb4db8ba8ca64f934"}]},"branch":"refs/heads/master"},"bc30c7e1fe4e75cc9b8a347f137535cd05921778":{"kind":"REWORK","_number":35,"created":"2020-10-10 06:43:08.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/35"}}},"commit":{"parents":[{"commit":"486666ac75ac13160a47b7b6262fc62f8594fe2d","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/486666ac75ac13160a47b7b6262fc62f8594fe2d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-10 06:43:07.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-10 06:43:07.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bc30c7e1fe4e75cc9b8a347f137535cd05921778"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bc30c7e1fe4e75cc9b8a347f137535cd05921778"}]},"branch":"refs/heads/master"},"c8179749dcf3e481bc1d7ad78a0ab1c8ce59ae00":{"kind":"REWORK","_number":36,"created":"2020-10-11 06:13:06.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/36"}}},"commit":{"parents":[{"commit":"486666ac75ac13160a47b7b6262fc62f8594fe2d","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/486666ac75ac13160a47b7b6262fc62f8594fe2d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-11 06:13:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-11 06:13:04.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c8179749dcf3e481bc1d7ad78a0ab1c8ce59ae00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c8179749dcf3e481bc1d7ad78a0ab1c8ce59ae00"}]},"branch":"refs/heads/master"},"1592b5869e3e9c0b85215a16232ad1005fc0e254":{"kind":"REWORK","_number":37,"created":"2020-10-12 06:27:41.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/37"}}},"commit":{"parents":[{"commit":"486666ac75ac13160a47b7b6262fc62f8594fe2d","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/486666ac75ac13160a47b7b6262fc62f8594fe2d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-12 06:27:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-12 06:27:39.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1592b5869e3e9c0b85215a16232ad1005fc0e254"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1592b5869e3e9c0b85215a16232ad1005fc0e254"}]},"branch":"refs/heads/master"},"f2b5b0349a75330e3d142b0e229650311f375f42":{"kind":"REWORK","_number":38,"created":"2020-10-13 06:16:14.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/38"}}},"commit":{"parents":[{"commit":"486666ac75ac13160a47b7b6262fc62f8594fe2d","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/486666ac75ac13160a47b7b6262fc62f8594fe2d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-13 06:16:12.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-13 06:16:12.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f2b5b0349a75330e3d142b0e229650311f375f42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f2b5b0349a75330e3d142b0e229650311f375f42"}]},"branch":"refs/heads/master"},"4bb3d1cf7dd06382552b9b4f24a6b93445649cd2":{"kind":"REWORK","_number":39,"created":"2020-10-14 06:13:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/39"}}},"commit":{"parents":[{"commit":"486666ac75ac13160a47b7b6262fc62f8594fe2d","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/486666ac75ac13160a47b7b6262fc62f8594fe2d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-14 06:13:27.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-14 06:13:27.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4bb3d1cf7dd06382552b9b4f24a6b93445649cd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4bb3d1cf7dd06382552b9b4f24a6b93445649cd2"}]},"branch":"refs/heads/master"},"72f00c7300443b9c72c935439a40aeab81a1fa3a":{"kind":"REWORK","_number":40,"created":"2020-10-15 06:20:41.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/40"}}},"commit":{"parents":[{"commit":"486666ac75ac13160a47b7b6262fc62f8594fe2d","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/486666ac75ac13160a47b7b6262fc62f8594fe2d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-15 06:20:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-15 06:20:39.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/72f00c7300443b9c72c935439a40aeab81a1fa3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/72f00c7300443b9c72c935439a40aeab81a1fa3a"}]},"branch":"refs/heads/master"},"9c374b74e9d6f09d25fd58dc2afe00c14ec6d080":{"kind":"REWORK","_number":41,"created":"2020-10-16 06:32:35.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/41"}}},"commit":{"parents":[{"commit":"486666ac75ac13160a47b7b6262fc62f8594fe2d","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/486666ac75ac13160a47b7b6262fc62f8594fe2d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-16 06:32:34.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-16 06:32:34.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9c374b74e9d6f09d25fd58dc2afe00c14ec6d080"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9c374b74e9d6f09d25fd58dc2afe00c14ec6d080"}]},"branch":"refs/heads/master"},"e9f95152a7a23559b59bbe077f7c356989118b5b":{"kind":"REWORK","_number":42,"created":"2020-10-17 06:22:03.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/42"}}},"commit":{"parents":[{"commit":"747ff02e2c3b9dc6bcc461731cec8758da7704a4","subject":"Merge \"update python-ironicclient to new release 4.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/747ff02e2c3b9dc6bcc461731cec8758da7704a4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-17 06:22:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-17 06:22:02.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e9f95152a7a23559b59bbe077f7c356989118b5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e9f95152a7a23559b59bbe077f7c356989118b5b"}]},"branch":"refs/heads/master"},"fbe60ab3478735aa6196f7ccd4418c717afc1822":{"kind":"REWORK","_number":43,"created":"2020-10-18 06:20:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/43"}}},"commit":{"parents":[{"commit":"747ff02e2c3b9dc6bcc461731cec8758da7704a4","subject":"Merge \"update python-ironicclient to new release 4.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/747ff02e2c3b9dc6bcc461731cec8758da7704a4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-18 06:20:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-18 06:20:50.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fbe60ab3478735aa6196f7ccd4418c717afc1822"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fbe60ab3478735aa6196f7ccd4418c717afc1822"}]},"branch":"refs/heads/master"},"c6217529620c9083bc4b5fcf78c59fbb59e101f8":{"kind":"REWORK","_number":44,"created":"2020-10-19 06:26:02.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/750084/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/44"}}},"commit":{"parents":[{"commit":"d447ce59fdd0c5ca992486af35dbd5d1008181a7","subject":"Merge \"update ceilometer to 15.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d447ce59fdd0c5ca992486af35dbd5d1008181a7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-19 06:26:00.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-19 06:26:00.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c6217529620c9083bc4b5fcf78c59fbb59e101f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c6217529620c9083bc4b5fcf78c59fbb59e101f8"}]},"branch":"refs/heads/master"},"6229a24e83737ab402e6ce9b42ad08f73d9effb1":{"kind":"REWORK","_number":45,"created":"2020-10-19 06:39:51.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/84/750084/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/45"}}},"commit":{"parents":[{"commit":"d447ce59fdd0c5ca992486af35dbd5d1008181a7","subject":"Merge \"update ceilometer to 15.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d447ce59fdd0c5ca992486af35dbd5d1008181a7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-19 06:26:00.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-10-19 06:39:37.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6229a24e83737ab402e6ce9b42ad08f73d9effb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6229a24e83737ab402e6ce9b42ad08f73d9effb1"}]},"branch":"refs/heads/master"},"7bb851d3866a567d378f5231590bd905fcc054b9":{"kind":"REWORK","_number":46,"created":"2020-10-19 08:49:01.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/84/750084/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/46"}}},"commit":{"parents":[{"commit":"d447ce59fdd0c5ca992486af35dbd5d1008181a7","subject":"Merge \"update ceilometer to 15.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d447ce59fdd0c5ca992486af35dbd5d1008181a7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-19 06:26:00.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-10-19 08:48:53.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7bb851d3866a567d378f5231590bd905fcc054b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7bb851d3866a567d378f5231590bd905fcc054b9"}]},"branch":"refs/heads/master"},"d8d2e848ec057c47963d33dbd44479fc28e6f38b":{"kind":"REWORK","_number":47,"created":"2020-10-19 13:36:08.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/84/750084/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/750084/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/750084/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/750084/47"}}},"commit":{"parents":[{"commit":"d447ce59fdd0c5ca992486af35dbd5d1008181a7","subject":"Merge \"update ceilometer to 15.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d447ce59fdd0c5ca992486af35dbd5d1008181a7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-19 06:26:00.000000000","tz":0},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-10-19 13:36:03.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1bc1fe988e3cfd4824dd7b4a9327a67db46c1533\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d8d2e848ec057c47963d33dbd44479fc28e6f38b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d8d2e848ec057c47963d33dbd44479fc28e6f38b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
