)]}'
{"id":"openstack%2Frequirements~750164","triplet_id":"openstack%2Frequirements~master~I315df6d0612df08cd9c99d373bc06156f3d27d5f","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I315df6d0612df08cd9c99d373bc06156f3d27d5f","subject":"update constraint for sushy to new release 3.4.0","status":"MERGED","created":"2020-09-07 12:34:09.000000000","updated":"2020-09-08 12:23:54.000000000","submitted":"2020-09-08 12:21:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"750164-1599567701624-f4e8b138","meta_rev_id":"03eca3f2217c1ba8d09bed321a7f69f91a5d07a0","_number":750164,"virtual_id_number":750164,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-08 12:21:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-09-07 17:38:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-09-07 13:11:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":2,"date":"2020-09-07 12:38:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-09-07 17:38:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-07 12:38:37.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-07 13:11:05.000000000","updated_by":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"reviewer":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"state":"REVIEWER"},{"updated":"2020-09-07 17:38:15.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-09-08 12:21:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"12ae42c29ee19bc5edf02b5777f38027d7260568","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-07 12:34:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2baa9f61b411d12c4b336dc65b171ef8dfa767cf","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-07 12:38:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"94eb5b80b868bd89871a5fa62aed9d6380f27aed","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-09-07 13:11:05.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"644f901c1503380b9b81c356560b73d4c3017cef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 13:50:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c917f87a590240cba99fa163fc674a44 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a9731f65da4473a96de25deedeca4ba : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6aa8b547c5164382bdbc16c71c246b70 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4cf751feb0354cfca31874ed82ecb151 : SUCCESS in 5m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ef54446b3907403eb26f7304221ea6c8 : SUCCESS in 4m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/031e7076cf1c4aacb30b1f6ad53da235 : SUCCESS in 3m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b926c09951d24353806cff0578cd76a2 : SUCCESS in 6m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/da6d6a17564b44d1a1c06f1e10bf9a48 : SUCCESS in 7m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0d2435b3e85d479fa0905c3140e2d5e9 : SUCCESS in 7m 31s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ddf30d17647d428cb1e1f4a9a9186a43 : SUCCESS in 5m 03s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1acc471fe4c94db2adbdebd141f68cbb : SUCCESS in 8m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cbf77de094e64b2b91fce6006acf2140 : SUCCESS in 7m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/13215d35207c4561b86c48558bb55dc0 : SUCCESS in 7m 20s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/21973f83a24041f0880df3b60fbe215c : SUCCESS in 19m 36s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f65967ad9f7a45ffb80180c397bc9796 : SUCCESS in 22m 59s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1734a8c151634adb9e5308a1857ebc9e : SUCCESS in 14m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/62e8485f309c4b6a9e679ab03ca59bb6 : SUCCESS in 3m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/48bc2ac257164c0e82f9110d7373deea : SUCCESS in 15m 31s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/39e72ef6993549e9a98efeeaa86fb0be : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ae4c5a2b6336443db311e6dc767ccb43 : SUCCESS in 7m 01s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4769cf62a22a40fb88960b52a94c720d : SUCCESS in 9m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b2164b716713459583d978fb348d622d : SUCCESS in 45m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4f363f3cdc4a48178e5b6173138dab70 : SUCCESS in 1h 12m 07s","accounts_in_message":[],"_revision_number":1},{"id":"f5bc3e2325423a75097c16834d73ef6916a5d9b5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 13:50:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6fea3f843c6c4e578864a79c66fda510 : SUCCESS in 1h 03m 20s","accounts_in_message":[],"_revision_number":1},{"id":"98e033aafd6fcefc143f0a150d61b4345fec0826","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-09-07 17:38:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"12a5e64b419ff0e084ea33c48bb8da146ac1bcba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 17:38:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fe6cb01ffaf58ab72eb63b44bc7703ceb07e96f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 19:04:02.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f293768c42814f92989b3754a622cedd : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94793e0d79794310a36cbb039e58a13e : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f11a73211104eb2bb3732ee6e3093d3 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2673ada694694f99a3ad476ea80c4a47 : SUCCESS in 4m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4b12d1fd839d49668b58362e3abd92cd : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/33b96c239e5a41fcbbd8e037825f63c6 : SUCCESS in 3m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cd3669384c28431e872c27447811f650 : SUCCESS in 5m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/714d0fc322be447fa63ff70c4774fca7 : SUCCESS in 6m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6c925f120ff54ad888ba1470a1f743ee : FAILURE in 6m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0b0bf6588e374626a10d86ddb64d3f64 : SUCCESS in 4m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/305eee3b9e864a6086ceca8b5b426de5 : SUCCESS in 6m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3ed1a33df2074614ab432205d1f3faed : SUCCESS in 5m 45s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2ba12d24cb504c8997d1f9f7e132d776 : SUCCESS in 6m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/28908cb1380c4ae78a739ba0063229eb : SUCCESS in 11m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d7e94a01243c4087a27efcebfa96bb09 : SUCCESS in 15m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3201ef8a5f574f04a695455553f31cff : SUCCESS in 9m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8e58005627954a16b9587b017adb74ea : SUCCESS in 3m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e2f872083da4412dbbc8b4c8d7b55697 : SUCCESS in 12m 26s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/54972538074a4747a97a56bb67ca56d7 : SUCCESS in 4m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5bb27b639f8c476cb56ece54db735846 : SUCCESS in 6m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1df99e5c5efa46a989b9e02761d847ec : SUCCESS in 8m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/37b015a6e4f247ebbd6eff0c31eb3c21 : SUCCESS in 1h 01m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7fd0c8db46f1453dafe0d222cdb563f6 : SUCCESS in 1h 24m 03s","accounts_in_message":[],"_revision_number":1},{"id":"2349e940d7cf0dfe77c09dc081c41734b24e0596","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-08 08:01:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6f1e9c5c76e0f3c44ef309b715db0736f55c1bcf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 09:10:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e122bfde36c44b95805aea4dc8d014b8 : SUCCESS in 1h 04m 25s","accounts_in_message":[],"_revision_number":1},{"id":"ce1dcfc9d1d65c495ef62ea9cc26d1c22d23fd86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 09:25:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17052986c1714efa930969d69fbf3e10 : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9eb04abc169465394a4e7820d98d8a3 : SUCCESS in 3m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8adda06fd754e5dbe43ce1a3410a82a : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97cbb919d125469e91da86f5d4f66025 : SUCCESS in 4m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d09c68c5f3c34f0f96f5b0420e5bfb85 : SUCCESS in 6m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/00df7eb61e29442db0b5c1d3bbebfbed : SUCCESS in 3m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c2719b38f96241c1bf578f76162fade8 : SUCCESS in 8m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/14f708aca851463eafa2e3118bcbeb35 : SUCCESS in 9m 54s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b8821f7706b14133864033b73387c4b1 : SUCCESS in 12m 31s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/30dc4ab8986d40a0860bf51d9a3145f3 : SUCCESS in 4m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bbaae864324a4d5ab235738654d7709d : SUCCESS in 10m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/de4f38ccee434bee93642bd5a260e901 : SUCCESS in 6m 13s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3bbc246dd31e45dc87c4159e6a45b194 : SUCCESS in 6m 10s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e78b384b49204165880cf38148914f3f : SUCCESS in 25m 18s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/83233f3935284b9aba1f51e924798a41 : SUCCESS in 25m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5e144549cd3a48539e49895bcd3217d9 : SUCCESS in 12m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/348b980eb80948f9bfe6987deba5c932 : SUCCESS in 3m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4c4aefa6db61479f880d3944d70a59fd : SUCCESS in 17m 41s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e9f5b53b00ee446d83427cd5c9006c79 : SUCCESS in 4m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8456f08def7a48e3908676151efbbc2c : SUCCESS in 7m 26s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5c8dc046c3f4470ba2fb8c5eec89da31 : SUCCESS in 10m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0fe86d4b90624383aecac082beb6437f : SUCCESS in 53m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/86dd7cd63ea24066b0a6b6e60f983341 : SUCCESS in 1h 19m 54s","accounts_in_message":[],"_revision_number":1},{"id":"891600fa3a522069d28b93f260ac48f5b5bb8eb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 09:26:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e519940c22ccfd43dc349a3d88d0c72a150f9b57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 12:21:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faa733750d9a4833a79301aedb04b608 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2a574fc23e7421f881e78fdd460c641 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0ba2ff4f35145ff8e039cfb3c269c35 : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23862ea0e56e428885b8b4af95c8286b : SUCCESS in 5m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9eeda05c4bda4dc198e9c41cf3572e08 : SUCCESS in 5m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d36d54cfe5954abab6175037d43e9374 : SUCCESS in 4m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c6aec66b27bc4dfd9e34cd9ef5c78db4 : SUCCESS in 9m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/db60daea9b6140a6940719abe9914425 : SUCCESS in 11m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/39d548c5d47f4411a77cbb43a0ff1deb : SUCCESS in 10m 57s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5a2e53f0a67e4d5e8841fff5a56cda2a : SUCCESS in 6m 45s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bbb3a31155ca46e88e910656b3643512 : SUCCESS in 10m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/48cdb2f26f1a428c8471d311fe0e80e2 : SUCCESS in 8m 07s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/826eda4d3e574a2caafe2f5c38c0b447 : SUCCESS in 9m 18s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/64b9386793dd4eefa88d2f1e7bd8346a : SUCCESS in 11m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ed04c34e788349e69ec60676a6748ed3 : SUCCESS in 25m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/76f07caff3064b4ab6de892dca33162f : SUCCESS in 11m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5e0b75d8f4a7460789c1066c04c79869 : SUCCESS in 3m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/954cea38c1c44ceeb21aef207f055502 : SUCCESS in 16m 19s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3459722b3efa4a26b94ea3fdd72c7a75 : SUCCESS in 6m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c09ebe11a0cb411bb9ab046c565c29d2 : SUCCESS in 7m 03s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b62ec489e90143b3b89a12f380ca064e : SUCCESS in 8m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2bd51b65bd04421a8daccae5476df301 : SUCCESS in 54m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/39197721757c4166b947324adfde204c : SUCCESS in 1h 37m 55s","accounts_in_message":[],"_revision_number":1},{"id":"278896ce2e17b8594f1d352751cab0c6a362b7b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 12:21:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"64ea73ed0eaa492772fee3bd0cca1c7beb433cc7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 12:23:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/095e9abf90464c509cb43365ad93e7c5 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"473cc2a6ab44d21c91bf1dba6572e075f01da4f7","revisions":{"473cc2a6ab44d21c91bf1dba6572e075f01da4f7":{"kind":"REWORK","_number":1,"created":"2020-09-07 12:34:09.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/64/750164/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/750164/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/750164/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/750164/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/750164/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/750164/1"}}},"commit":{"parents":[{"commit":"d2c8b9d472320a8e916f38df4580a7f7bbab2bda","subject":"Merge \"update constraint for ovsdbapp to new release 1.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d2c8b9d472320a8e916f38df4580a7f7bbab2bda"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-07 12:34:06.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-07 12:34:06.000000000","tz":0},"subject":"update constraint for sushy to new release 3.4.0","message":"update constraint for sushy to new release 3.4.0\n\nChange-Id: I315df6d0612df08cd9c99d373bc06156f3d27d5f\nmeta:version: 3.4.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Julia Kreger \u003cjuliaashleykreger@gmail.com\u003e\nmeta:release:Change-Id: I81144f983b64fd0ad2ac215b08fc9ce38a023559\nmeta:release:Code-Review+1: Julia Kreger \u003cjuliaashleykreger@gmail.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: wu.shiming \u003cwushiming@yovole.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/473cc2a6ab44d21c91bf1dba6572e075f01da4f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/473cc2a6ab44d21c91bf1dba6572e075f01da4f7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
