)]}'
{"id":"openstack%2Frequirements~750714","triplet_id":"openstack%2Frequirements~master~I2648ffa14121230085da5d6a37307fa0eb947ab4","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I2648ffa14121230085da5d6a37307fa0eb947ab4","subject":"update constraint for python-glanceclient to new release 3.2.2","status":"MERGED","created":"2020-09-09 16:27:32.000000000","updated":"2020-09-09 23:12:03.000000000","submitted":"2020-09-09 22:57:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"750714-1599692249372-0b22d573","meta_rev_id":"3dcee3f58454bcc56fb87a8067a392ee19d60d92","_number":750714,"virtual_id_number":750714,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-09 22:57:29.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-09-09 16:33:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-09-09 18:09:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-09-09 18:09:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-09 16:33:04.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-09 18:09:38.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-09 22:57:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"027d3106367228cee3e88006c85c164d920d855e","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-09 16:27:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"04e23890aa21d436a0811157dd4e7ac54d4387b9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-09 16:33:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"63b6c77b15265d25a6accc12c863992648925048","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 17:42:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/9bb184b992bc4ac0a647ebf95ef6bafb : SUCCESS in 1h 08m 47s","accounts_in_message":[],"_revision_number":1},{"id":"e946efeaacbee9d928f350e10fc6c0ae739b72de","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-09 18:09:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cb5bfe7f36a492f9443eb504d102f84ae1fca82d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 20:06:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e99351edeeb4610885ea311cba3b7bd : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c83ede218ad84b6a99ec30db1967ed84 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/051cb0f16f49405d81356741febf18ee : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d4ebfbe148445cfbc6541036f252d28 : SUCCESS in 7m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5b90ff60589a4daab2f699da20c04dc0 : SUCCESS in 6m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6619ecfc128a43ffa23bf5777eaab025 : SUCCESS in 6m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6ef3929b2e8f489bbc224818af5d26e6 : SUCCESS in 8m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e025e90c068749e881f323315c273aef : SUCCESS in 10m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1ae95d4f89794c01a65f8331d336705c : SUCCESS in 9m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9b1dadfbc2084d78bf70a47a05b8127c : SUCCESS in 6m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d97537b76a8b45e9bc452e4232511df7 : SUCCESS in 10m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/564c22035c62447ba9f7fa420d915918 : SUCCESS in 8m 09s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/41b81014f88743788a0f9ca5a274513c : SUCCESS in 7m 26s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5059c13d841a4cf981601efbd1e783ea : SUCCESS in 14m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/944bc63f724e4a138c297e08dc85f77b : SUCCESS in 26m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/469a4ee04086432cbb96c276b3cca07c : SUCCESS in 13m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c8bd9813fcba4b39b94cedf4548c872a : SUCCESS in 4m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/78600141e68b4ceb8f7c306a9e87a164 : SUCCESS in 21m 00s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/129607146e7b49d496f478ebafbacbd0 : SUCCESS in 6m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/593f305e68aa4348b48a9261c5fc0125 : SUCCESS in 7m 16s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ab511462050347039fcfe722133f42d2 : SUCCESS in 11m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cf847e690b534d6ba682eed099cd2ea7 : SUCCESS in 55m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8984846f066b47f18bdba1af638f99de : SUCCESS in 1h 27m 22s","accounts_in_message":[],"_revision_number":1},{"id":"a60730a99a692bfeed1cee64ec564ca1711f5a1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 20:06:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6a5eaefe45f1145ada8ba048d6a55e3f90a738db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 21:50:28.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d3c641d19163fb2fdc1772dc9fbf565c55e97cb3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 22:57:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0f3f1e8ef6b836c1a7121a9dd9fa16c00de92968","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 22:57:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3dc0e84ad3d433bbe2bc0dbfcc74cd5 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ae60fb133e0450781c595c3ab1bb2eb : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e84d4cf6c5b49cbb661f40f6d2e69b5 : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8bf0e7231c1443428c4381067c63069f : SUCCESS in 4m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/89aade8c9cd447349c007e4168c5f0ad : SUCCESS in 4m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9168d413dbd440b0891b19d663686ebb : SUCCESS in 5m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5a7c5b290d314e52aa95aef47173563e : SUCCESS in 9m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a2cf11c7df6944b0bb5df6488c14964e : SUCCESS in 7m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3b7828f0662240d6a4cac171d2420e4d : SUCCESS in 10m 02s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/393159f347604d8ab3c562bc1e053a98 : SUCCESS in 5m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ca944ecb2b50486fb71b137d7c0db8ff : SUCCESS in 11m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1d5c33e7c66e4afd920b1921afb55c33 : SUCCESS in 7m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/405ab334c651414b9ef7aeb857c6b36d : SUCCESS in 9m 49s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2e5547213dfb43569bafbda69f29115d : SUCCESS in 15m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cccf65df9d434104a9dd37ab1018028b : SUCCESS in 25m 41s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8ef3dba0b83c4ac3ad2deee034b7ec4e : SUCCESS in 14m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/364dfb2d22f341cd858cf512b014a060 : SUCCESS in 4m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6edaeec11755476cabb12600787c2a35 : SUCCESS in 17m 13s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e7bc4ed8be4040968487ea7d952e705d : SUCCESS in 6m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4e7e9fd17c724d54a4a34095c9b2406d : SUCCESS in 8m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/936bdd0b9e5a42aab40f0d4b9ef47467 : SUCCESS in 11m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/13f44778cb4444238dcde2e6da26dd89 : SUCCESS in 51m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5cab889b9bc649f18908c7f0e00a4658 : SUCCESS in 54m 28s","accounts_in_message":[],"_revision_number":1},{"id":"2aa99ed91845be831fe55624f4daa4033d813e66","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 23:12:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/068f5244b07c48849944eb608bdbd252 : SUCCESS in 1m 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e20c475f098aa38dd9ed51e0d41de024e50c8f15","revisions":{"e20c475f098aa38dd9ed51e0d41de024e50c8f15":{"kind":"REWORK","_number":1,"created":"2020-09-09 16:27:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/14/750714/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/14/750714/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/750714/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/750714/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/750714/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/14/750714/1"}}},"commit":{"parents":[{"commit":"85d17024d4296ffdc9d6b8301fc41ff773d73764","subject":"update constraint for python-cloudkittyclient to new release 4.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/85d17024d4296ffdc9d6b8301fc41ff773d73764"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-09 16:27:30.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-09 16:27:30.000000000","tz":0},"subject":"update constraint for python-glanceclient to new release 3.2.2","message":"update constraint for python-glanceclient to new release 3.2.2\n\nChange-Id: I2648ffa14121230085da5d6a37307fa0eb947ab4\nmeta:version: 3.2.2\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I66a77c7f4881b8c13fbbc64e03d8431316278d5d\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: Abhishek Kekane \u003cakekane@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e20c475f098aa38dd9ed51e0d41de024e50c8f15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e20c475f098aa38dd9ed51e0d41de024e50c8f15"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
