)]}'
{"id":"openstack%2Frequirements~750739","triplet_id":"openstack%2Frequirements~master~If7975abd6415777f5022ccbe2d43757c090c097c","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"If7975abd6415777f5022ccbe2d43757c090c097c","subject":"update constraint for python-ironicclient to new release 4.3.0","status":"MERGED","created":"2020-09-09 17:07:22.000000000","updated":"2020-09-09 23:35:06.000000000","submitted":"2020-09-09 23:32:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"750739-1599694356091-9fd2f9ed","meta_rev_id":"2bfa2e9d442e08c91e23082439e4a5bc4b5b293d","_number":750739,"virtual_id_number":750739,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-09 23:32:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-09-09 17:21:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-09-09 18:10:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-09-09 18:10:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-09 17:21:22.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-09 18:10:47.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-09 23:32:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"42187e82fe1aa45ddb75d9eb7eb1c54f4ac9b5a1","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-09 17:07:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"072c4beac937f1f79360b8ce934ca41450200995","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-09 17:21:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"02db9d3a3dd876e5e9116d69f5c573f8b51d81cf","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-09 18:10:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b4d2cfea91daf167e046769ee83dc7c4253a1ef1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 18:30:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/603f19b13dc74c1a9c292f868ce868d4 : SUCCESS in 1h 09m 40s","accounts_in_message":[],"_revision_number":1},{"id":"cc482538328d74b2d7313571b6cffd3b94af47b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 20:17:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48f972b28cb745f99c4f942aef63815d : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11381b0bee704b2981419c9735b1949d : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f932cd6161f440f3a31111d52e448a0d : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4ea200b6ed24261aa49debc8f33eab9 : SUCCESS in 6m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0290fc6e553d4b419f08c7b67ecd685a : SUCCESS in 3m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/21aa45fb33f544e1b80cebd8ba43eeff : SUCCESS in 4m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8abee30af2764511afc650d79194bdbc : SUCCESS in 7m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4fef575fd18b4102a2ecef6c0b9a863b : SUCCESS in 9m 06s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2e7634fdac944ee1a7df68c515e3d6e0 : SUCCESS in 10m 01s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/db5beaecc5764112965a1f056525fdeb : SUCCESS in 6m 09s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0bfe31f19e674e06a301b717c13546da : SUCCESS in 10m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/19c322f6360d4cddb26030b540e3187f : SUCCESS in 7m 28s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7f29761590c048bb8199d34e6a38d0c2 : SUCCESS in 7m 26s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c888b24e54894f20818297a3bd64aaf7 : SUCCESS in 12m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/de6739000825499ea49c125e1a5ef223 : SUCCESS in 26m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/37cda89372c14793a71e900f150c9dbf : SUCCESS in 15m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6d5b3fd2457344f988dc8a1647741fdf : SUCCESS in 3m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a49d1080b4344be7b11f9d34fd5a4a32 : SUCCESS in 15m 54s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f07f24af46c34b46a23bb5efd1bf280b : SUCCESS in 4m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/335cd359f9b04c5c923a586a4d91499b : SUCCESS in 5m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/92f9484a111e47d79d506a5013b52b13 : SUCCESS in 10m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7fb97d3d96824e57bd1834a5ad1924b5 : SUCCESS in 55m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/446776dd93014f1bb324b3cc21c1d1cf : SUCCESS in 1h 16m 20s","accounts_in_message":[],"_revision_number":1},{"id":"4ca33beae557bf83508a419b94d38c85456092d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 20:18:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"54935aff3f6b3162258ec894b4f99b1672126602","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 21:50:30.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fd646025d5f83a6beef33fe4671c33c77a9a2cdd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 23:32:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d3ed86b30974f50b11eeeff12c2295e : SUCCESS in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/666c942b92c14096aaf00c037557cb05 : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2da807ed4b9f4b75af73cf7c72d33061 : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02034747437d4a95921621a857ceb200 : SUCCESS in 5m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5dcbaec628a3445a8674e84718c288f4 : SUCCESS in 5m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4818d44bba9f4d4a9eb2c0cd00887460 : SUCCESS in 5m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/29e1e48023ce4192ac09d4fd5ac656a3 : SUCCESS in 7m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/71b48220de31406d9a233b50dad1203d : SUCCESS in 7m 25s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2124fe8ab60a422083eee56a773b0058 : SUCCESS in 8m 27s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1e3e0c6e5d0a4d5798aa8e30408c192c : SUCCESS in 6m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/136998c0c22549899a1ba479e4350c0e : SUCCESS in 8m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/41c7652622834c2a83535d9b1bb0a788 : SUCCESS in 7m 21s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/416e6618911a4fa395b9b60afc0efb4a : SUCCESS in 7m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2ab86efbdb404774a91051b8d9dad154 : SUCCESS in 12m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e11b3e6a554d4617a8ade58888150774 : SUCCESS in 28m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/46c851d6784a48a49be04c96bf38b4cf : SUCCESS in 13m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/50cbb74a553f413b83aa06c7c3ca77a6 : SUCCESS in 6m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/416d4348e6db44bfa5eac85e719fa395 : SUCCESS in 19m 48s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/27a77445faa74e04be3b456d2f0467fc : SUCCESS in 5m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7984eb31a0c64222bb7abb009e77210c : SUCCESS in 6m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2c5933391f8b48ce9722e2d8c7b1ed82 : SUCCESS in 12m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/62ce7ff2c870480f90dc34041bb3b14d : SUCCESS in 51m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b300419509bc4fd09a781149b7191678 : SUCCESS in 1h 27m 41s","accounts_in_message":[],"_revision_number":1},{"id":"11243d89b5ac96ec18c8ce0565bcc726567c5472","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 23:32:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"640bf211b69a63c792bb3db03bdd0e9aa861617c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 23:35:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07fe66da293b405b962889893c5bdae7 : SUCCESS in 1m 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f7a0a6aff0785d92d807d48563f7be9c086ab93d","revisions":{"f7a0a6aff0785d92d807d48563f7be9c086ab93d":{"kind":"REWORK","_number":1,"created":"2020-09-09 17:07:22.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/39/750739/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/39/750739/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/750739/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/750739/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/750739/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/39/750739/1"}}},"commit":{"parents":[{"commit":"85d17024d4296ffdc9d6b8301fc41ff773d73764","subject":"update constraint for python-cloudkittyclient to new release 4.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/85d17024d4296ffdc9d6b8301fc41ff773d73764"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-09 17:07:19.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-09 17:07:19.000000000","tz":0},"subject":"update constraint for python-ironicclient to new release 4.3.0","message":"update constraint for python-ironicclient to new release 4.3.0\n\nChange-Id: If7975abd6415777f5022ccbe2d43757c090c097c\nmeta:version: 4.3.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I4deec2aade22c268cdc9e5e914ca032154108089\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7a0a6aff0785d92d807d48563f7be9c086ab93d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7a0a6aff0785d92d807d48563f7be9c086ab93d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
