)]}'
{"id":"openstack%2Frequirements~750982","triplet_id":"openstack%2Frequirements~master~I3a83cfe519a36031be458f26ed6b0e6fc4c880b6","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I3a83cfe519a36031be458f26ed6b0e6fc4c880b6","subject":"update constraint for aodhclient to new release 2.1.1","status":"MERGED","created":"2020-09-10 12:59:31.000000000","updated":"2020-09-10 18:52:26.000000000","submitted":"2020-09-10 18:49:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"750982-1599763764409-c09e01d6","meta_rev_id":"46a39252c97be7141895305416472ff96799c4a4","_number":750982,"virtual_id_number":750982,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-10 18:49:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-09-10 13:02:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-09-10 13:03:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-09-10 13:03:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-10 13:02:19.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-10 13:03:17.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-10 18:49:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"239b0bcf97b498b33413cc1c9b33c339afd5fa85","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-10 12:59:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b65427179de0a543cfd8fb37799cc7f1d172a14b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-10 13:02:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"77ce7db31915d49cb1d2908197774691768dc7a3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-10 13:03:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c82cf2896b425bb11cc125271b43740e105557f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 14:14:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/983a1838e5014e32b3e883d1524cf84c : SUCCESS in 1h 06m 43s","accounts_in_message":[],"_revision_number":1},{"id":"8e3c743ce14e77fc04c93b51c7edb73376f87c85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 14:59:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97fe97c74b6b41259d64ac59e7fc415d : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d53ee09842e24163ac1c83f18699377c : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b3dd6ea246d437a830869930e5e80e6 : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb937e13a2ee425c88b2d7f185c14189 : SUCCESS in 5m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/99126272700f468290d87c84d8dd11b9 : SUCCESS in 5m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a435dc2dbd63484b835497149c1905b9 : SUCCESS in 5m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d6f8fbdbe3cf4f46ae8da6d7bc385d7f : SUCCESS in 8m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5a6c1ae4752e4de5863223bb43dbec74 : SUCCESS in 8m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/abc9fea9beea41e78e644f508af02890 : SUCCESS in 9m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c2df355ddc984ee68f258409f041f206 : SUCCESS in 6m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/104f8b4bd6234c2b8d25f8c6c208f36a : SUCCESS in 12m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c6d85ad1e39943a39db0bde27ecb3f4a : SUCCESS in 11m 47s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f3388237f1164287aa33084fba6bb092 : SUCCESS in 9m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/830f517aaaaa4bfeafc8dd203e34e806 : SUCCESS in 14m 00s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ead52f24cbbb45e49640e0a694a3bdb1 : SUCCESS in 27m 30s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f16ada06dee34dbf8a4612e15f70fef6 : SUCCESS in 20m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3de9191d5743474292143d2a59103b07 : SUCCESS in 4m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a0a094f0b2bd4c36a83a3d3509c6d11f : SUCCESS in 14m 25s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9c82e727b55a4ca1a979eee5a395574e : SUCCESS in 5m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a434a43ba65f488ba3675a1ca889372c : SUCCESS in 9m 16s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/84e2742a51c7438cbda806bd26964bf9 : SUCCESS in 11m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/746869af9a4345568402587e1696944a : SUCCESS in 59m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4c1632a26f83404196e3ee6a001e687e : SUCCESS in 1h 25m 04s","accounts_in_message":[],"_revision_number":1},{"id":"6b168f67c5a1a1457fd64a3954583301fa8c8159","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 15:00:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c2a9aba669142f313dbfc2590ee5a9c4642fdbe9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 18:49:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f59ab3c3cb3c49d3be90c226e7bbb15d : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5f7ac96d22b4866962544013d227d19 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7dcb2aa89977487f95ae5e3f402dd8af : SUCCESS in 6m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/536c1e708b7f4e3696c1ea4ad867746d : SUCCESS in 6m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c1f71d828b5d4f7e9cbac1ae542277c0 : SUCCESS in 5m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/631a8c54616a4af391c1cd93b7e31276 : SUCCESS in 7m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/00f12b9addc14212b0379bd739c36af8 : SUCCESS in 11m 32s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/57b97251268742b584ab1ced3b7c71ad : SUCCESS in 9m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9f2b3b9c54094b3386bed0110c354926 : SUCCESS in 11m 33s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ba2fbea52e4d4e59b6306d008679f6e4 : SUCCESS in 8m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/640e7926adc2483a8069fb58092eda7a : SUCCESS in 11m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7c34a208f5234edcb27178d20302f1dc : SUCCESS in 7m 36s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/53f329e6723d4e9cb42b0441acea0c8e : SUCCESS in 9m 09s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/30b750ee41844a5ba5b13a5d5ef7c284 : SUCCESS in 13m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f55b0bced15544f6a2f1b634c7847cc2 : SUCCESS in 24m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/838e203dd5da4b9d9e4de2991786d07f : SUCCESS in 12m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7c3780658eba4017b2e038bd6b43c351 : SUCCESS in 4m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2a2cdc6c84524f9fa5402b4f63dfb8ef : SUCCESS in 18m 03s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d1a9b87de6e14af1bcabeddcaa40abc2 : SUCCESS in 6m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d14c2b4832c14bc6b38d1f00d9ea2422 : SUCCESS in 8m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c34f8b95bcd04982a1461759d4eba08e : SUCCESS in 11m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a0e0ebbe71984ec1866a1cd380a5cb37 : SUCCESS in 47m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0ad483c9b7c54c6cb484c4b8375b3ac0 : SUCCESS in 1h 07m 27s","accounts_in_message":[],"_revision_number":1},{"id":"2bbde37df58a65f167e5e489af8341adb534fbee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 18:49:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f59d039446b0e5fb06766017c4a2180068dfd6d8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 18:52:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9f10434e3f64b1eaddcb41dfa169f1b : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f7dd81b6ded09d15749ddab46ec6bdda98a7f9d6","revisions":{"f7dd81b6ded09d15749ddab46ec6bdda98a7f9d6":{"kind":"REWORK","_number":1,"created":"2020-09-10 12:59:31.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/750982/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/750982/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/750982/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/750982/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/750982/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/750982/1"}}},"commit":{"parents":[{"commit":"896e4c1843cf1ec1db2d81ddffc8ca05f62792bf","subject":"Merge \"update constraint for python-magnumclient to new release 3.2.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/896e4c1843cf1ec1db2d81ddffc8ca05f62792bf"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-10 12:59:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-10 12:59:28.000000000","tz":0},"subject":"update constraint for aodhclient to new release 2.1.1","message":"update constraint for aodhclient to new release 2.1.1\n\nChange-Id: I3a83cfe519a36031be458f26ed6b0e6fc4c880b6\nmeta:version: 2.1.1\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: Iee5c63c48bfca2a8dad3621a5f57b19258db720a\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: zhurong \u003caaronzhu1121@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7dd81b6ded09d15749ddab46ec6bdda98a7f9d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7dd81b6ded09d15749ddab46ec6bdda98a7f9d6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
