)]}'
{"id":"openstack%2Frequirements~751086","triplet_id":"openstack%2Frequirements~master~I64218152535c9947a205fc424e8203215935fb40","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I64218152535c9947a205fc424e8203215935fb40","subject":"update constraint for python-octaviaclient to new release 2.2.0","status":"MERGED","created":"2020-09-10 19:02:06.000000000","updated":"2020-09-10 23:23:11.000000000","submitted":"2020-09-10 23:21:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"751086-1599780109371-bbaa24f1","meta_rev_id":"fabcce6ac61a9c44c0f7ae8ab8c11c839ae0d870","_number":751086,"virtual_id_number":751086,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-10 23:21:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-09-10 19:07:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-09-10 20:12:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-09-10 19:35:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-09-10 19:35:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-10 19:07:57.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-10 19:35:55.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-10 20:12:01.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-09-10 23:21:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"65c56613efbc6428179109bb201a851bc103e15e","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-10 19:02:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b77ea8b8c5d151318a75f058cb338594de844fd0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-10 19:07:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f2fc369f0bd4bcbdfb64c0a7cca6ef8f51ed1aec","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-10 19:35:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d60ba2c5067eeffe19d881cb62f0b60e99274705","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-09-10 20:12:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2fb1a25d5c99e7773fd4e0fa03dd49bcad2dbecb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 20:18:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/78383e83ac054eadabd8d8b1957e35fd : SUCCESS in 1h 07m 38s","accounts_in_message":[],"_revision_number":1},{"id":"477f7877c42fb81f2a8683fe1fc499dd884f76ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 20:32:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e334738f67614faea118348b79e92088 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88568b74663c4d558387561cf775a547 : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/369e9d494cd84181ba4896f88b767d24 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd4ff72657aa4d0bbfe55ade0a6a0543 : SUCCESS in 3m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5bf2cc6d59e047b2bedf5013b3640fc7 : SUCCESS in 6m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/131b857383ef4a06b4010e6e16851bca : SUCCESS in 4m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/27c96ad38a3348c5b18ab570658bbf0e : SUCCESS in 6m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/66e8dc8779d7439f80085981d683c1dd : SUCCESS in 7m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fa395dd517ab42578991004309c2800f : SUCCESS in 8m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fdbd6e21991142f2bf6cddd8dcce27ab : SUCCESS in 5m 08s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/89e32e764981401e8a4dff63839eb6d5 : SUCCESS in 14m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/df33dc478e5249f0baf5e2026a7b4f61 : SUCCESS in 12m 23s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ba6fe82185914f2bbb52dad395a9eebc : SUCCESS in 6m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c4e7d617b9084f63988c4a22236ad766 : SUCCESS in 11m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3143bd458fc743788715545aa7e6beb1 : SUCCESS in 28m 53s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b1fc38b900e144359162a16b904d8185 : SUCCESS in 11m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/afffb2eb88664831a372c7e531d4ea6c : SUCCESS in 3m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/de464ee3a3064289b7b855993e0b6467 : SUCCESS in 16m 35s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/7310040877f941c5bd23f40ffa330a98 : SUCCESS in 4m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2c4098487a6347bcb2ccf62d58ee39b0 : SUCCESS in 6m 39s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ab4e3705e84c4659940356e2620f6412 : SUCCESS in 13m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ad3d50596e1d4934b22b2ff51360a8dd : SUCCESS in 54m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/623cda8234aa45fd97ae8fa044d0c21c : SUCCESS in 1h 23m 56s","accounts_in_message":[],"_revision_number":1},{"id":"f88990ac4d3d9c0ab4328e7ef491a151222f4a49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 20:33:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fc964ac26f07371cc93ba33aef88df67b1c05f1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 23:21:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c34b858094549bb93dd567e4f53b1dc : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6776e18cf6043b694c07e0a11dc4408 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4b02e5878284ebf84eb7e0af5d580f6 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b11642b511904e0fb4654f63443a23a3 : SUCCESS in 3m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/67f463e76288454f93023ae54a8908e5 : SUCCESS in 4m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/20c7d68d23be4b82a391d81ff1f920d0 : SUCCESS in 4m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/37d98ac0eee14351a4442e0f9b23faba : SUCCESS in 6m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/beb06f72c71c4e298a4944123bad1502 : SUCCESS in 6m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/91a9a9ebb7674a41ac8c39935036165a : SUCCESS in 8m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ebeca442ce1d481b8d8901a8d672815f : SUCCESS in 4m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f4cb951c17f64b4d8c1b544a062bab78 : SUCCESS in 7m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/921957b193ea4aa48b6c8650d6e68313 : SUCCESS in 7m 49s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/45bcaffdb18b4f219805810ce24e738d : SUCCESS in 8m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2fe5d43257344538aa8c718d1fdf423a : SUCCESS in 22m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a46631c1caa746d39ac8165ff85a4c03 : SUCCESS in 24m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/799ffdf373f04b82bb3e77582d620ea0 : SUCCESS in 20m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9b3e29e098aa4499a768c7ed0f7b9d4c : SUCCESS in 4m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d78374cfa22143ad8a54d80fc9290668 : SUCCESS in 18m 24s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d6657b1a9da04654b8e29dce05905f10 : SUCCESS in 5m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9ec0fa43c0724373b7841e5d1f98f51e : SUCCESS in 13m 15s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/510a0a3a18ed49d3b11a14434d8c8fae : SUCCESS in 9m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eec0fb95009b463985ce5648a2e1915b : SUCCESS in 58m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0848d11b7fb74f668b9dd51e343bb824 : SUCCESS in 1h 07m 14s","accounts_in_message":[],"_revision_number":1},{"id":"19b46a29e5f76f65b8f6e20e3c1ca0bf301dbbbe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 23:21:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2a056a36f38a3f44d15cd2086d71c6de29c62c2c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 23:23:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94fce242d68f4d9285eb21ea3d382603 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ff08af3c62d005ae3b491ee7e562e7119045ab31","revisions":{"ff08af3c62d005ae3b491ee7e562e7119045ab31":{"kind":"REWORK","_number":1,"created":"2020-09-10 19:02:06.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/86/751086/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/86/751086/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/751086/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/751086/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/751086/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/86/751086/1"}}},"commit":{"parents":[{"commit":"2e9271383035e1a2f351de29f4a4e4577d120ee3","subject":"Merge \"update constraint for sphinx-feature-classification to new release 1.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2e9271383035e1a2f351de29f4a4e4577d120ee3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-10 19:02:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-10 19:02:04.000000000","tz":0},"subject":"update constraint for python-octaviaclient to new release 2.2.0","message":"update constraint for python-octaviaclient to new release 2.2.0\n\nChange-Id: I64218152535c9947a205fc424e8203215935fb40\nmeta:version: 2.2.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nmeta:release:Change-Id: I2dc6bec2699427ca630958252bedc978223157c2\nmeta:release:Code-Review+1: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ff08af3c62d005ae3b491ee7e562e7119045ab31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ff08af3c62d005ae3b491ee7e562e7119045ab31"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
