)]}'
{"id":"openstack%2Frequirements~751098","triplet_id":"openstack%2Frequirements~master~Iabc9ead0c60c1cbb1ed05345fc91558fae025b24","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Iabc9ead0c60c1cbb1ed05345fc91558fae025b24","subject":"update constraint for python-blazarclient to new release 3.1.1","status":"MERGED","created":"2020-09-10 20:05:53.000000000","updated":"2020-09-10 23:42:59.000000000","submitted":"2020-09-10 23:41:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"751098-1599781294912-bbaa24f1","meta_rev_id":"03886dc3052a041fe41886b59a203910e3496f9b","_number":751098,"virtual_id_number":751098,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-10 23:41:34.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-09-10 20:06:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-09-10 20:11:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-09-10 20:19:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-09-10 20:11:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-10 20:06:47.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-10 20:11:58.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-09-10 20:19:24.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-10 23:41:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8cc514b73b285c264f18fb510c395a4cc38bbbbe","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-10 20:05:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"156db292801c9d4bc64cfaacbd415a3fc762a253","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-10 20:06:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9c2a6200164e7868c9bff4ed53eb260095624e28","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-09-10 20:11:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1b1eee32add401f701178d32a12a0763c0543a36","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-10 20:19:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d48f125b68f275479bf9965dd0750886559ba04e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 21:08:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fa8ba7c356c40b48e04b5b1429809e5 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73cc74a0feb145f8b50ed70e85043178 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbce12b4484d424c84f3ce1b8298dd08 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2578809ac0c46afbefd08748cfb560a : SUCCESS in 5m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6811d2daf19844cabaa913db685967cb : SUCCESS in 6m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/381d2b14f8f6493ea1c71d1f3593c3ff : SUCCESS in 4m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d04226e241ae48918d99a84a0fd2ed79 : SUCCESS in 9m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e24b89118ad249848535588f9263ea33 : SUCCESS in 9m 43s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/375ac4116af94e6eb5c5baaf984f6034 : SUCCESS in 10m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/82245f6409f4469994e980a00b297d15 : SUCCESS in 5m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ff3edb3c13b7417494fa66880d3f95e4 : SUCCESS in 8m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/60dfeb74ab5242e7a38efe7ed96e5ef4 : SUCCESS in 10m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3c90a9263e7748a9a12b4c5b01b2673b : SUCCESS in 6m 30s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/80fc47f2796f44a7b4433fdfc3d72935 : SUCCESS in 15m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9f71f5e81fc64b0087ffd4f19149360f : SUCCESS in 27m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/76202aa3ec0d4a80a87b93277a02429b : SUCCESS in 12m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e5dfb7a9862f4cd79a23401a29826a72 : SUCCESS in 3m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3d819caaad5442239b59a271f621048f : SUCCESS in 19m 11s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f9b010c4718b45edb9fbfb09b1ed6a42 : SUCCESS in 4m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/318a0c9090db4f3d8bec1cfb3a031442 : SUCCESS in 7m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1e78acf302224f20a6b3cd7df45f1124 : SUCCESS in 8m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2a7a3cebf28341ba832fd4b8dbf7bbfe : SUCCESS in 48m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/601e65808f4249b6b13c33efc8bc9352 : SUCCESS in 53m 20s","accounts_in_message":[],"_revision_number":1},{"id":"7b383b099e1f932ccd078a604656f63e021b6943","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 21:08:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"06b22e6671e40327960b0decd354d73ba73b59a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 21:25:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/fdf11af43c6c4feda64d57b918c36569 : SUCCESS in 1h 06m 35s","accounts_in_message":[],"_revision_number":1},{"id":"8f4798d4ca7750f2f1bb28b1181afc6bb4681aa4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 23:41:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50080ea8e0e640a1ab66367e395fdcf7 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ac73b0b5cb848d0bee52507a61cb006 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8edad5e52e9c4700aa781293cbf3a0e6 : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72195c3580ec44878f6c90d2e852d03f : SUCCESS in 3m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3f85811d8a6f427eac92673d8a759488 : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2e8b7a65157a4ef290a63eb59f562355 : SUCCESS in 3m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/777d66b437064dab8e808adec084e145 : SUCCESS in 7m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f299ec8253074e54b66f405189446967 : SUCCESS in 7m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/56253ca284494957ad36c865a5c1b6ce : SUCCESS in 8m 27s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/784f9386424e472aab1fbc4dd99c8c01 : SUCCESS in 6m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4e11461d96084481b0f6681ce9c4506e : SUCCESS in 11m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5cdcced1b07c4a08b52c85eed5b7f4db : SUCCESS in 15m 30s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1498fdaaa97844a0ba69827a3f6e6d57 : SUCCESS in 17m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fb12348d3edb43889290c934a85a6836 : SUCCESS in 14m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5291b5fe07ed424da7cc2e6480dd7c2f : SUCCESS in 25m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d172c1f56bbc4036bdca3a1716c88d20 : SUCCESS in 12m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2652d9f9b1f249a9a7a9a206eef0cf89 : SUCCESS in 4m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e517a50d5e604ed988d017555f130274 : SUCCESS in 16m 51s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ae5377c5e17947eb92ff32156cf9a3e4 : SUCCESS in 5m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/019567c542434e56ac191658a34a7ba5 : SUCCESS in 6m 35s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fd38a2adebcb432a9b757b083bc95b16 : SUCCESS in 11m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ccf3aa6b11fe4e6cb330211e6efe7348 : SUCCESS in 49m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8725b28c755e415d8ef04216b3d7b829 : SUCCESS in 1h 24m 03s","accounts_in_message":[],"_revision_number":1},{"id":"e328a88a17a4476c602c263009a517a32b099776","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 23:41:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"46599af58df64f973f5b8f172ed9003c0fbdea56","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 23:42:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/daf666aeac5f40ab9c675458aa52038e : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b0ad31e8350c753ec4810df6089fdb5025b38119","revisions":{"b0ad31e8350c753ec4810df6089fdb5025b38119":{"kind":"REWORK","_number":1,"created":"2020-09-10 20:05:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/98/751098/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/98/751098/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/751098/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/751098/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/751098/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/98/751098/1"}}},"commit":{"parents":[{"commit":"affd8e0aa99e19629572170c00b1e5332e784b35","subject":"Merge \"update constraint for python-novaclient to new release 17.2.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/affd8e0aa99e19629572170c00b1e5332e784b35"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-10 20:05:49.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-10 20:05:49.000000000","tz":0},"subject":"update constraint for python-blazarclient to new release 3.1.1","message":"update constraint for python-blazarclient to new release 3.1.1\n\nChange-Id: Iabc9ead0c60c1cbb1ed05345fc91558fae025b24\nmeta:version: 3.1.1\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nmeta:release:Change-Id: I280a1f8188cb9612ada09ab11c4f7842a20958f8\nmeta:release:Code-Review+1: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nmeta:release:Code-Review+1: Ivan Kolodyazhny \u003ce0ne@e0ne.info\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b0ad31e8350c753ec4810df6089fdb5025b38119"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b0ad31e8350c753ec4810df6089fdb5025b38119"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
