)]}'
{"id":"openstack%2Frequirements~751303","triplet_id":"openstack%2Frequirements~master~I442857131b4f0b973ad207274149b7ea72b7f6e6","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I442857131b4f0b973ad207274149b7ea72b7f6e6","subject":"update constraint for validations-libs to new release 1.0.4","status":"MERGED","created":"2020-09-11 14:27:01.000000000","updated":"2020-09-12 00:01:27.000000000","submitted":"2020-09-11 23:59:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"751303-1599868759112-87da3b05","meta_rev_id":"bbc661a13b1f89f24061c6d582b7cb0ab61542f3","_number":751303,"virtual_id_number":751303,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-11 23:59:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-09-11 14:28:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-09-11 14:29:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":2,"date":"2020-09-11 14:40:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":1,"date":"2020-09-11 14:40:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-11 14:28:30.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-11 14:29:54.000000000","updated_by":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"reviewer":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"state":"REVIEWER"},{"updated":"2020-09-11 14:40:10.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-11 23:59:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f789e4c5290452d8f37483b3f6004408affd750a","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-11 14:27:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f7f56df69f0fc8630fb8730519a9d6d0701448db","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-11 14:28:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e41bd942869ba3378f1324db62f45c24249fefcd","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-09-11 14:29:54.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"95f2a8e28312ca55b65d0776b2eabb1fff807664","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-11 14:40:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b6ab5057dcbb54b47c78b80360c85f38fd2f943a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 15:48:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/976647c936544b74a9fd782cfe669271 : SUCCESS in 1h 02m 42s","accounts_in_message":[],"_revision_number":1},{"id":"fb558d1a0f7d44018b04227866eb1bf1d5af77a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 16:19:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33bd9f601310437cbab245480dde3d7e : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17836ebc74a140e7bf021f720c363e9d : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/028fb020b9f04a669e32dc6b61660ead : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e99854b632e843dfa7f19248653a88e1 : SUCCESS in 6m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d785a764195e4b7a829e41b0303b906e : SUCCESS in 4m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/86f3454d949348f985aeea40c2658bd8 : SUCCESS in 3m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/81397f9da57e43d6b7c05c8a2a398dad : SUCCESS in 6m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fbe46cd3d6ec4be79e9ec6fc32ef778b : SUCCESS in 11m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/814c3fa18ebe4c6e8da8b2a72fdd933e : SUCCESS in 9m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9efadec1b75045809dca1787b91dd40b : SUCCESS in 5m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0730d4e8b0c74738839aa8e0d2a2cad4 : SUCCESS in 9m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5bfa0a4b0c6c4a33b8857702155ba978 : SUCCESS in 7m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ef4f8009e38c4d239b6a6c01ea00eccc : SUCCESS in 7m 56s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/335e37a1f73a4e30b9b3725c7ca69112 : SUCCESS in 15m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/60bc25334842404080853d043355988e : SUCCESS in 26m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c821fc7116b64b38ade27d7700b10bfb : SUCCESS in 21m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/671b59f85935435aba415032e2c81db1 : SUCCESS in 4m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/92b13605f25949c1a899c0960fc02f0d : SUCCESS in 12m 35s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a746fec5879a47f8873a58d029f8bafd : SUCCESS in 10m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fabaec22dd5c48168bb3ee967eaca7ea : SUCCESS in 8m 11s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7c0fc380e7c54a66a6fa9f82ef5bbfe6 : SUCCESS in 13m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c5d5262e763b4ea1ba27798fb491c2ae : SUCCESS in 57m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ef0d37bf2c6b4030a27ef116cddb7ff8 : SUCCESS in 57m 18s","accounts_in_message":[],"_revision_number":1},{"id":"9545bc0536af9324a8b5780ed9ff04ebc1a36ea5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 16:19:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c6923ed9759c842d2b30be8e393758019914c029","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 23:59:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43f22a5b5e9648bb8bb9f3404385ce95 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbe409a3a48e43d8900e8dac9de55fd3 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4c32ab008de42a6a394aa15c565cadb : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c72975fd5d2400696b3634240a35d10 : SUCCESS in 4m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c201d2cb3e34471f8e59044f274f1d52 : SUCCESS in 4m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4e1f018c31b742deae3ef25ad397f5fb : SUCCESS in 4m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/931ddd0287274445b5319ba4152279eb : SUCCESS in 7m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/38df045b9e6444a9a8cfa13ebc23a536 : SUCCESS in 8m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/95eea23b2f9b4f67b786cea7cd394f99 : SUCCESS in 8m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0b72e18736a7440e96eefe08dfd718dd : SUCCESS in 4m 45s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f8679ab5c4424aaeba8966d857f81d24 : SUCCESS in 10m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ddf3e31f2b4940af9af80e4b2836e55a : SUCCESS in 5m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/204d9b90879142b1b3d81f77d969a037 : SUCCESS in 12m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8beb5f418baf422790aad0733a3f09d6 : SUCCESS in 13m 24s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0d0081adb62f4426a837ab1f8475aaa3 : SUCCESS in 28m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5bee31b372c0425484c126dcd7c5ba7a : SUCCESS in 14m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9190512abfda447ca037f7287a9fed32 : SUCCESS in 3m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a1773478bbf44e3398c0a91eabd36961 : SUCCESS in 17m 36s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3ff5aae5ed0246279071dbf0cc82ff5c : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b1daeb4929d6468d9bde838c2e85ef7c : SUCCESS in 6m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/75a0cd89c3624bd5be581d176b82616a : SUCCESS in 10m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b5193d084824214bffee0440e5c9dfd : SUCCESS in 54m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f75339725c0c4400a8d5948b4b640eea : SUCCESS in 1h 37m 58s","accounts_in_message":[],"_revision_number":1},{"id":"9ed2e242ad526aeb31af1b95194bfc481276a557","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 23:59:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b261937e33b0650772ba9b04de4422d397d7a33e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-12 00:01:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fea5e2b203244ec83930caed80a3101 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d52d41610cab76b76d6cea03232c6b4be21442b8","revisions":{"d52d41610cab76b76d6cea03232c6b4be21442b8":{"kind":"REWORK","_number":1,"created":"2020-09-11 14:27:01.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/03/751303/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/03/751303/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/751303/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/751303/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/751303/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/03/751303/1"}}},"commit":{"parents":[{"commit":"69f74c548ae4dbb7f980c15b199121f67d725005","subject":"update constraint for python-ironic-inspector-client to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/69f74c548ae4dbb7f980c15b199121f67d725005"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-11 14:26:59.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-11 14:26:59.000000000","tz":0},"subject":"update constraint for validations-libs to new release 1.0.4","message":"update constraint for validations-libs to new release 1.0.4\n\nChange-Id: I442857131b4f0b973ad207274149b7ea72b7f6e6\nmeta:version: 1.0.4\nmeta:diff-start: -\nmeta:series: independent\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Mathieu Bultel \u003cmat.bultel@gmail.com\u003e\nmeta:release:Commit: Mathieu Bultel \u003cmat.bultel@gmail.com\u003e\nmeta:release:Change-Id: I84492a539d1c0f460da77ed24f54af89d5c823ca\nmeta:release:Code-Review+1: Cédric Jeanneret (Tengu) \u003ccjeanner@redhat.com\u003e\nmeta:release:Code-Review+1: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: wes hayutin \u003cwhayutin@redhat.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d52d41610cab76b76d6cea03232c6b4be21442b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d52d41610cab76b76d6cea03232c6b4be21442b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
