)]}'
{"id":"openstack%2Frequirements~751308","triplet_id":"openstack%2Frequirements~master~I321629d66411f3e46a0e0257afed906b871ec81a","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I321629d66411f3e46a0e0257afed906b871ec81a","subject":"update constraint for python-heatclient to new release 2.2.1","status":"MERGED","created":"2020-09-11 14:36:47.000000000","updated":"2020-09-11 23:59:20.000000000","submitted":"2020-09-11 23:59:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"751308-1599868760707-c6eda660","meta_rev_id":"34bd3a69ea9e768ff7f622b1e2c7555b62b59dd9","_number":751308,"virtual_id_number":751308,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-11 23:59:20.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-09-11 14:46:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-09-11 14:39:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-09-11 14:46:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-11 14:39:44.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-11 14:46:44.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-11 23:59:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"82089c5ecb21e708f65d62fad5704578d00eeda5","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-11 14:36:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"202cc8c5a088a7bb87d3a25f95b80177525398fb","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-11 14:39:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9ed509258885d2b814bc93e7d4e840002e9df532","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-11 14:46:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8550f83fb2ffbd1ebc4db33016fa5afebf12d821","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 15:57:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/028001e6080e47e2b4c344327f024e1c : SUCCESS in 1h 03m 49s","accounts_in_message":[],"_revision_number":1},{"id":"31ed8e3a3059237606401f7965e3f979984a980e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 16:48:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8507c37b88b14abcb18ab63180a353ca : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64b49f41b1ce4590bc573e697d22d3df : SUCCESS in 8m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/401fd746f4f84d458a91af6543c04d51 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a629d7d81d174f1e8d77b53cc7bdd002 : SUCCESS in 5m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0faa5476047749cd85d8603c8f78b4fd : SUCCESS in 4m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b87c3e0cc8ed41898c41aab2cf502a53 : SUCCESS in 4m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ccb94cdd311b4e9f98bf75c9b779d954 : SUCCESS in 8m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4c47df06dd8b456b9ef4bffce74b235d : SUCCESS in 9m 27s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fc43985752e24244a72b2eb48e93a2f3 : SUCCESS in 8m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4711eb2da6274aeba4fa1c0da7cd2271 : SUCCESS in 5m 10s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c88e4dd1ff8b449fa3f4dd1084d00104 : SUCCESS in 15m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c05e3e1c5d734f8aa136c63f378a88a9 : SUCCESS in 7m 14s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c50e43c728a6466985a23f5a8acdc676 : SUCCESS in 8m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1159bbf02b5e4eb2bc3386ca158d91f2 : SUCCESS in 13m 25s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0bad98029ee64caabc7f80e37ed88603 : SUCCESS in 33m 11s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5b19867bd4b0439a8f197c71d5e4a10d : SUCCESS in 13m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a670e5ebc1ff45d0afccfc4741f8eac7 : SUCCESS in 5m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5657b887e71044e383e1c9251f535741 : SUCCESS in 18m 57s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/743a3cac180c4409b18962217b1e2e6c : SUCCESS in 5m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/82309131c1af4a929ea0bbc2be10ecd1 : SUCCESS in 6m 50s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cf68fc09a9f84b32bdcd7b049e02ee11 : SUCCESS in 10m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/58a9963e96504f72a4d1393df4e45529 : SUCCESS in 1h 00m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/629efc2545884454befbf916d357a074 : SUCCESS in 1h 17m 18s","accounts_in_message":[],"_revision_number":1},{"id":"6a6fbaa4398b3eccbfcd982fdeb296eb3026eba8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 16:48:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ced0a62e4e5b93889550590794b63f65d74e9635","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 23:59:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c2424182e9536c365a259f276243791ad1437a6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 23:59:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d98fa0fee1814c2a9be4044256fe0048 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebdd0e7ea7854cb58a0e4ef021c144ea : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a630fca7ef3542e285bdbc27b82bc8c4 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f42857ea6c844a0c94d8be0e90364230 : SUCCESS in 4m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3a64d1044eb743f6a28169be7b5eb8ee : SUCCESS in 4m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0347edb07349476c989901bb82412d35 : SUCCESS in 3m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/207e51890cc145c4ad0da71a9f7ee5f8 : SUCCESS in 6m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3f10778eba0145a8ac99d1664220f3d5 : SUCCESS in 8m 20s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5baa95b0576f468593560f3708a8fe1b : SUCCESS in 7m 12s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4f388aeb635e420593822672b79679e1 : SUCCESS in 5m 06s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/107ab0306e1f430185b5048adc454fd1 : SUCCESS in 10m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/73c8c72aceb449468f2685e8dbd9adad : SUCCESS in 6m 59s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/82ef8baa85d34c4f85678fa9062f36f1 : SUCCESS in 6m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/119779e2d8bf48c99126cf717875ad05 : SUCCESS in 43m 41s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/93986e70840849b5bb74e2d8583e8fd1 : SUCCESS in 26m 24s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a773af9381cc4bd8b8f451144cac0c73 : SUCCESS in 23m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5240556b7f4a46609849077e3edc1dc1 : SUCCESS in 4m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d8f245dea75e47c594316fd0dc558bea : SUCCESS in 17m 32s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/222c8fd4dd4f47cbbadb4431062abd45 : SUCCESS in 4m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9f95afdf6dfa4668b01efb4413a090e3 : SUCCESS in 7m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/93d636536fdf43cdab9917bf8dc2f05b : SUCCESS in 10m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0437e7d6db7b4776a4d386d61eefeafb : SUCCESS in 1h 00m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c1a109052dc44421a09e3f64401930df : SUCCESS in 1h 07m 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9965b8f766635c9f0371b04772d91ef896328b34","revisions":{"9965b8f766635c9f0371b04772d91ef896328b34":{"kind":"REWORK","_number":1,"created":"2020-09-11 14:36:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/08/751308/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/08/751308/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/751308/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/751308/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/751308/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/08/751308/1"}}},"commit":{"parents":[{"commit":"69f74c548ae4dbb7f980c15b199121f67d725005","subject":"update constraint for python-ironic-inspector-client to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/69f74c548ae4dbb7f980c15b199121f67d725005"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-11 14:36:44.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-11 14:36:44.000000000","tz":0},"subject":"update constraint for python-heatclient to new release 2.2.1","message":"update constraint for python-heatclient to new release 2.2.1\n\nChange-Id: I321629d66411f3e46a0e0257afed906b871ec81a\nmeta:version: 2.2.1\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: Iae559270442d4882263ce1753f2f2269c0610435\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: Rico Lin \u003crico.lin.guanyu@gmail.com\u003e\nmeta:release:Code-Review+1: Ivan Kolodyazhny \u003ce0ne@e0ne.info\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9965b8f766635c9f0371b04772d91ef896328b34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9965b8f766635c9f0371b04772d91ef896328b34"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
