)]}'
{"id":"openstack%2Frequirements~751329","triplet_id":"openstack%2Frequirements~master~I07c935d825505e154b31c53a199d34b0413e82d6","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I07c935d825505e154b31c53a199d34b0413e82d6","subject":"update constraint for osc-placement to new release 2.1.0","status":"MERGED","created":"2020-09-11 15:36:58.000000000","updated":"2020-09-12 00:02:54.000000000","submitted":"2020-09-11 23:59:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"751329-1599868768388-b4a3470e","meta_rev_id":"ead2ac8ff354f737c154fd86b898bec17e080c3d","_number":751329,"virtual_id_number":751329,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-11 23:59:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-09-11 15:39:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-09-11 15:53:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-09-11 15:53:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-11 15:39:20.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-11 15:53:11.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-11 23:59:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0531ea1835e55f8fde2e20f8dff7bf3f15260840","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-11 15:36:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2196378d93a478a7ea8639df3d294684f39cdf1b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-11 15:39:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"900fc55b30f26d2d84ed69972d0c66ff455b65d5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-11 15:53:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"72fdeabadbc7296d8dcdf7322ec77619ca8f13ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 16:48:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d09856c94cb14553afa89b384996bbf8 : SUCCESS in 1h 05m 28s","accounts_in_message":[],"_revision_number":1},{"id":"17d9b30684b842a767e3fe44c639477100bcee27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 17:27:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7367fe77ec7f4de4bb3ac32eed87244b : SUCCESS in 8m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6ee64b0bf1a4f9ab66780f185456aee : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57f2413193154016a0d3019665c0fad5 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9ededf7c49a43b7b136b1f63eea4a01 : SUCCESS in 5m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/112b755cc0414315a13c2914821e4581 : SUCCESS in 6m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/73fa18e31d2e423988d4467fd3414f8b : SUCCESS in 6m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/eba17371d3d54fd9b16a05f7f9fa6d38 : SUCCESS in 13m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a231d38a7faf48c0a7dbf6d8421fabf5 : SUCCESS in 7m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1f5ca059413a443a9bfce64817f5bbc1 : SUCCESS in 8m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/aff5c40a22ef49c6a0ffa7d956bf5aae : SUCCESS in 5m 15s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3d80ccb30b064394b45e1adc39b36354 : SUCCESS in 10m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f8400e47ec65490f866ce898002d70eb : SUCCESS in 7m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5ec2991315cb429482478eaa82f2a147 : SUCCESS in 7m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d76b55106cad4789800879f1063374c8 : SUCCESS in 18m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4472605d14b740d996a2773570a35c68 : SUCCESS in 25m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/73100943f2b3422db51546620740519e : SUCCESS in 15m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/de5573ac864c4f268d019117a6e92096 : SUCCESS in 4m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/13dbf790673844ce8aa52298769604b0 : SUCCESS in 19m 18s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/938ea40a44164eeb94e0293df23eae9a : SUCCESS in 5m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e91455b3e68e4f5387e6ba3cb52b7348 : SUCCESS in 7m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3bf1f3f96ba2446e9e98d2b6c67bc95c : SUCCESS in 10m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/62592845c20b44d0a379ba1b35f814d9 : SUCCESS in 55m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3150567c276d43f9b4e6686d6f1ea226 : SUCCESS in 1h 24m 13s","accounts_in_message":[],"_revision_number":1},{"id":"3937a0dfe4d914bdfd9727fc75916f00732e76a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 17:27:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aed1e334787e04b04d0c347ab75b33493a12e0c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 23:59:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/588ae8c23b7546a4915fe67a591ad134 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b103d658a7642f088a9c577c9511d17 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54043c23f7794a3c965cce0deaaa7804 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/17d01b6f756941569296c5a1262b63a6 : SUCCESS in 4m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1f00852f004c49dc9070f5694179f4f9 : SUCCESS in 5m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/676376fe159841d7b53336aba7409bd1 : SUCCESS in 4m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9ce916ef1e584bbba65ffca693f49605 : SUCCESS in 6m 32s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/950dc636a5204988a0a5add63ec3bf36 : SUCCESS in 8m 16s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b3655cc74c8440059bff810dea33283e : SUCCESS in 8m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/18b9a439ca044b4d9527f866250c1010 : SUCCESS in 5m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/531be31d1e434a93a55bc6ae28ea2e78 : SUCCESS in 7m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/97f6f24dc2ae4695b3b7e5e34783c031 : SUCCESS in 7m 07s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5cf02cd574c840338b8c672c1fddaaf0 : SUCCESS in 7m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e6174576b0a44c8fb7d2512052f507f1 : SUCCESS in 19m 35s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0f226ff4ccde4cf1886432f05b50e3ce : SUCCESS in 28m 21s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3c7e13023dd14ed0a617fdbe24149b02 : SUCCESS in 16m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7fd3734aea184f3d96ae801b84462290 : SUCCESS in 4m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/582f4ebaf4a64e7db2a074e81e081c00 : SUCCESS in 21m 22s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/824aa00148f24f7d97859fc9c4479052 : SUCCESS in 5m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/953fd8d6f3c848d28b51c46cdbb008e3 : SUCCESS in 6m 22s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3442d3b024e04282a4431d8eef647d57 : SUCCESS in 10m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a8949490bb841fa9a75d85190d9216e : SUCCESS in 53m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3a2a7f37ba0949d4b08f008c665220cc : SUCCESS in 1h 11m 08s","accounts_in_message":[],"_revision_number":1},{"id":"d7463acd6d6caff8a243dc088115b324003c8511","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 23:59:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4ddda8b62b4ef367c2f0d13d3c07c1192f647486","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-12 00:02:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53b1f9522dac4c1f8f565b49638dbbe9 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b47b208811158a380849bd9eda209251ea55a37f","revisions":{"b47b208811158a380849bd9eda209251ea55a37f":{"kind":"REWORK","_number":1,"created":"2020-09-11 15:36:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/29/751329/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/29/751329/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/751329/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/751329/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/751329/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/29/751329/1"}}},"commit":{"parents":[{"commit":"69f74c548ae4dbb7f980c15b199121f67d725005","subject":"update constraint for python-ironic-inspector-client to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/69f74c548ae4dbb7f980c15b199121f67d725005"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-11 15:36:56.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-11 15:36:56.000000000","tz":0},"subject":"update constraint for osc-placement to new release 2.1.0","message":"update constraint for osc-placement to new release 2.1.0\n\nChange-Id: I07c935d825505e154b31c53a199d34b0413e82d6\nmeta:version: 2.1.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Tetsuro Nakamura \u003ctetsuro.nakamura.bc@hco.ntt.co.jp\u003e\nmeta:release:Change-Id: I1a2df751d8241ed331061b7e3641a826aa584564\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: Ivan Kolodyazhny \u003ce0ne@e0ne.info\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b47b208811158a380849bd9eda209251ea55a37f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b47b208811158a380849bd9eda209251ea55a37f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
