)]}'
{"id":"openstack%2Frequirements~751393","triplet_id":"openstack%2Frequirements~master~I79ee4f7f5352594a68a44332c20038bb1dc7a9e8","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I79ee4f7f5352594a68a44332c20038bb1dc7a9e8","subject":"update constraint for oslo.serialization to new release 4.0.1","status":"MERGED","created":"2020-09-11 19:25:34.000000000","updated":"2020-09-12 03:58:16.000000000","submitted":"2020-09-12 03:58:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"751393-1599883096348-a980bdbc","meta_rev_id":"89d5f8bc2dfd97beaa813459123f4aeebe70723d","_number":751393,"virtual_id_number":751393,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-12 03:58:16.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-09-11 19:26:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-09-11 20:00:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-09-11 20:00:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-11 19:26:27.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-11 20:00:11.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-12 03:58:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f543c85b0cafe99623abf872d06a52915ce62407","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-11 19:25:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"15d363f7bf831ca1131151c44afd3b5ec3d5f83b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-11 19:26:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fb3b055b7379e15b0310eeb01e0c4872d0d62d80","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-11 20:00:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0c39e8acb7993b1b95c3198bdf604834e23ac885","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 20:31:45.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f654f9c6fb0143efa8243b9c8a22f88f : FAILURE in 39m 14s","accounts_in_message":[],"_revision_number":1},{"id":"42de7476cec8abbfff23d0f4a431c1f4451a8dea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 21:06:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8c2983453514904b658cd60b98c136d : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/408dd43cbd314107a7738677d5729d96 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ddfebaddac840cfafe8e800bf063c3e : SUCCESS in 9m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e960e18882e442c5ade19ed2c43f6172 : SUCCESS in 5m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/db5c85bb33c9498284e3848ed2530f26 : SUCCESS in 4m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a60f61ce5c7c465780a3f7f20f4a277a : SUCCESS in 5m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7716df8b574045e99e75efd8757873e1 : SUCCESS in 7m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c1967a4305b34ff7bd58d839904b9dc1 : SUCCESS in 8m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9f164d8467df4f88b6282163f94e51c7 : SUCCESS in 11m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fbc139e85a014d0c8367bd1e90de746b : SUCCESS in 7m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/698d989471fd44548c37c22cb4b7744f : SUCCESS in 11m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/aa7d845cf9c84f1ba4ced5e9bbfac8a1 : SUCCESS in 8m 35s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f45b255ba2b74581b877d6ad6547380d : SUCCESS in 8m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/46cc8fbf009b4f0f88547b8d812016c4 : SUCCESS in 16m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8fe2072b6ebb4a96a5bae1bd16cabc74 : SUCCESS in 25m 31s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/186a6b517eba465fa06b111be1a40ea6 : SUCCESS in 13m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d58cc65a869a47b3a2b7e2d42a71da1d : SUCCESS in 4m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3f9a9f5fd34a4964893e71ac003995ce : SUCCESS in 17m 40s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/82c5b230ba9e43f68936d3125bbea358 : SUCCESS in 7m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4c5b88e8dc2149bcbc487b88313417cd : SUCCESS in 9m 02s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0081100f61ed4402898b56c4d7ef7a87 : SUCCESS in 11m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/244b7179a263495a869bb95c3e0447fc : SUCCESS in 54m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5ac9f1e800774c48a80598d91110df76 : SUCCESS in 1h 28m 06s","accounts_in_message":[],"_revision_number":1},{"id":"4b4d1a587f24d4a0fe81664f810e8a739ce85359","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-12 00:00:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"078f8865e870b8bc6f30ed7bcd68b5971c1ce973","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-12 03:58:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a3a06262a6fd78c3f85a6d949ef11af9fc2a3f53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-12 03:58:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3278182421864a9887af798922006566 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71f7adec786a427f8cbbe47908614c9e : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/988bfa29a435456987347d5c6058324f : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a1f16aa7bb7547c48ea5ae8d610f9bc7 : SUCCESS in 4m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c921ca4afdb545fe9ad821bc676520cb : SUCCESS in 4m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1c3861f2ac37450ca04081d76588dbde : SUCCESS in 3m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b604402b6fd847d9aa6eb014b1f2ed7d : SUCCESS in 6m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5f61955e368d4766abe12dab65d4c802 : SUCCESS in 10m 02s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d9a787cee43b4be290ad8b541ca75aed : SUCCESS in 10m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a700a7c2ad5e447d9f0d9a72196f0f87 : SUCCESS in 5m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a3e053e36b974a519ad7d4eba8632198 : SUCCESS in 10m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/82a280179a0a42cc8bdae0caec1cf722 : SUCCESS in 8m 10s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/24da4887f6274d229ad914f413861514 : SUCCESS in 6m 29s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d7b82fd7e90c408a990816ffa02da268 : SUCCESS in 13m 18s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e690df6c4c374c14b5655f8e8542df5a : SUCCESS in 22m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1f455033f77f4d60986d26cafb78cb35 : SUCCESS in 9m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/efee6d174ea341b8a9d59550f22ae628 : SUCCESS in 3m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f459595f8acf4116918fa0987738eb7d : SUCCESS in 17m 26s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1b8f0476f525491ab67713b9c2d6f83a : SUCCESS in 4m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6f2a4d22b8ac403da01281bd9127014e : SUCCESS in 5m 44s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/972efba0207147d8b2eff7679b10a214 : SUCCESS in 12m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/36bbf78bbe0d4649af33879c3e0f2534 : SUCCESS in 1h 07m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/33881638b82b4b2999b02e69840dfe21 : SUCCESS in 1h 40m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9c74b0e478a0c65e5d2f1c95cd31eba795723951","revisions":{"9c74b0e478a0c65e5d2f1c95cd31eba795723951":{"kind":"REWORK","_number":1,"created":"2020-09-11 19:25:34.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/93/751393/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/93/751393/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/751393/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/751393/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/751393/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/93/751393/1"}}},"commit":{"parents":[{"commit":"69f74c548ae4dbb7f980c15b199121f67d725005","subject":"update constraint for python-ironic-inspector-client to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/69f74c548ae4dbb7f980c15b199121f67d725005"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-11 19:25:32.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-11 19:25:32.000000000","tz":0},"subject":"update constraint for oslo.serialization to new release 4.0.1","message":"update constraint for oslo.serialization to new release 4.0.1\n\nChange-Id: I79ee4f7f5352594a68a44332c20038bb1dc7a9e8\nmeta:version: 4.0.1\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I3ae7c0913864b949dfb01fd31e1c7567fc4b7fd7\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9c74b0e478a0c65e5d2f1c95cd31eba795723951"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9c74b0e478a0c65e5d2f1c95cd31eba795723951"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
