)]}'
{"id":"openstack%2Frequirements~751397","triplet_id":"openstack%2Frequirements~master~I3312a99ee12ac53de5777ca3071aa265fa28449a","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I3312a99ee12ac53de5777ca3071aa265fa28449a","subject":"update constraint for osprofiler to new release 3.4.0","status":"MERGED","created":"2020-09-11 19:27:45.000000000","updated":"2020-09-12 03:28:35.000000000","submitted":"2020-09-12 03:26:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"751397-1599881167887-e889b14a","meta_rev_id":"85d250a1ec23ab356bfbc341cea52daaeaf12a67","_number":751397,"virtual_id_number":751397,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-12 03:26:07.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-09-11 19:28:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-09-11 20:01:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-09-11 20:01:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-11 19:28:05.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-11 20:01:36.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-12 03:26:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4dba1e765fb956edb9ac628522e6aa4264061486","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-11 19:27:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"efd7e3bfbb4be98a9b81a06aa96fea9ebc524a21","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-11 19:28:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e7fed20002a6938c8952f4deca472e031405928e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-11 20:01:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"977405c35a51aee8bbe7fb32b8d31fe1c0eda326","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 20:28:42.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e0594986a1134340ad09d4ce7ba1552f : FAILURE in 36m 11s","accounts_in_message":[],"_revision_number":1},{"id":"33cd2e7e95753ff5974072b8dfef5e6af8cd8d28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 20:45:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f38ba1a13634d95913d9643bfd55f91 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d323c722014f46349e85d2da25d385f0 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be9a6c83ee82464a9d87567943d18865 : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a0c64c564e04676858d4f85f6bc8937 : SUCCESS in 5m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/18b903a03f3c479cb59f399b0347b8bc : SUCCESS in 6m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9e2c1dd2fe59406882078a6e3fb79809 : SUCCESS in 6m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/117488455d3c46cda9536970ac73e696 : SUCCESS in 7m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/14eda58588da4597917098afb51e2a2a : SUCCESS in 7m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/25132ed15a614475a15b1f7468f135d5 : SUCCESS in 7m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8306331499dc41aeb9d65469afdaf099 : SUCCESS in 5m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/de12473c87254ad49ec4f46819345da2 : SUCCESS in 9m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f2b555a6d0df46218bda9d881d4658d3 : SUCCESS in 8m 59s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b525190afaf94955b793d97af2630d71 : SUCCESS in 11m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b85773e2b70b42bd9083f6bdf75f4207 : SUCCESS in 13m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0b174ef287a343aaae5f8dd6026c72d5 : SUCCESS in 25m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/793406685feb4986841d603972948ae0 : SUCCESS in 13m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f87fee1ae0004f80b55ec90418ac52e8 : SUCCESS in 4m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/183c406393414162bdd0c0d47828afb5 : SUCCESS in 19m 47s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0933f1e51fe2429cb0720ecedbac3e53 : SUCCESS in 6m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a7894fac5f254c979ec56fcafbd756e8 : SUCCESS in 9m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/23f92126bbd24ccbbdb86698ce1fd171 : SUCCESS in 12m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cbd5f7b332e9444cb0ab4dd6f93dd428 : SUCCESS in 54m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9e545249f63c472496becb541d6e469c : SUCCESS in 1h 10m 50s","accounts_in_message":[],"_revision_number":1},{"id":"8bcfa64847dbd07231d2d03fda7e0cdf6e55d9e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 21:57:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"67831746d5d2b0e5e5488bfee659a65413d2a06d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-12 03:26:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"259b8d2ecc2987559b9b5fa403bd2d9a4d4e0d2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-12 03:26:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd2b60b68f274894b41341c0413fb0ce : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e7d7f23c2594621957315c7f08dff04 : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35db21b7cd1443379ecbdac6f897cd14 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db6ffd9aa9b04a1fb2025b08fd817068 : SUCCESS in 4m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8e04f34480db4024b81835c5a2befb87 : SUCCESS in 4m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5909eed6de644b42b32ae2ad8ede2c62 : SUCCESS in 5m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1bbf4aaeff55480fbf95695865fa1f80 : SUCCESS in 6m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/01e173404c9c4523a51472469f59b4e2 : SUCCESS in 6m 55s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8ea79af0cfd9476faddbf9f25f88db94 : SUCCESS in 8m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/315af8477cc045faac51420b6d1593cb : SUCCESS in 5m 06s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6f11aae59ae74ac28fac87dee1405975 : SUCCESS in 10m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bcf8c0ee405d4f318152760d0e0fd638 : SUCCESS in 6m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/658e92f6942b4f169be8dd8cda5ef787 : SUCCESS in 7m 03s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/154818f6e5b64819875007df6de1a624 : SUCCESS in 12m 27s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/02e08d98cd8d454da5e20c84b045d264 : SUCCESS in 26m 26s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/60b9daa5d3f34bc690cda8ff335fb7e6 : SUCCESS in 13m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1abbedcc31104a7d863d9d738a6cd151 : SUCCESS in 4m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/87a5eeff52dc4d55b415c43c3b61c71b : SUCCESS in 16m 07s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/896980f20fa84772a30cf09d81cb415c : SUCCESS in 5m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4bce86f9896e4d0280327babe8aca0a6 : SUCCESS in 7m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/18e6dfcf4a774396b9f930ab90fe39f6 : SUCCESS in 11m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f0f87579fe6245b8a2e2ed6bc85fd556 : SUCCESS in 50m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/02c9a0cc9c0c4e89ae12d60874a10c62 : SUCCESS in 1h 07m 42s","accounts_in_message":[],"_revision_number":1},{"id":"3b453e420169e82be72d51691b9986ebb3a06943","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-12 03:28:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb4e9d14e71c4f76820371e71beec357 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ca6114d986b0baed828d83619701e332504323df","revisions":{"ca6114d986b0baed828d83619701e332504323df":{"kind":"REWORK","_number":1,"created":"2020-09-11 19:27:45.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/97/751397/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/97/751397/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/751397/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/751397/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/751397/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/97/751397/1"}}},"commit":{"parents":[{"commit":"69f74c548ae4dbb7f980c15b199121f67d725005","subject":"update constraint for python-ironic-inspector-client to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/69f74c548ae4dbb7f980c15b199121f67d725005"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-11 19:27:43.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-11 19:27:43.000000000","tz":0},"subject":"update constraint for osprofiler to new release 3.4.0","message":"update constraint for osprofiler to new release 3.4.0\n\nChange-Id: I3312a99ee12ac53de5777ca3071aa265fa28449a\nmeta:version: 3.4.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I3ae7c0913864b949dfb01fd31e1c7567fc4b7fd7\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ca6114d986b0baed828d83619701e332504323df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ca6114d986b0baed828d83619701e332504323df"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
