)]}'
{"id":"openstack%2Frequirements~751399","triplet_id":"openstack%2Frequirements~master~I89b1a2371f5e9471768a5a001a2f71a35d0225af","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I89b1a2371f5e9471768a5a001a2f71a35d0225af","subject":"update constraint for oslotest to new release 4.4.1","status":"MERGED","created":"2020-09-11 19:28:12.000000000","updated":"2020-09-12 03:58:13.000000000","submitted":"2020-09-12 03:58:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"751399-1599883093176-06ebbb73","meta_rev_id":"719885c0d438798237dfe5f8d512fd69b0ca2a91","_number":751399,"virtual_id_number":751399,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-12 03:58:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-09-11 19:28:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-09-11 20:06:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-09-11 20:06:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-11 19:28:23.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-11 20:06:22.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-12 03:58:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"00809c02c4e5f199d0ecc7d06ba15caded7b4312","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-11 19:28:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f4ee0fe4ad95695722d285a1989206d66cd1319d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-11 19:28:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"49eb4c3d8c24f2e2da34b252bd1142a23f233b18","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-11 20:06:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f007ba6074282375732ec07b58b5c85640784fdd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 20:53:59.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/951da60014bb44e28c9f2d6411c4ddcd : FAILURE in 1h 18m 38s","accounts_in_message":[],"_revision_number":1},{"id":"bd053f86509f4bcd08c9a19cd178bad02e7e309a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 21:01:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/804b00c6cf0b4bb2b112378e6335dc2f : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3622ee9205ae4711ab9306f1252426d6 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a8c9f386dfc4336b43ae3c3d69de0b7 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2156ead0aeb549058d9e9b5a754cf8ce : SUCCESS in 5m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/399dd3883c804b78a3b4654251f9be8c : SUCCESS in 5m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b3eb3a17c9354739a4d0e2597f348625 : SUCCESS in 4m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/08cfc2d3e0dc4b2aae1e588262b33f9f : SUCCESS in 9m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9e5c1ed3f5d942539d3dd74982c30097 : SUCCESS in 8m 06s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8b2b30575bda453695aa4abb02792fa0 : SUCCESS in 9m 06s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/98521aab75594434bef94ac51f33320c : SUCCESS in 7m 46s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4590ed4fccc043dc9e14f9ae19f1c579 : SUCCESS in 11m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6ec0211b6dd4499396736ab3f7020e34 : SUCCESS in 8m 23s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7a15d0e9ad8648dfb6efda3b9bf66f06 : SUCCESS in 8m 25s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ce90c27b6b6a407bbc8cd801a10ffed5 : SUCCESS in 14m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2b18b552142c4aa6b058b5abb0a64b88 : SUCCESS in 18m 20s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9b6be0784c7f4f73bb4116b3c1ae9de9 : SUCCESS in 11m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6b111e0631d14dc0bbbc313905e45885 : SUCCESS in 6m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0b1e49feee304c9c9294c57c62744959 : SUCCESS in 14m 46s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ceaa479a0c274744a65c07e970422a5d : SUCCESS in 5m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c7f70fda87bc47bfbf019075270c4dc3 : SUCCESS in 7m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/45c80327b610441a9d8f3fb312f7fe11 : SUCCESS in 10m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2efc5c7df13146ff857bf839296222b9 : SUCCESS in 56m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0e814b78ebfa40b0b758893454473a7f : SUCCESS in 1h 25m 48s","accounts_in_message":[],"_revision_number":1},{"id":"9d11137a8587a114845e84300f7069c8f9fab14a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-12 00:00:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"86c2376dfb37495b1a60ed41c2fa6701482793c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-12 03:58:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0705251319a44f1095a4230e376d2fcc : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd4cb904289f44ab9f0aa5b447753ddf : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2dfd7ef6179f44fa89e15d9a7055cc13 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e8c831fffb04a01b183560760937fb0 : SUCCESS in 4m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/08e3473cedeb45d2b0a6a337bcc00962 : SUCCESS in 4m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a2f8c1a4cc1e480798513438b4156971 : SUCCESS in 3m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/631cd8efdda24a77a126ffc1a708e05b : SUCCESS in 7m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cdf50d31f9b24e95a5a3386791592bde : SUCCESS in 7m 07s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3bd4c4ada5874d7e9d2872bee889c66b : SUCCESS in 8m 05s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a434a80239114f0f8dd97a3b046989ab : SUCCESS in 5m 09s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8e020bff1e2f4209957a5e71dcceefad : SUCCESS in 10m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b32c981af7114b268d39c857a1a201b4 : SUCCESS in 7m 07s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ac8483a078604e108416938d2429a83a : SUCCESS in 7m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/12b6672e5d7e407a9554770859d883d5 : SUCCESS in 11m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/30a46b099894436197ea2d143e5d7a58 : SUCCESS in 27m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b42196d22939409dbf5e6636ecddd238 : SUCCESS in 12m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/596c6654203c48aea7c2a2166bc4a655 : SUCCESS in 3m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6d98d52c69514d58853329a3fc3c2bd7 : SUCCESS in 12m 23s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/25686d86cf7a4a99ab9cc4c0af6ff2cf : SUCCESS in 4m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d6e39a31dcb24691a5ead55763de28c4 : SUCCESS in 5m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d30270b87c0840e3852993c49ca3127a : SUCCESS in 10m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e33b710df9034f9e86866644493fa01a : SUCCESS in 52m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/31cac73ef5bf426f964e514dfc017fc4 : SUCCESS in 49m 16s","accounts_in_message":[],"_revision_number":1},{"id":"1940320d69170bb057a5f0a458e2ba706192e8dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-12 03:58:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7d11daa923aa74bc3ad6db00efbc8d8357b861ab","revisions":{"7d11daa923aa74bc3ad6db00efbc8d8357b861ab":{"kind":"REWORK","_number":1,"created":"2020-09-11 19:28:12.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/751399/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/99/751399/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/751399/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/751399/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/751399/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/99/751399/1"}}},"commit":{"parents":[{"commit":"69f74c548ae4dbb7f980c15b199121f67d725005","subject":"update constraint for python-ironic-inspector-client to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/69f74c548ae4dbb7f980c15b199121f67d725005"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-11 19:28:07.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-11 19:28:07.000000000","tz":0},"subject":"update constraint for oslotest to new release 4.4.1","message":"update constraint for oslotest to new release 4.4.1\n\nChange-Id: I89b1a2371f5e9471768a5a001a2f71a35d0225af\nmeta:version: 4.4.1\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I3ae7c0913864b949dfb01fd31e1c7567fc4b7fd7\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7d11daa923aa74bc3ad6db00efbc8d8357b861ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7d11daa923aa74bc3ad6db00efbc8d8357b861ab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
