)]}'
{"id":"openstack%2Frequirements~751405","triplet_id":"openstack%2Frequirements~master~I681b3aa261d0a03137cc4418abda4268a93549b0","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I681b3aa261d0a03137cc4418abda4268a93549b0","subject":"update constraint for openstack-doc-tools to new release 3.3.0","status":"MERGED","created":"2020-09-11 19:29:31.000000000","updated":"2020-09-12 08:40:37.000000000","submitted":"2020-09-12 08:38:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"751405-1599899931281-a980bdbc","meta_rev_id":"3074fc2390d7fa80beba5d42ede9192a04e2aafa","_number":751405,"virtual_id_number":751405,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-12 08:38:51.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-09-12 05:50:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-09-12 07:49:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-09-12 05:50:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-11 19:29:38.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-12 05:50:00.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-09-12 07:49:15.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-12 08:38:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"82ca7e0fc5cca408067d2b6fa995a09d702eedf9","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-11 19:29:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"142604ef9700285bfe07b2eb77cfdd5adc44bda8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-11 19:29:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"051db9f66201e56bc1919d4506f9d722afddc611","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-11 20:05:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1b25112db0e4fb8d967a38bc47555d8a57830cc8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 20:29:40.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/63642d8308574f61b9d58061046ddec7 : FAILURE in 35m 18s","accounts_in_message":[],"_revision_number":1},{"id":"000c575b1bf4451d4953a5f657da3c73093427b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 21:06:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d92f42fd545049729da60ecf76add7d8 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43d26231f99a492d8839c6dbc8b0d375 : SUCCESS in 7m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f214a685421a4e48b4a57225eaf66435 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5061dfb3fee646cfa804ff0d1823ff0a : SUCCESS in 6m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/59734f9e8863474298c9560e8931fa14 : SUCCESS in 6m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/76cb2188bf514c44b71e45493b60ef35 : SUCCESS in 5m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cf68b86111ae4ef8b638ce521491b21a : SUCCESS in 9m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/32c580329b63470f81b68c6598f6555a : SUCCESS in 8m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c4b882549b79486da312106c1c4b592f : SUCCESS in 11m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9551f03f037d4eb5b5196d207f721575 : SUCCESS in 7m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/47f9bce1343941c78033fa2eb05c0a0d : SUCCESS in 9m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3082b5a15cd94dc5935915d872cfc131 : SUCCESS in 10m 14s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/963b00cc3b1c4471b1674a765bd43c9a : SUCCESS in 8m 26s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2a97e1ed2a10405bba706c8c26e48c06 : SUCCESS in 13m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c1875a5af39a4d179d73e18cb6fad7b3 : SUCCESS in 25m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b3a07a93c4e14f4f8f4b0a36e737832a : SUCCESS in 15m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6e2de339ecb14e8bb5029c6bf5c1bdd4 : SUCCESS in 5m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/30a48d2319a74b25b7929be3c0c55cc8 : SUCCESS in 19m 10s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/19888508cfa84af1a15ff32d55fd88cc : SUCCESS in 8m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1f06bd587081409099ae497314598ce3 : SUCCESS in 9m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6570106e1b654ddfb8631d74776d562e : SUCCESS in 9m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/671c5476df1f4770ae8e48e81d999d07 : SUCCESS in 42m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d44992b620be464db1db319d4eb1f0d5 : SUCCESS in 1h 09m 34s","accounts_in_message":[],"_revision_number":1},{"id":"de3844d5fd778030d78d54929c712b0d4d1b0c0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-12 03:58:17.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"497a2ba78ddafeda6a20cabd3b24b51af1ffb991","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-09-12 05:49:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"30473007eaf2c8697ae191b7d0394c4a0b38fe38","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-09-12 05:50:00.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bc3ffa47e2a4580ed6c157e49e2cab55b0a29c0b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-12 06:21:29.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5848a9cd1f5a4d7ca2f07c9c7c768efc : RETRY_LIMIT in 3m 00s","accounts_in_message":[],"_revision_number":2},{"id":"2e1b0ed28f245ddd75130c0b9ba659b6293c34d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-12 07:21:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e47b437911ad4abd86625a29c5f8314c : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff1ac8c48451414faabdbc4384bb470e : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/70ca97f5f4584c6ba73b8e9f58ed874e : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c443ab28e45c4f268ba6adc80c8720f5 : SUCCESS in 4m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0f7b692ca9e5494d82953a5a85db0bce : SUCCESS in 4m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/872ead711bfd4bd28e5a6465bd92e751 : SUCCESS in 3m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/21e1e15fc57443da8c7237e7367c85af : SUCCESS in 7m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5712b287d3bc4424831dc873793903cc : SUCCESS in 6m 56s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/eb2d99f6f3c44f5aa293962faab52d7d : SUCCESS in 11m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b4a887744bd448f9b00a405be7e9fc57 : SUCCESS in 5m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b30404638fb7448dbebb0860e687fd50 : SUCCESS in 6m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b876a51c75914999b2851ba7fbe5856c : SUCCESS in 6m 35s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a4586ec74b6e48c39002d67c7ad1486a : SUCCESS in 7m 09s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/47c91e82907d497caa3652924abc8a89 : SUCCESS in 25m 51s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ee7534df79034cb8af7529be75ac59f6 : SUCCESS in 25m 30s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6695e197745e48bd97b3e6afbbe3b5ac : SUCCESS in 12m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8f89c3fe9ba54429a5894b863b93422e : SUCCESS in 3m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8f443d1108e44b07a48a30cc11294089 : SUCCESS in 15m 15s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d790c9ea33bc4ae486689b3848afac72 : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2f18c328948744239c457811483a2a7d : SUCCESS in 5m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/67227c00dded49eb80ebacb560015a33 : SUCCESS in 9m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/271e7dbb93aa4cae9e33750fb4350da3 : SUCCESS in 52m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ded4f0d7e9c84aadbf9bde425a68b081 : SUCCESS in 1h 25m 06s","accounts_in_message":[],"_revision_number":2},{"id":"fd8487702e194de807803952708287f9e8986db3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-12 07:21:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ce269af75fcb1f662beb4019b3416d36df2f3ae5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-12 07:49:15.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"68649595142c12ae305f20dae03dbed28b8b051f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-12 08:38:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3703c7cece846bd957ec342f35f60be : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48600e21adbc4f788da1aa35a8069cc6 : SUCCESS in 7m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/802544f891f7468ea4e4c211b8677a9d : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a86e6ec154848dbad987e0e6a68d2a0 : SUCCESS in 7m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/391dcce409a64f398b5dd0ac62d517d4 : SUCCESS in 3m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/af28444622ad4a9aaa34340f00ff60e4 : SUCCESS in 4m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/abb680f9ca5d42b985c7cb3416d844f3 : SUCCESS in 6m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0ad111879f394d23908040ba198b419f : SUCCESS in 9m 25s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b6e726f714a34c848400df8818ef12ea : SUCCESS in 8m 40s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9622b7f5651245009e8e9dc475557905 : SUCCESS in 5m 54s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/394b839707c54369bb4873fe5229b4f6 : SUCCESS in 9m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/81d4c41459344617a243b767b1bacb37 : SUCCESS in 5m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0a0f398d2ae34f64b00be0661c696379 : SUCCESS in 6m 25s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/167d60b351d14eada0fef8a7a7a14552 : SUCCESS in 31m 25s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/463224e82ca34146a2a30acae0426d34 : SUCCESS in 27m 16s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0e2d55d86cfe4e2c86c28fdd917140f0 : SUCCESS in 14m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1ffde6517df8443da7a4e4cd41307b0b : SUCCESS in 4m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/071891a1f90f4e338b090753b7376fc5 : SUCCESS in 19m 49s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1e7bc3e3ab6b427daec0b45996d2ba2d : SUCCESS in 6m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0e26c8ea5efe48db8594bdfcbd890b36 : SUCCESS in 9m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/69c03a9e6a7443b080275308018c8efe : SUCCESS in 11m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a8823fd5d5ac45be952d92607730d495 : SUCCESS in 58m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6ee7911ecb2c4670b483559ef3307e26 : SUCCESS in 1h 15m 28s","accounts_in_message":[],"_revision_number":2},{"id":"b93f98f0a45abc6d6b73daca56eb5206a95cea65","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-12 08:38:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"5187bd0f05c2a6839f9c5dd2c756451a3564239a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-12 08:40:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/091fa6d60fae47bca573b9ecf6570659 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4b0a01d06173a829a61990b0b5f9d57e3b1d136c","revisions":{"bbd86d00911a1142cae22548945fc2324bbac789":{"kind":"REWORK","_number":1,"created":"2020-09-11 19:29:31.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/05/751405/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/05/751405/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/751405/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/751405/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/751405/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/05/751405/1"}}},"commit":{"parents":[{"commit":"69f74c548ae4dbb7f980c15b199121f67d725005","subject":"update constraint for python-ironic-inspector-client to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/69f74c548ae4dbb7f980c15b199121f67d725005"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-11 19:29:27.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-11 19:29:27.000000000","tz":0},"subject":"update constraint for openstack-doc-tools to new release 3.3.0","message":"update constraint for openstack-doc-tools to new release 3.3.0\n\nChange-Id: I681b3aa261d0a03137cc4418abda4268a93549b0\nmeta:version: 3.3.0\nmeta:diff-start: -\nmeta:series: independent\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I3ae7c0913864b949dfb01fd31e1c7567fc4b7fd7\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bbd86d00911a1142cae22548945fc2324bbac789"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bbd86d00911a1142cae22548945fc2324bbac789"}]},"branch":"refs/heads/master"},"4b0a01d06173a829a61990b0b5f9d57e3b1d136c":{"kind":"REWORK","_number":2,"created":"2020-09-12 05:49:43.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/05/751405/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/05/751405/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/751405/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/751405/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/751405/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/05/751405/2"}}},"commit":{"parents":[{"commit":"9d7cf2ced23eeaf1a8690f9cefa22c6423c476a4","subject":"Merge \"update constraint for python-manilaclient to new release 2.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9d7cf2ced23eeaf1a8690f9cefa22c6423c476a4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-11 19:29:27.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-09-12 05:49:35.000000000","tz":-300},"subject":"update constraint for openstack-doc-tools to new release 3.3.0","message":"update constraint for openstack-doc-tools to new release 3.3.0\n\nChange-Id: I681b3aa261d0a03137cc4418abda4268a93549b0\nmeta:version: 3.3.0\nmeta:diff-start: -\nmeta:series: independent\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I3ae7c0913864b949dfb01fd31e1c7567fc4b7fd7\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4b0a01d06173a829a61990b0b5f9d57e3b1d136c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4b0a01d06173a829a61990b0b5f9d57e3b1d136c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
