)]}'
{"id":"openstack%2Frequirements~751416","triplet_id":"openstack%2Frequirements~master~Icb0f5723847bd9067b6b53b772a60a8b0ce98692","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Icb0f5723847bd9067b6b53b772a60a8b0ce98692","subject":"update constraint for castellan to new release 3.6.0","status":"MERGED","created":"2020-09-11 19:37:57.000000000","updated":"2020-09-12 04:53:17.000000000","submitted":"2020-09-12 04:53:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"751416-1599886397188-2f639b6f","meta_rev_id":"09e05983dc45c3b1ecf59908e0f63cd9204f068d","_number":751416,"virtual_id_number":751416,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-12 04:53:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-09-11 19:40:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-09-11 20:04:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-09-11 20:04:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-11 19:40:38.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-11 20:04:16.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-12 04:53:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fd6125d19f86f2cff9b51c1e6fe66715ab0b95db","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-11 19:37:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48e289db46da44d6fd27adc5b7e6470ec8caf4c7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-11 19:40:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"005bdcdfca39fbf97b9cc55da076e02539be226d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-11 20:04:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2b3467d956fe192d6a0b7117a4b68d1972bbaf15","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 20:31:00.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/20f7ef8907544f278c0cc18cd4f18406 : FAILURE in 38m 12s","accounts_in_message":[],"_revision_number":1},{"id":"4e755165f42cc0afe415c7bc1afddf72cf6fd694","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 21:31:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aba4127a170f4e4ebe6abada5168c0fa : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd8a2d4f1a4540a39ca8686faaff35d0 : SUCCESS in 8m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f77eab6b40d43a7ae66795bc7513a3c : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/662366a33c874dcb87f8e336c916648d : SUCCESS in 7m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f93e8d0dbd0d4646819fa0cb8af94050 : SUCCESS in 4m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0ffde1e68e4b4c078ceb5842ba12a1e2 : SUCCESS in 5m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/12bb9eafcf774d9d85537a1d8b9afdd8 : SUCCESS in 8m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e273b0a13291413aa3045cbd0058edbc : SUCCESS in 8m 38s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4c6c350d34094b3584a5c1347b330c66 : SUCCESS in 18m 50s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1732ebfa118c4ba3abca1f50571a0d3e : SUCCESS in 7m 21s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ce0d1a54c76b4892882315e7313e18ae : SUCCESS in 11m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/323480f74d87415fb59500ec6ab3980e : SUCCESS in 8m 18s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/fda2673240cf4b5f874cac12823bf807 : SUCCESS in 8m 30s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/28b6e22be7cb44d68c0037a986ce5507 : SUCCESS in 21m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/815afee416fb48a5b22d96f830909b7c : SUCCESS in 28m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f8013923910647799259c8be062eb217 : SUCCESS in 13m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e69bde5af678486aa1736cbc46432343 : SUCCESS in 4m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a1b5dc4ac9504b509b9bd87d182e8ca8 : SUCCESS in 19m 02s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c8c103c13b5842ffa7e3cfa69129da27 : SUCCESS in 5m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d40e5cfdda5049a0b0f66e16b01e62af : SUCCESS in 7m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/62bb7aceab7040c0a214b8e20c80dd1d : SUCCESS in 10m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cd876260b00f4516bd129da32bb5f699 : SUCCESS in 1h 03m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b7b50ef8b66b463b8f1d1740330cf743 : SUCCESS in 1h 19m 43s","accounts_in_message":[],"_revision_number":1},{"id":"9b703ea47eff87f9b860b88682d3dbb76b42ced2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-12 00:02:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"11a083d6138f967f394d1122fb1c31a0026c3880","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-12 04:53:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b63be64bf9254479921ab0a1ee1fcffb : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68ec57fceca9419c8d650792bd1fa96a : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39903b3b4b0a4800bba94b1632f26ed2 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41752fd935114281bd90ed75aff2dd40 : SUCCESS in 3m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/55b111d39f864c029410b4c136604017 : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3aaa0b66a55946649fe58ff1162df4b1 : SUCCESS in 3m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0af26067fbfa4eb4b8d702d0c6517a83 : SUCCESS in 5m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/64308ff89f5649d2b98694586c693219 : SUCCESS in 6m 05s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1de104a86c594fef93f5d741eeffaf30 : SUCCESS in 7m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4707454439894cc89c80b0537dca6603 : SUCCESS in 4m 19s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/dd8ba6ce3648414fa6bc5926a148f395 : SUCCESS in 6m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a93802deb3a34088aba7820a3d1e9cbc : SUCCESS in 5m 40s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/87cd71d0a02c4de3a734d38f102b0198 : SUCCESS in 6m 04s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5d71b79b8aa24b40a2a6a89eb7ee9bcc : SUCCESS in 10m 44s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5e4c74e07d9c4433b85b80041a522dea : SUCCESS in 16m 30s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b4984fb0546640da924fb1a5f7592824 : SUCCESS in 12m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/373320943a8c4bc68a0abaef573706aa : SUCCESS in 3m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b092daf4cc4448649a5590322230903b : SUCCESS in 16m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/518b154eabee4e49b13fb6770a2653ef : SUCCESS in 4m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5e91fe00f335482cbedc101df73aa865 : SUCCESS in 5m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7b65967d6989415f88ff36f632dd6406 : SUCCESS in 9m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e766fa6c79444a368e04db6ef889f244 : SUCCESS in 1h 00m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ae2dfe60d58b4029a9de3b7f5b4c31c8 : SUCCESS in 1h 15m 24s","accounts_in_message":[],"_revision_number":1},{"id":"289bceadc25f917df81ee31c29c6b323c2261b63","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-12 04:53:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1ba25e57aec184bdf52906a81ddebe7942abe5f5","revisions":{"1ba25e57aec184bdf52906a81ddebe7942abe5f5":{"kind":"REWORK","_number":1,"created":"2020-09-11 19:37:57.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/16/751416/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/16/751416/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/751416/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/751416/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/751416/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/16/751416/1"}}},"commit":{"parents":[{"commit":"69f74c548ae4dbb7f980c15b199121f67d725005","subject":"update constraint for python-ironic-inspector-client to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/69f74c548ae4dbb7f980c15b199121f67d725005"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-11 19:37:56.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-11 19:37:56.000000000","tz":0},"subject":"update constraint for castellan to new release 3.6.0","message":"update constraint for castellan to new release 3.6.0\n\nChange-Id: Icb0f5723847bd9067b6b53b772a60a8b0ce98692\nmeta:version: 3.6.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I3ae7c0913864b949dfb01fd31e1c7567fc4b7fd7\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1ba25e57aec184bdf52906a81ddebe7942abe5f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1ba25e57aec184bdf52906a81ddebe7942abe5f5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
