)]}'
{"id":"openstack%2Frequirements~752280","triplet_id":"openstack%2Frequirements~master~I1d7e4025484d46fcfae2d22b3cec4277cfbf7b50","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I1d7e4025484d46fcfae2d22b3cec4277cfbf7b50","subject":"update constraint for python-troveclient to new release 5.1.1","status":"MERGED","created":"2020-09-16 17:18:37.000000000","updated":"2020-09-16 22:00:02.000000000","submitted":"2020-09-16 21:49:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752280-1600292966830-e889b14a","meta_rev_id":"fb615778cf7ccd07aca8295d6da319dc2e8ce248","_number":752280,"virtual_id_number":752280,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-16 21:49:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-09-16 17:19:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-09-16 18:54:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-09-16 18:54:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-16 17:19:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-16 18:54:57.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-16 21:49:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bcac012c6eded49e9a94d44b6690e97daedc384c","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-16 17:18:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4a62283203e2cb1f704e69dd8a2c3d21a8e13de6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-16 17:19:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"301cea98f658efbe6eed1030e5a51323209d95d3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 18:29:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/33fd88b44d944255b7b984a86e7374e7 : SUCCESS in 1h 05m 58s","accounts_in_message":[],"_revision_number":1},{"id":"68a01d37c870bc36f4d2f8935fe461be0606b00c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-16 18:54:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8b70f46aed64986d8f8d854bb1d08ac8f497bf83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 19:09:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b495f5f581fa411cbed499816eb5c325 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bbfbacdb3064b08a3b248ead92ac1fa : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a492cbfaa6e649878250e6901c8d2b18 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac09d7e4089f459f82b9551b1c08e481 : SUCCESS in 7m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/20ac913286124813a7b7ce37abd22346 : SUCCESS in 5m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ab0a937644d44cf2a643da79b5f60950 : SUCCESS in 4m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5764057889b147a9aaed71e19545f0dc : SUCCESS in 11m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ba6d8dcc62bf475286fa9015082a68f5 : SUCCESS in 8m 38s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0bcff1df0f774d45bd65c99f85dc5f73 : SUCCESS in 10m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d104d8c892354908857cd89aa047de07 : SUCCESS in 6m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3c950d1be4994a08bd9e84839bd8d5db : SUCCESS in 11m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5b1ea66b518e439f96c8001fc542c56f : SUCCESS in 10m 38s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b0fca969ed604d598b847cb98f37233d : SUCCESS in 11m 13s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e8cc4fa43291483789bdd1a687a1e78c : SUCCESS in 19m 19s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b34d856997af42fab248d80be38a1829 : SUCCESS in 25m 24s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b476783abd6f4784853f7635141c92fb : SUCCESS in 12m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1babb7cf413d4c32976669e557d32c13 : SUCCESS in 5m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4027533879954c52ba62f3efdf071a35 : SUCCESS in 18m 19s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a3157ee8341642c093f2efc1f6ecc1d6 : SUCCESS in 6m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3613c7d946e24cd483aa8fb381353956 : SUCCESS in 9m 46s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/afb39ec6a8ae4bb9900d69d4926497be : SUCCESS in 13m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/13d3ba6ef9654820b8f73241a5027218 : SUCCESS in 1h 04m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8982992b1f6d485bb894929f0207fb74 : SUCCESS in 1h 21m 16s","accounts_in_message":[],"_revision_number":1},{"id":"cb33d9f973b7701b38cffc8cb47a577257545b76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 19:10:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f05f6bd8c123cc4a85c61086e7c06bad2aad1769","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 21:49:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea14ecd7f5544c9484d3bae884a12d11 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06547d621bb94ac6b466d6323c987142 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b46b1c01fed8492d8a42adb77192e731 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9c4446ebfb041e599b3815c7a987b6a : SUCCESS in 5m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/75f461aad1ec4f25a57ffbf87dc3689a : SUCCESS in 3m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5e65d7e726004579beee8a0d07308633 : SUCCESS in 4m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f770df52b4e34f628c492c44a08bc87f : SUCCESS in 6m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b5d73bb9a3c84fbdb6bbfccf8ffac1e9 : SUCCESS in 8m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e28623d77eb4493e9edbe3911e48a340 : SUCCESS in 9m 01s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d9a827b69f0f4921a3c1b10a23acb446 : SUCCESS in 5m 27s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/22ff65df1a1f4826bc03655f61bd6bfd : SUCCESS in 10m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7f9cd5176c704fdb92934e4e94aefb01 : SUCCESS in 9m 40s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1bd3f38897394c66b8b805d85e3c402a : SUCCESS in 8m 00s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4f8954b2723647b382fece3b789bd759 : SUCCESS in 11m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a3fd123c5bc9412e8191ef632910c654 : SUCCESS in 28m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c3959c79b6f7491ca002bd28a4353988 : SUCCESS in 12m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f166559d679e4322b4eb8f3a8c0a906b : SUCCESS in 3m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/22fcb1bbb65049fba308424f648373ab : SUCCESS in 19m 36s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/add6e240b89c4b43be437eebe06dad75 : SUCCESS in 6m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d3737eedeeee4527aa397b7c57ae006a : SUCCESS in 5m 53s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e711923d1c5f4cd6a07ea03f1fe3e025 : SUCCESS in 10m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6d73ee1ab00c4e31b35364c2a834264c : SUCCESS in 57m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cc90229ec3c248afa7f02787b6fbf0b6 : SUCCESS in 1h 06m 22s","accounts_in_message":[],"_revision_number":1},{"id":"37b72652c1259b99ba77bdfd8014181813387688","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 21:49:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c81af163dc4ccedf0631428bfebfe7f5db4d07a0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 22:00:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a3a73d00fd145b6b437b19f0c5f67f4 : SUCCESS in 2m 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b34454c0c655a562fa05393603d04f798ea254d9","revisions":{"b34454c0c655a562fa05393603d04f798ea254d9":{"kind":"REWORK","_number":1,"created":"2020-09-16 17:18:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/80/752280/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/752280/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/752280/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/752280/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/752280/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/752280/1"}}},"commit":{"parents":[{"commit":"f38aa325e5fa08c24a0a6307a74588ec972bf9c1","subject":"Merge \"update constraint for python-senlinclient to new release 2.1.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f38aa325e5fa08c24a0a6307a74588ec972bf9c1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-16 17:18:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-16 17:18:36.000000000","tz":0},"subject":"update constraint for python-troveclient to new release 5.1.1","message":"update constraint for python-troveclient to new release 5.1.1\n\nChange-Id: I1d7e4025484d46fcfae2d22b3cec4277cfbf7b50\nmeta:version: 5.1.1\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Lingxian Kong \u003canlin.kong@gmail.com\u003e\nmeta:release:Commit: Lingxian Kong \u003canlin.kong@gmail.com\u003e\nmeta:release:Change-Id: I8340ab5a7c3faf3995512b23cf5e9e8cdf7b7be6\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b34454c0c655a562fa05393603d04f798ea254d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b34454c0c655a562fa05393603d04f798ea254d9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
