)]}'
{"id":"openstack%2Frequirements~752409","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I76959ea2735b06d55e1b76cf9fd70f261e6b7000","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"I76959ea2735b06d55e1b76cf9fd70f261e6b7000","subject":"update constraint for python-troveclient to new release 3.0.1","status":"MERGED","created":"2020-09-17 09:11:23.000000000","updated":"2020-09-17 19:04:17.000000000","submitted":"2020-09-17 19:04:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752409-1600369457686-dddb1e4e","meta_rev_id":"0517812eab199ace88a3bd7c74618ab984a902f9","_number":752409,"virtual_id_number":752409,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-17 19:04:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-09-17 17:11:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-09-17 16:23:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":2,"date":"2020-09-17 09:25:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-09-17 17:11:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-17 09:25:33.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-17 16:23:49.000000000","updated_by":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"reviewer":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"state":"REVIEWER"},{"updated":"2020-09-17 17:11:29.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-09-17 19:04:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2595887456cdb4cad4b8e596452949ed8b47c8f4","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-17 09:11:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8e62ae23dde88eb3b0832e33306c1102e0fe61b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-17 09:25:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bbff871a8e9d5c60879d14b916db4104f6198130","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-09-17 16:23:49.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9be57ffdcc6d03e2db46a4a354162bad561f7be5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 16:24:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71bd6e0863b3485a93222809127cab49 : SUCCESS in 6m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7c0ef490afe04bd2b2966cdefc73494d : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6d6e14bfad84817848edcefd93b893b : SUCCESS in 6m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7e10f9d0c1344b33a0dec5295642c9b3 : SUCCESS in 6m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/306660259da54431bf0a5eb81d41d96e : SUCCESS in 5m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c9744f9f752240c4b479bbbbf733c099 : SUCCESS in 4m 20s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b73c78a9d08a4a28abfaffa795ac0dfa : SUCCESS in 8m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/95460fea02c34a82b2e9838c9fc5408c : SUCCESS in 11m 15s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/5835c6d9b2c04b35a4e9193396fea1cc : SUCCESS in 11m 27s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/42012680b2fc4d4fae06d7677207c48a : SUCCESS in 13m 40s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/a59d185f1d524fcd8654aab4a878c78b : SUCCESS in 6m 27s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/30d5a0fcfe4f4c8a9d1a96be23ab0f1a : SUCCESS in 6m 30s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/585fe2fba0a74b8c853e4e4246d05274 : SUCCESS in 12m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/79fdd39a4eea49f7abc2bc04056dc9df : SUCCESS in 12m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3f7e32fb82c6445780207b3ddafec7f1 : SUCCESS in 13m 33s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/288412ed1d164140a5c2792a1fbdd0f4 : SUCCESS in 17m 20s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4969ff239f1043e395c3fd35477e47af : SUCCESS in 13m 16s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/a2291f07385d4be2a76a20e6eb38b282 : SUCCESS in 27m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8a7b3b165c2f4e94a0ef8d016552069b : SUCCESS in 42m 10s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/829add312886471a9449e8f6c03cd835 : SUCCESS in 15m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c36c798502124950a7ace68fd8cbeeaa : SUCCESS in 15m 44s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d423e7160ee24132b318d6e635a9fdaa : SUCCESS in 14m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ca0cb929d03a4c6aa35cdf81aca4551f : SUCCESS in 19m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/795d68f56fe141c9a2595aa5cf5fc584 : SUCCESS in 7m 04s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d8891e7d5c424e4abbb02f718687fe24 : SUCCESS in 9m 44s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1372dc34c22148b6b19d2921ab250e38 : SUCCESS in 13m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c857d8ecf7944488a2b1835b6cbaf021 : SUCCESS in 1h 07m 28s\n- tempest-full https://zuul.opendev.org/t/openstack/build/33586edf90654727bba98be305817974 : SUCCESS in 1h 50m 26s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8a02e6d91d7b45ebba57b4d4b34a4786 : SUCCESS in 1h 11m 29s","accounts_in_message":[],"_revision_number":1},{"id":"e269a36f9fb9932b7bb78f56c46751d99da3887c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-09-17 17:11:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9b624999f73a6270b748fce5640212e8b61a2edc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 17:11:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d6a0f3a26cc9183611a5a1e61273ca9dd2fc6dd3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 19:04:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9dad48bfb86971d86f65d1d9b07e9900d2acedf8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 19:04:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/284f6d8e64e044ffb1275246e8cfdb40 : SUCCESS in 4m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/99f5d3d67bef48f0854e56c347f6b10a : SUCCESS in 10m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/194b7c5e32bb4e968a969d01f2a31e08 : SUCCESS in 6m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/03518be14004448ab049d7dbfc933e21 : SUCCESS in 4m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bbba9f9a040441998aa455b0d607ec70 : SUCCESS in 4m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/edd0068b66fa4301b8e417bd131979df : SUCCESS in 5m 03s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/053cdbea2f85473c92fc13f38cb1e901 : SUCCESS in 6m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b49acc84d0f24241b01ecf9507b97492 : SUCCESS in 8m 56s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/6b4242b77cf148a2a336bd944255d54a : SUCCESS in 10m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2cab7b5f571b47e1b9e8216ded2958f3 : SUCCESS in 9m 06s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/ab40299d9f554a6d82daff5ffff0699a : SUCCESS in 5m 36s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/25ca9758a90a4ecdbf061f552b7de9eb : SUCCESS in 5m 25s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/2fd90beb55f6404b9ee94540f392adcd : SUCCESS in 10m 36s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8c18357be3b6477583f2371736eda4f7 : SUCCESS in 10m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f0f434a495584dbd8e0648db96cec496 : SUCCESS in 7m 25s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/89924697b1324a14b8fa94df15f9daa0 : SUCCESS in 13m 36s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8345bd4a66294776a439c41774b0939a : SUCCESS in 27m 45s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/e8a903cdf09b4bd9bba4a44ada3b0e1c : SUCCESS in 24m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4f8ed6ba71b14df78460384e54ead321 : SUCCESS in 24m 29s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b7678073c30c48aebcfba4b865c3075f : SUCCESS in 17m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/28415720865e492cafd7f1461e71eb9c : SUCCESS in 16m 39s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/91f912576b8840fd8eb4603c1ab8af47 : SUCCESS in 17m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/276586dd5c114843b9f3a5805a47d530 : SUCCESS in 23m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/684d32f6b46448d08841ef6cae07607b : SUCCESS in 8m 06s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/031b94bd11904444bf1f0416e14d9b4d : SUCCESS in 9m 29s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0ea0d0dd520347259254396e70ac1f13 : SUCCESS in 9m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/40f5ce686b9e4e37a59dc271f4587e8d : SUCCESS in 1h 11m 05s\n- tempest-full https://zuul.opendev.org/t/openstack/build/91ad38f5e5f24a80a377188813f1137f : SUCCESS in 1h 34m 58s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b1b46203ed084972bbbd3750cb73db1c : SUCCESS in 59m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0cfc2ef6318b639bb4287fa051eae8f3ac8cc426","revisions":{"0cfc2ef6318b639bb4287fa051eae8f3ac8cc426":{"kind":"REWORK","_number":1,"created":"2020-09-17 09:11:23.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/09/752409/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/09/752409/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/752409/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/752409/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/752409/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/09/752409/1"}}},"commit":{"parents":[{"commit":"0ffbc751e486c92782f2a06b0684a77c1a8aca15","subject":"update constraint for os-brick to new release 2.10.5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0ffbc751e486c92782f2a06b0684a77c1a8aca15"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-17 09:11:21.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-17 09:11:21.000000000","tz":0},"subject":"update constraint for python-troveclient to new release 3.0.1","message":"update constraint for python-troveclient to new release 3.0.1\n\nChange-Id: I76959ea2735b06d55e1b76cf9fd70f261e6b7000\nmeta:version: 3.0.1\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Lingxian Kong \u003canlin.kong@gmail.com\u003e\nmeta:release:Commit: Lingxian Kong \u003canlin.kong@gmail.com\u003e\nmeta:release:Change-Id: I8113f6e81589017b49f16793c576a14ea77f895b\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0cfc2ef6318b639bb4287fa051eae8f3ac8cc426"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0cfc2ef6318b639bb4287fa051eae8f3ac8cc426"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
