)]}'
{"id":"openstack%2Frequirements~753239","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I140f89beb2e73740d9c56d090fc8a6b1b32b3c81","project":"openstack/requirements","branch":"stable/train","topic":"bump-up-pifpaf","attention_set":{},"removed_from_attention_set":{"29071":{"account":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"last_update":"2022-08-02 05:19:07.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I140f89beb2e73740d9c56d090fc8a6b1b32b3c81","subject":"Bump up pifpaf version for train","status":"ABANDONED","created":"2020-09-22 07:25:34.000000000","updated":"2022-08-02 05:19:07.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"9ecd6d77e03e05a425f4ae8714f2f4ccb7b2430e","_number":753239,"virtual_id_number":753239,"owner":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-10-02 05:13:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"date":"2020-10-02 03:24:22.000000000","_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"recommended":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":-2,"date":"2020-10-02 00:40:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-10-07 11:16:22.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":2,"date":"2020-10-21 11:15:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-07 10:38:29.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"CC":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-22 15:32:15.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-09-28 06:26:04.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2020-10-02 05:13:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-07 10:38:29.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2020-10-07 11:16:22.000000000","updated_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"reviewer":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"state":"REVIEWER"},{"updated":"2020-10-21 11:15:46.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-07-18 23:28:57.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"CC"}],"messages":[{"id":"e1563ff135bd8754846fc4e3f5b19efc57b3d4b4","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-09-22 07:25:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96acf86fc28af71963ef305aef6a7905920515ea","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-22 08:11:00.000000000","message":"Patch Set 1:\n\nDid you seen an issue related to pifpaf?\n\nUsually we try to avoid to bump third parties requirements on stable branches [1].\n\n[1] https://docs.openstack.org/project-team-guide/dependency-management.html#upper-constraints","accounts_in_message":[],"_revision_number":1},{"id":"6426b6371e19fa42f411e3ba42f1ccd3155c40cc","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-09-22 08:26:37.000000000","message":"Patch Set 1:\n\n\u003e Did you seen an issue related to pifpaf?\n \u003e \n \u003e Usually we try to avoid to bump third parties requirements on\n \u003e stable branches [1].\n \u003e \n \u003e [1] https://docs.openstack.org/project-team-guide/dependency-management.html#upper-constraints\n\nsorry, I don\u0027t. I will try to find out if there is anything related.","accounts_in_message":[],"_revision_number":1},{"id":"72a24368a5376be14b180a17201358f8a206e926","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 09:08:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0584b9401a2e4425a67c26e561bff974 : SUCCESS in 4m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c1b231f394724ecc923b3d6159d56dac : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/551436d6340c49768811a22f60cf84b6 : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6a099442468d4c8992845152dd48ac9a : SUCCESS in 5m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b1d4074597d046f6914c5e2f4d2f4ce3 : SUCCESS in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/71ac79cc1f3c4447b611b3260c417c42 : SUCCESS in 4m 10s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/8a6c0b162c014ed486eca1dbaeb0708d : SUCCESS in 8m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b1a272f7583a46c5914c482d82c465d7 : SUCCESS in 8m 38s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/bd2767b1f3a842f598f64c7e51d4c417 : SUCCESS in 9m 43s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5ed85dc4bbbc439ba41f3b8f46b79ee5 : SUCCESS in 9m 59s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/3935e3d44a974b66b67972869cb5a9e1 : SUCCESS in 5m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2c0a19a23975475a9bf57683df2dd6e4 : SUCCESS in 6m 08s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/ffd9e8d6eb014c7f8684a3331fb4fcd7 : SUCCESS in 11m 54s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8551189fcedc4d65b0a4b69d56a45da9 : SUCCESS in 10m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2b6a6080153147cd8f5339f9748ebaaa : SUCCESS in 8m 19s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/8d1eb4aa20c7434b834ad14b2c4c529f : SUCCESS in 13m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/28ab81212c424cf9b3b5a54e649fc4c0 : SUCCESS in 13m 28s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/0d0b524e8c44405a9c3c9f5242cc52e0 : SUCCESS in 23m 30s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/801edece6f3b4d25a5cf30bcf8cfb9d7 : SUCCESS in 21m 24s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/841fde41ffad41299b68f82ba9a1c330 : SUCCESS in 13m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3cb93989970c466888d780585eb8697b : SUCCESS in 12m 26s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9f9a0c39a37a4f2d9f1ca0eea386a495 : SUCCESS in 12m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a96b662a1e7347629b4ad30144739f22 : SUCCESS in 19m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4af10a26aa2c42deb6e75c87cfe7c864 : SUCCESS in 5m 56s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/174515c05f2b4d4a88285d786de4233e : SUCCESS in 9m 53s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/500bbc8bfb204cd7bdd39c3edefb5a34 : SUCCESS in 10m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9054e660a8b04846ab08f46841b16aa7 : SUCCESS in 1h 04m 32s\n- tempest-full https://zuul.opendev.org/t/openstack/build/810d5c49d21c468d87b265f789971d46 : SUCCESS in 1h 24m 43s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/80cc90add79f4f2da9233c4d8c027068 : SUCCESS in 49m 59s","accounts_in_message":[],"_revision_number":1},{"id":"ad97867d4cd1a617f36e5aac692c1a156edd4094","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-09-22 15:32:15.000000000","message":"Patch Set 1: Code-Review-2\n\nuntil we have a specific reason I\u0027m going to keep this as a -2","accounts_in_message":[],"_revision_number":1},{"id":"f7e68868e0d0d7db447a32261d2bd63923c5a6aa","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-09-23 01:48:17.000000000","message":"Patch Set 1:\n\nI thin is issue https://github.com/jd/pifpaf/pull/124 is suspicious. could you please have a look at it? thx","accounts_in_message":[],"_revision_number":1},{"id":"46a4b704b7f758cd0de65c28b324e159e78e609e","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-09-23 01:49:27.000000000","message":"Patch Set 1:\n\nmoreover, this bug is fixed only for version \u003e\u003d 2.4.0 ...","accounts_in_message":[],"_revision_number":1},{"id":"c67003efb0ed4fe45422477174bab8c996f525fa","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-09-23 09:38:53.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c41f7ed3dfece45bfcec6b9442b29ec982066573","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-23 10:16:17.000000000","message":"Patch Set 2:\n\n\u003e I thin is issue https://github.com/jd/pifpaf/pull/124 is\n \u003e suspicious. could you please have a look at it? thx\nHello,\n\nHave you faced this issue?\n\nFrom what I understood of this topic the main reason behind this patch is too keep version of pifpaf aligned between ussuri and train, and we don\u0027t observed similar issue on openstack, isn\u0027t?","accounts_in_message":[],"_revision_number":2},{"id":"07c298b4314e48f80b91673411896d9716f688d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 13:11:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2267c996909b4ea29c865cf40c10e24f : SUCCESS in 3m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4cc4f583e8214a7291c6dfd84b1aaec2 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a483e6f1117844acba9e6e0b1ba3a029 : SUCCESS in 3m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/796717ae5fc94041baa0a761ca075968 : SUCCESS in 4m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/97b15f14257a43b1b31ac3c6acf1326e : SUCCESS in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0f70c13aecfa42df9b4d9daf10d72447 : SUCCESS in 3m 21s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/5a632838977d47ad96ab1e83a253e058 : SUCCESS in 7m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3472f816b51b4d299ba80cf4aabe23db : SUCCESS in 8m 42s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/be1696d98a1045dfa1f9cef51c7a57b5 : SUCCESS in 9m 22s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9aa02f26c4e84b36bf4fd3ffa8a844af : SUCCESS in 8m 40s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/7469256fb0164b5b90fa084604930c49 : SUCCESS in 4m 12s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/43b829bc111c463cbe0623cc4373925a : SUCCESS in 5m 36s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/3282d3eefadc4774b7e0c2476f354881 : SUCCESS in 10m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/602b1d7989fc494383bcabfd3613fdf1 : SUCCESS in 9m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c854436e8dee46e98d3687badd0163fb : SUCCESS in 7m 08s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/e00061088baa476da23c0b1199f168cf : SUCCESS in 15m 20s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/500c08f4c9c14f6293477745cb53f614 : SUCCESS in 20m 38s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/1857a69ddddf44bfb5efa76b8484309b : SUCCESS in 22m 18s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/787f6f24449e4dc4a424ffaa2cfd9bd7 : SUCCESS in 26m 07s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/3e18230345d74cddb7e958b98a10f9fd : SUCCESS in 14m 46s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8e8226bc6e5f44de880acf5859706d9c : SUCCESS in 12m 21s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e8716beb38ca456081ca93b8fb19bd90 : SUCCESS in 12m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8f3c33490e4e4fa383c51bf21710ade3 : SUCCESS in 18m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a47081336a594550b0c4c3c604fc7b0e : SUCCESS in 6m 27s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6fe54e2177bc4f87b480a1d17506de76 : SUCCESS in 12m 22s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cda4949ca4ad45cfb1bf62c33254e304 : SUCCESS in 8m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a05f1678f78b44b2a14137862286a9f7 : SUCCESS in 57m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/85f873ecfba247819e1058ed741438d7 : SUCCESS in 1h 22m 20s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c7fb504e58c74aaba93e8be8c5eaac15 : SUCCESS in 1h 04m 29s","accounts_in_message":[],"_revision_number":2},{"id":"913566200d171733195582e0c817496999c782b7","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-09-23 13:42:28.000000000","message":"Patch Set 2:\n\n\u003e \u003e I thin is issue https://github.com/jd/pifpaf/pull/124 is\n \u003e \u003e suspicious. could you please have a look at it? thx\n \u003e Hello,\n \u003e \n \u003e Have you faced this issue?\n \u003e \n \u003e From what I understood of this topic the main reason behind this\n \u003e patch is too keep version of pifpaf aligned between ussuri and\n \u003e train, and we don\u0027t observed similar issue on openstack, isn\u0027t?\n\nHi the main reason is not to align version, the issue I am facing is the https://review.opendev.org/#/c/747892/ test kept failing which I thought related to the bug I link in the commit msg.","accounts_in_message":[],"_revision_number":2},{"id":"356c06738d84f7c3b172463689587e3777d9f92e","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-09-24 00:29:18.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"eb8ca0882fb7ab7d59cbdb50bbe55cfa76302110","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 02:03:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89102f06229f46448c1e3e721fe501ef : SUCCESS in 4m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f4e8f0340ec445ebaf188ff52e578311 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bac3b6f0655c457697adab2a9d1bb72a : SUCCESS in 3m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/26c5b074af6d41b89be6e537bb5796dc : SUCCESS in 4m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/14b61ce291904124b6d558f986f0f3f5 : SUCCESS in 4m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/09c67ed7c5204ae98ee55aa5b2ca652d : SUCCESS in 4m 40s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/84e5b0a975464cbb842c12fad02b1d5e : SUCCESS in 7m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/aa0e5023f31e4829a88662ca5c9c1074 : SUCCESS in 8m 10s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ff6fcfcc90af4a66a45808c63dca9d48 : SUCCESS in 9m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/50df7c64b1c0432db15e9ec06376398b : SUCCESS in 9m 27s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/56d2475c647b445693a433c9a619da15 : SUCCESS in 5m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/425d9bf6f1d2430b930acc2abca58125 : SUCCESS in 5m 06s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/913740ab51cf441bab995ed51d43749b : SUCCESS in 9m 16s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/83b99523c3214688ad022d4fc1c14b2d : SUCCESS in 8m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/49f0cfba7640491e8405f079ba5fafbb : SUCCESS in 6m 56s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/cc5e30668b8b4deab4ad55eab841c5c2 : SUCCESS in 12m 25s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/31aabd24668b4a01b5a83513c74ea9cd : SUCCESS in 12m 01s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/269c500b96cb4898ac6ab87ad0c4142f : SUCCESS in 22m 07s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c79568f02f334dfd94bb9c2dbca94d27 : SUCCESS in 21m 44s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/781ddc6fd512462e9abc49c4cf84c631 : SUCCESS in 13m 09s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/faa350583da540fb803ddcf6494357fc : SUCCESS in 11m 35s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/623e238a24cb40bea4dcba1ebb9271c4 : SUCCESS in 32m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/21b9657fc07e442f954b1bbd471d4b83 : SUCCESS in 21m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d90155b766394d44a51b7837a1b77b6e : SUCCESS in 5m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ac7f0587eeae4870a4c355cbd2f02151 : SUCCESS in 17m 56s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5f32e2fd97884ebfb3d8ac74f95433cf : SUCCESS in 9m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/08bb4e07bc054affb16c9733607562e9 : SUCCESS in 1h 05m 09s\n- tempest-full https://zuul.opendev.org/t/openstack/build/15516b47dd724cee9c38566889649a49 : SUCCESS in 1h 25m 47s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/aa75da116d364b78a85daef5e4175ef9 : SUCCESS in 1h 07m 24s","accounts_in_message":[],"_revision_number":3},{"id":"88711b4aea7e714c81c65700fc74962fca28b5e4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-24 12:06:10.000000000","message":"Patch Set 3: Code-Review+2\n\n\u003e \u003e \u003e I thin is issue https://github.com/jd/pifpaf/pull/124 is\n \u003e \u003e \u003e suspicious. could you please have a look at it? thx\n \u003e \u003e Hello,\n \u003e \u003e\n \u003e \u003e Have you faced this issue?\n \u003e \u003e\n \u003e \u003e From what I understood of this topic the main reason behind this\n \u003e \u003e patch is too keep version of pifpaf aligned between ussuri and\n \u003e \u003e train, and we don\u0027t observed similar issue on openstack, isn\u0027t?\n \u003e \n \u003e Hi the main reason is not to align version, the issue I am facing\n \u003e is the https://review.opendev.org/#/c/747892/ test kept failing\n \u003e which I thought related to the bug I link in the commit msg.\n\nThanks for the heads up :) sorry I didn\u0027t dig more than that on oslo.messaging and at first glance I didn\u0027t see the possible link between these patches.","accounts_in_message":[],"_revision_number":3},{"id":"81acc92360823a48148e0d4ae716ef182ba2f0f2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-09-24 15:38:24.000000000","message":"Patch Set 3:\n\nI\u0027m not seeing a link between the patches as well and that\u0027s the only real patch in https://github.com/jd/pifpaf/compare/2.2.2...2.4.0","accounts_in_message":[],"_revision_number":3},{"id":"6c154baa2eceb9b5a6050c16794e56acfec34dba","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-09-25 00:25:58.000000000","message":"Patch Set 3:\n\n\u003e I\u0027m not seeing a link between the patches as well and that\u0027s the\n \u003e only real patch in https://github.com/jd/pifpaf/compare/2.2.2...2.4.0\n\nI am sorry but I am confused, what link do you want? I think the commit https://github.com/jd/pifpaf/pull/124/commits/5f74964a1e01fbadc6cbee19503f31b37e200894 is clearly contained in this link https://github.com/jd/pifpaf/compare/2.2.2...2.4.0?","accounts_in_message":[],"_revision_number":3},{"id":"dbadbaeddbef46d8bc099ac968ef68cc964ba93c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-01 17:21:18.000000000","message":"Patch Set 3:\n\nthe reason I have not approved this is because I do not see what the comit you linked has to do with the review you linked (747892)","accounts_in_message":[],"_revision_number":3},{"id":"7aa1e6f2c7479cffc3ee394e93fa190b91a51a50","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-10-02 00:40:55.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"fbec5c74596cf95064c5180616c24a2c2ceaa2b9","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-10-02 00:44:51.000000000","message":"Patch Set 4:\n\n\u003e the reason I have not approved this is because I do not see what\n \u003e the comit you linked has to do with the review you linked (747892)\n\nI have updated link to the failed test case. Hope it makes more sense this time.","accounts_in_message":[],"_revision_number":4},{"id":"c2262ba2018e520f833139cb764b4cd5960e2541","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 02:46:34.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15d4da19ca0e4d35a0e4b899ccb3947e : SUCCESS in 3m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0ab23f1ed97c406fa00134121bd7cafa : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37fd3f2f853641bbad66f7ea3281aa31 : SUCCESS in 3m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/afe5ba8694fd4a9c9cbe82f90365c151 : SUCCESS in 4m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/06df2e75013747df9b9caf4ce7cdda0e : SUCCESS in 3m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f295bd092a6a4a979c024ffd344901e4 : SUCCESS in 3m 19s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/32643db626fd498391fd99a315254f9e : SUCCESS in 9m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/299a90b7e047430e963bf6dc95d37bd6 : SUCCESS in 9m 56s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/3a732a3f53ad432292c0575afe7771d7 : SUCCESS in 9m 02s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/76d3060b38e34d58a5b10b9762a51b1f : SUCCESS in 8m 39s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/51165a7687794f94af3c5a771ac829c4 : SUCCESS in 4m 31s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e50d1c70ef054bc895b9f5a7290b8105 : SUCCESS in 5m 21s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/8a9f4416479c4395b6decedcf1aaccca : SUCCESS in 11m 21s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fcf32298fd5e4c589498b0787e348350 : SUCCESS in 9m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7f158409a6044ea38c3ba08158281410 : SUCCESS in 8m 56s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/b03635b7c19d4665a00f8d53a50698b4 : SUCCESS in 19m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b4bfc8dbca894e6dbc670493bd667533 : SUCCESS in 19m 54s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/b52018ad08484be7af05afcaeed4b305 : SUCCESS in 20m 22s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/72983caf4d954be6a3cc90a657b28979 : SUCCESS in 22m 32s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/eb38218e99e44b23b9ca2b66c4304fc2 : SUCCESS in 14m 37s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/86e73f92d0894b1cbb705a17de17be4c : SUCCESS in 13m 30s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/743d88e6288a48d3b518c3022d12ed58 : SUCCESS in 12m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c906769186d14820a724b80aba8d5a5d : FAILURE in 29m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6476afd370954f77b8edfe91e0869d92 : SUCCESS in 5m 29s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/01d40ebc380e425dbd942f33556d6e8f : SUCCESS in 11m 28s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/898bffb7d3bd4a4fa32c9472ad282fa9 : SUCCESS in 11m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/158b37f9352a47e2abd87db2f0aa3d27 : SUCCESS in 1h 00m 19s\n- tempest-full https://zuul.opendev.org/t/openstack/build/278d69b1411d4a9c94eb5ce56800b6df : SUCCESS in 1h 24m 44s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/964f71b470274cebba2d87788cd3b3e5 : SUCCESS in 1h 01m 09s","accounts_in_message":[],"_revision_number":4},{"id":"6e5a20b0c9040d87c3f58a3b74f89c233ce493e2","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-10-02 03:24:22.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"ecea9bdccf91c5747bbee1035b70b60cfc60f7f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 05:13:10.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/985aef3fd5844c61af4a2724f03a615d : SUCCESS in 3m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9b1d9a5af7764f42a0bc3fd88663d8ed : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f29b6f324fe54b31838ee3f51e908f76 : SUCCESS in 3m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c0000a78fc754d04b798460ba4b451a2 : SUCCESS in 4m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d32673c7cf45414a8ca01ca3ef29bc40 : SUCCESS in 3m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/13e22ad9c0334f35a97c32bf90d30443 : SUCCESS in 3m 16s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/decd57f2e4d445ab858c11004684c4b2 : SUCCESS in 6m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/92e69e13eb604e34930e1bc450226b67 : SUCCESS in 7m 51s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/20c609a6255b4bce8be541f370e856d4 : SUCCESS in 9m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/06b170e1153341cead5e778e6806af62 : SUCCESS in 8m 19s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/cebf6a7b162a4ec791bfbad4399b1831 : SUCCESS in 4m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/00e57e8869f44614b99f645e78b9f40e : SUCCESS in 4m 09s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/8107972d83ce4aea9333b91ba83e0c54 : SUCCESS in 12m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a3c4891203794d79b62859215c564ac4 : SUCCESS in 10m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/35f0ebcc9f924f14ab4a63318b0ad057 : SUCCESS in 6m 19s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/41e65db1c909460ba84d4edfc6d57d9e : SUCCESS in 13m 01s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/96077db22b2b4406b5242f5ac53f1677 : SUCCESS in 12m 06s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/13c7b076bf19419b8dc33732ed3f50ad : SUCCESS in 20m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e44d3df6fa544ebe8df3dcf2eeffc6cf : SUCCESS in 20m 49s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/30dc0d8205464a5d9597581cdbe37943 : SUCCESS in 13m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/14c060429b194b00adb3f90648fc3bb1 : SUCCESS in 12m 33s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c43b180d9e064923a066f3fd72e365f3 : SUCCESS in 12m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e3f6e672e7634250ad90671f0bea1003 : SUCCESS in 19m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/17229a66b67442a3b32be209f1c84b4a : SUCCESS in 5m 21s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/beb88ce640994dc99f8cbeb3e8718bb4 : SUCCESS in 8m 46s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b45df25e13d64b0da7fa572ee416e5d5 : SUCCESS in 8m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/78f992f51cd44178a7d08f3a209a7c06 : SUCCESS in 1h 07m 34s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5da393d7e5114aa981a2384025259021 : SUCCESS in 1h 42m 03s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/02e76c392fc24d058878857a5931f848 : SUCCESS in 55m 27s","accounts_in_message":[],"_revision_number":4},{"id":"17b2786c7066ae9b11a2c88d97a42621d3dfc052","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-06 15:44:44.000000000","message":"Patch Set 4: Code-Review+2\n\n@Matthew: Any update about [1]?\n\n[1] https://github.com/jd/pifpaf/commit/90a9cc2fdbe167469a9a45800f49deca2f426547","accounts_in_message":[],"_revision_number":4},{"id":"56ae47ec18fc3e6d72fee9b279ae6f447ef96e50","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-10-07 10:38:29.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"22d86c5f2e1519f06cb51fca3fd138967c910e36","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-10-07 11:16:22.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"331e6a79d6b2ae6eca30b184acc5cd612c895c17","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-08 16:37:43.000000000","message":"Patch Set 4: -Code-Review\n\nWe did some test with pifpaf 2.4.0 and this patch and unfortunatelly it doesn\u0027t help us to solve the oslo.messaging issue.","accounts_in_message":[],"_revision_number":4},{"id":"121d7bf46c5a48c0e4507192b765bd836f37233a","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-10-09 01:30:42.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4},{"id":"d62df099f56208448e745a5ba06f37e1ed6f854c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-14 13:33:35.000000000","message":"Patch Set 4:\n\nFYI after lot of unsuccessful retries we successfully fixed our gate by adding a depends-on on this patch after only 2-3 retries, so we think that the SIGTERM patch improve the situation.\n\nWithout this \"depends-on\" all our retries failed, and at first glance we haven\u0027t difference between our branches on this part (patch/code)","accounts_in_message":[],"_revision_number":4},{"id":"8d96325c49cb70ae5bb7a7a0804075bedc56035f","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-10-14 14:51:39.000000000","message":"Patch Set 4:\n\n\u003e FYI after lot of unsuccessful retries we successfully fixed our\n \u003e gate by adding a depends-on on this patch after only 2-3 retries,\n \u003e so we think that the SIGTERM patch improve the situation.\n \u003e \n \u003e Without this \"depends-on\" all our retries failed, and at first\n \u003e glance we haven\u0027t difference between our branches on this part\n \u003e (patch/code)\n\nhmm.. so do you think it is valid to backport this...","accounts_in_message":[],"_revision_number":4},{"id":"e19eba2da0f434d1481678b13696c5a94a4025d1","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-10-15 00:04:03.000000000","message":"Restored\n\n...","accounts_in_message":[],"_revision_number":4},{"id":"8b4385e6139c9421f1c30041a089be69ca9f281f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 02:50:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4db1a68b4746469a9f58ad2618a0c09e : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3bd610db82ca43f4913e894e5a9d2b64 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/481ccb448f134ae8a2a9c11e8445beed : SUCCESS in 4m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/32068f8b220441e5a877d207564adfd4 : SUCCESS in 4m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/40cbe1c4e0274ea692df03c8eb17f26e : SUCCESS in 4m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cc14bf189108453f89b29c20c412d3e5 : SUCCESS in 4m 04s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/6bdb761e5ada4790bf604483cf1303eb : SUCCESS in 7m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c42c40d766b74ccfa5bffffb19c718df : SUCCESS in 7m 48s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c6cdb96c848746fda8d30f1cad4b5201 : SUCCESS in 9m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/eb08abf54e37430d8acece3b66b607b8 : SUCCESS in 10m 36s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/05f3ecf52ad2432d966fc9fa590d23c9 : SUCCESS in 5m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/75a200dfc4e945bd91c86e8b66284f62 : SUCCESS in 5m 18s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/3d0cf42c89db409a974ee497c232fa5e : SUCCESS in 11m 46s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6a7089e4fcc84e5383e5c6c928c89559 : SUCCESS in 9m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e435182851e445acb83e781167f9bd4b : SUCCESS in 7m 10s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/67d8c8a24d3c4183a9edbacfee86cd32 : SUCCESS in 17m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c8858a169ae24027a58b7561dc811786 : SUCCESS in 15m 58s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/3079b8ac235d4ad8b4f2c66933405dd1 : SUCCESS in 24m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/78297f3f5297451e8db30836c0691eb2 : SUCCESS in 23m 33s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/47f1687b8409433484dd9fe15e870dea : SUCCESS in 14m 21s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2166cc70822b428abea9814155a7532b : SUCCESS in 13m 36s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b832e33182884241a66bfe7e6578722d : SUCCESS in 12m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/796f5298b6dd462aa85cff91b5256b6d : SUCCESS in 20m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5fff53afd29f491c8788b13be676420e : SUCCESS in 6m 07s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9fca6451f01342ebaf018ee655201815 : SUCCESS in 10m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ec70725b995540249b07d50c119a8ed8 : SUCCESS in 10m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/dd763b7d304642a683f77b349780bd44 : SUCCESS in 1h 10m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/db51ba029559406a941a0fd30315b0b9 : SUCCESS in 1h 30m 34s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/776318d88144461bb5f1c85ffb5110d8 : SUCCESS in 1h 00m 37s","accounts_in_message":[],"_revision_number":4},{"id":"1cb5bf7ecfb0719f4344e4cbf67e288269fd07c1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-15 07:53:08.000000000","message":"Patch Set 4:\n\n\u003e Restored\n \u003e \n \u003e ...\n\nThanks, Daniel (damani) and myself will bring more context and debug info to help us to move further on this topic.","accounts_in_message":[],"_revision_number":4},{"id":"36d9771a23607352eb05c342f8845d9770e62ba3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-21 11:15:46.000000000","message":"Patch Set 4: Code-Review+2\n\nHello,\n\nWithout this patch/bump we get lot of errors [1] on oslo.messaging [2].\nWe tested twice to add a depends-on on this patch and we fixed our CI twice with that [3].\n\nPlease Mathew without wanting to harass you, can you take a new look to this topic, the oslo.messaging patch is stuck since August.\n\n[1]\n```\n2020-10-12 09:10:43.723605 | ubuntu-bionic |\n\n2020-10-12 09:10:43.723623 | ubuntu-bionic | Captured traceback:\n\n2020-10-12 09:10:43.723642 | ubuntu-bionic | ~~~~~~~~~~~~~~~~~~~\n\n2020-10-12 09:10:43.723660 | ubuntu-bionic |     Traceback (most recent call last):\n\n2020-10-12 09:10:43.723679 | ubuntu-bionic |       File \"/home/zuul/src/opendev.org/openstack/oslo.messaging/.tox/py27-func-rabbit/lib/python2.7/site-packages/fixtures/fixture.py\", line 197, in setUp\n\n2020-10-12 09:10:43.723698 | ubuntu-bionic |         self._setUp()\n\n2020-10-12 09:10:43.723716 | ubuntu-bionic |       File \"/home/zuul/src/opendev.org/openstack/oslo.messaging/.tox/py27-func-rabbit/lib/python2.7/site-packages/pifpaf/drivers/rabbitmq.py\", line 136, in _setUp\n\n2020-10-12 09:10:43.723735 | ubuntu-bionic |         self.start_node(n1)\n\n2020-10-12 09:10:43.723753 | ubuntu-bionic |       File \"/home/zuul/src/opendev.org/openstack/oslo.messaging/.tox/py27-func-rabbit/lib/python2.7/site-packages/pifpaf/drivers/rabbitmq.py\", line 86, in start_node\n\n2020-10-12 09:10:43.723772 | ubuntu-bionic |         wait_for_line\u003d(\"completed with .* plugins\"))\n\n2020-10-12 09:10:43.723791 | ubuntu-bionic |       File \"/home/zuul/src/opendev.org/openstack/oslo.messaging/.tox/py27-func-rabbit/lib/python2.7/site-packages/pifpaf/drivers/__init__.py\", line 259, in _exec\n\n2020-10-12 09:10:43.723810 | ubuntu-bionic |         % (wait_for_line, b\"\".join(lines)))\n\n2020-10-12 09:10:43.723828 | ubuntu-bionic |     RuntimeError: Program did not print: `completed with .* plugins\u0027\n\n2020-10-12 09:10:43.723847 | ubuntu-bionic |     Output: ERROR: node with name \"pifpaf-1\" already running on \"localhost\"\n```\n\n[2] https://review.opendev.org/#/c/747892/3\n[2] https://review.opendev.org/#/c/747892/4","accounts_in_message":[],"_revision_number":4},{"id":"b28bb269e06f51d3007d0ad6f39276a24ec0dfbe","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-07 01:20:06.000000000","message":"Patch Set 4:\n\nis this still needed?","accounts_in_message":[],"_revision_number":4},{"id":"99af11bb301f9df6db2a805db73ae93bac348372","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-09 13:41:06.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 4:\n\u003e \n\u003e is this still needed?\n\nOn oslo.messaging we got around the problem by disabling functional tests, but I think we could  answer yes.","accounts_in_message":[],"_revision_number":4},{"id":"74574a5f0ed9a9220a6b16984c82d8c808128e4b","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-07-18 23:28:57.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9ecd6d77e03e05a425f4ae8714f2f4ccb7b2430e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-08-02 05:19:07.000000000","message":"Abandoned\n\nAs per: https://review.opendev.org/c/openstack/requirements/+/753239/4#message-74574a5f0ed9a9220a6b16984c82d8c808128e4b\n\nI\u0027m abandoning this.  It it is still needed we can restore it.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d811cc3746dbf2c35c696f1dd6a9b760c72722e9","revisions":{"1442eebf0e059816a7cf52a9095c230da0dc6309":{"kind":"REWORK","_number":1,"created":"2020-09-22 07:25:34.000000000","uploader":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"ref":"refs/changes/39/753239/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/39/753239/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/753239/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/753239/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/753239/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/39/753239/1"}}},"commit":{"parents":[{"commit":"0cfc2ef6318b639bb4287fa051eae8f3ac8cc426","subject":"update constraint for python-troveclient to new release 3.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0cfc2ef6318b639bb4287fa051eae8f3ac8cc426"}]}],"author":{"name":"shenjiatong","email":"yshxxsjt715@gmail.com","date":"2020-09-22 07:22:04.000000000","tz":480},"committer":{"name":"shenjiatong","email":"yshxxsjt715@gmail.com","date":"2020-09-22 07:25:20.000000000","tz":480},"subject":"Bump up pifpaf version for train","message":"Bump up pifpaf version for train\n\nThis ps bumps up pifpaf version to be same\nas stable/ussuri.\n\nChange-Id: I140f89beb2e73740d9c56d090fc8a6b1b32b3c81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1442eebf0e059816a7cf52a9095c230da0dc6309"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1442eebf0e059816a7cf52a9095c230da0dc6309"}]},"branch":"refs/heads/stable/train"},"efd58d3f0001b76dcce5785167fbd57a659dc4f2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-09-23 09:38:53.000000000","uploader":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"ref":"refs/changes/39/753239/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/39/753239/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/753239/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/753239/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/753239/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/39/753239/2"}}},"commit":{"parents":[{"commit":"0cfc2ef6318b639bb4287fa051eae8f3ac8cc426","subject":"update constraint for python-troveclient to new release 3.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0cfc2ef6318b639bb4287fa051eae8f3ac8cc426"}]}],"author":{"name":"shenjiatong","email":"yshxxsjt715@gmail.com","date":"2020-09-22 07:22:04.000000000","tz":480},"committer":{"name":"shenjiatong","email":"yshxxsjt715@gmail.com","date":"2020-09-23 09:34:34.000000000","tz":480},"subject":"Bump up pifpaf version for train","message":"Bump up pifpaf version for train\n\nThis ps bumps up pifpaf version to be same\nas stable/ussuri.\n\nPlease take a look this issue [1] which is\nfixed after 2.4.0.\n\n[1] https://github.com/jd/pifpaf/pull/124\n\nChange-Id: I140f89beb2e73740d9c56d090fc8a6b1b32b3c81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/efd58d3f0001b76dcce5785167fbd57a659dc4f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/efd58d3f0001b76dcce5785167fbd57a659dc4f2"}]},"branch":"refs/heads/stable/train"},"32273399eda3074044820cdda57c87836ce262e2":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-09-24 00:29:18.000000000","uploader":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"ref":"refs/changes/39/753239/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/39/753239/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/753239/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/753239/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/753239/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/39/753239/3"}}},"commit":{"parents":[{"commit":"0cfc2ef6318b639bb4287fa051eae8f3ac8cc426","subject":"update constraint for python-troveclient to new release 3.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0cfc2ef6318b639bb4287fa051eae8f3ac8cc426"}]}],"author":{"name":"shenjiatong","email":"yshxxsjt715@gmail.com","date":"2020-09-22 07:22:04.000000000","tz":480},"committer":{"name":"shenjiatong","email":"yshxxsjt715@gmail.com","date":"2020-09-24 00:23:43.000000000","tz":480},"subject":"Bump up pifpaf version for train","message":"Bump up pifpaf version for train\n\nFuntional test of oslo.messaging [1] kept\nfailing. looks like it is related to issue [2]\nwhich fails to clean up rabbitmq cluster when\none unittest done. It is fixed after 2.4.0.\n\nThis ps bumps up pifpaf version to be same\nas stable/ussuri.\n\n[1] https://review.opendev.org/#/c/747892/\n[2] https://github.com/jd/pifpaf/pull/124\n\nChange-Id: I140f89beb2e73740d9c56d090fc8a6b1b32b3c81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/32273399eda3074044820cdda57c87836ce262e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/32273399eda3074044820cdda57c87836ce262e2"}]},"branch":"refs/heads/stable/train"},"d811cc3746dbf2c35c696f1dd6a9b760c72722e9":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-10-02 00:40:55.000000000","uploader":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"ref":"refs/changes/39/753239/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/39/753239/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/753239/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/753239/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/753239/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/39/753239/4"}}},"commit":{"parents":[{"commit":"0cfc2ef6318b639bb4287fa051eae8f3ac8cc426","subject":"update constraint for python-troveclient to new release 3.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0cfc2ef6318b639bb4287fa051eae8f3ac8cc426"}]}],"author":{"name":"shenjiatong","email":"yshxxsjt715@gmail.com","date":"2020-09-22 07:22:04.000000000","tz":480},"committer":{"name":"shenjiatong","email":"yshxxsjt715@gmail.com","date":"2020-10-02 00:36:38.000000000","tz":480},"subject":"Bump up pifpaf version for train","message":"Bump up pifpaf version for train\n\nFuntional test of oslo.messaging [1] kept\nfailing. By looking at the unittest log [2],\nit kept failing because pifpaf fixtures fails\nkilling the cluster after one unitext completes.\nIt is related to issue [3]\nwhich fails to clean up rabbitmq cluster when\none unittest done. It is fixed after 2.4.0.\n\nThis ps bumps up pifpaf version to be same\nas stable/ussuri.\n\n[1] https://review.opendev.org/#/c/747892/\n[2] https://zuul.opendev.org/t/openstack/build/3512e9c53b59406abaf5e5aa15531d12/console\n[3] https://github.com/jd/pifpaf/pull/124\n\nChange-Id: I140f89beb2e73740d9c56d090fc8a6b1b32b3c81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d811cc3746dbf2c35c696f1dd6a9b760c72722e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d811cc3746dbf2c35c696f1dd6a9b760c72722e9"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
