)]}'
{"id":"openstack%2Frequirements~754180","triplet_id":"openstack%2Frequirements~master~Ieb278dff5cd155cdc51d312d9d490411456866b3","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ieb278dff5cd155cdc51d312d9d490411456866b3","subject":"update constraint for sushy to new release 3.4.1","status":"MERGED","created":"2020-09-24 16:58:10.000000000","updated":"2020-09-25 05:55:14.000000000","submitted":"2020-09-25 05:53:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"754180-1601013237708-e805a381","meta_rev_id":"bb40e87e96572f9f0fa5c8fb7b95bef2ffcc7b11","_number":754180,"virtual_id_number":754180,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-25 05:53:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-09-25 04:22:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-09-24 17:27:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-09-25 04:22:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-24 17:27:11.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-25 04:22:01.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-09-25 05:53:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"669dd9728014127f3714f63c8b84dcc2c5aee92c","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-24 16:58:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a0ee1583a8a9739ce1e5c7a30865081a21ba47a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-24 17:27:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"23208b413e448e2f7e29a13be95d3970c43c8c0c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 18:20:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/de2c2431d05d4e5d931e3cec71755752 : SUCCESS in 1h 12m 03s","accounts_in_message":[],"_revision_number":1},{"id":"7e0fb9dcec62aff858ce03370a2fd4b67a12e6bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 00:23:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/285dd31da40141638cbbce05b6bb4e9c : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/073134f082534a089ff82eb960f3839c : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0dda9cdfc8d24485879df4178257371d : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/daab3110407c4ab989f1dda82aacb8f0 : SUCCESS in 4m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e9857bdf9a99401f965186960b901649 : SUCCESS in 4m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ba6741bec2f34912ad9135763f367dcd : SUCCESS in 3m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9b231b93726c45b58fd16321df1f57b8 : SUCCESS in 7m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/82432566daa9417a8d78ffe958949067 : SUCCESS in 6m 54s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f212906374434c50a991294d1c9ed070 : SUCCESS in 9m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fec874469e6e4568ab82cc90e4443ac7 : SUCCESS in 4m 54s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/53e1688783f34a9db461470afd0eea47 : SUCCESS in 10m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/369772bfbd2f43a19e3a6c1f0ce29586 : SUCCESS in 8m 01s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6c0f88a92fde45f8830cdda3b126507d : SUCCESS in 9m 16s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/68eea932cef84cedbdea645828444ece : SUCCESS in 13m 54s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/12e27c9dbcb946968005adcc2d272549 : SUCCESS in 26m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/17da241156aa41cfb5ea82e0fe2afbfa : SUCCESS in 14m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6afce10174bb486880934cd790783811 : SUCCESS in 4m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cd7f2cc9b93b422dbde98abfbf68964a : SUCCESS in 19m 05s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/842d94fe10564a8e91293681c3454709 : SUCCESS in 5m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bbac106794784e81ab9310f68a3d9eff : SUCCESS in 6m 56s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fb4776447ada41b4899420ae52e529ac : SUCCESS in 12m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cea74b6efb8143fea8fa1da98735907d : SUCCESS in 54m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/59e0594e801149468e6997fcfa2dc7f7 : SUCCESS in 1h 19m 21s","accounts_in_message":[],"_revision_number":1},{"id":"3f0e4cc9724b399008d2b86d5bed0d4d5e5a3ed0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-09-25 04:22:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nI said OK in the releases channel already","accounts_in_message":[],"_revision_number":1},{"id":"af5f30462b9f71b58f2a865f4bfa8dab9d3a578f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 04:22:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"890e47d91f25832a3886057ba6555588e6def396","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 05:53:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e8b0342212c229e3b846b43c6bc227dd93daa7e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 05:53:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8a7ed24fc714910a107dfe331b45e3b : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c792619ec004aea8cc38fe9a0d9fbb1 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cea367389a034039ae96bed8f796effa : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a2fbf2630164915af9d4d5e5be263d9 : SUCCESS in 4m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/59ebc5905d8e432da18c628de240f1ed : SUCCESS in 4m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/864a396bfec14174bcaa12007aa7c389 : SUCCESS in 4m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bbdf0259588b40798e9dfbb3e9819a3c : SUCCESS in 6m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3416132271d4428ba8606cfdc6bd9378 : SUCCESS in 8m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2ac95a4ba5464eb4be721764e43a5712 : SUCCESS in 8m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4a16da9118c347aa8803de83b55da829 : SUCCESS in 4m 57s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3efe1a4d22314b87ba1a9339294fc714 : SUCCESS in 10m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cc62c70e53f14b9a8aeb95ac91a4b264 : SUCCESS in 11m 36s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b5aab9e918b04f93b7a47bf927dc6d7b : SUCCESS in 6m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2c3d0221f1744211989ae93059b1f924 : SUCCESS in 15m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b31eb559d8a64ca28c02a308863eafad : SUCCESS in 22m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2da0b5dde2b64a0d89869e309655372f : SUCCESS in 14m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9522f01ab72b46148a891b9a62b91386 : SUCCESS in 4m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/08132b6058e049ae950c95e4dad992ae : SUCCESS in 17m 05s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9295abe5324f464a8bc5a1a055fe6607 : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1025fefcd2154bab94896e6b21b72244 : SUCCESS in 8m 08s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b2cfed160b344d8081c387635f341819 : SUCCESS in 12m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b5acb88d3d5e4081bf93f4d373b51366 : SUCCESS in 56m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/946cb8479071439fa91b04060a977aab : SUCCESS in 1h 01m 55s","accounts_in_message":[],"_revision_number":1},{"id":"972c9d68342f918ee6dedcc51439ee517f667633","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 05:55:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d358e3d4aa5417eba3ad6a43c82b7f6 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0ab52034a436d63b336c1ea6d4d9952903c5c444","revisions":{"0ab52034a436d63b336c1ea6d4d9952903c5c444":{"kind":"REWORK","_number":1,"created":"2020-09-24 16:58:10.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/80/754180/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/754180/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/754180/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/754180/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/754180/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/754180/1"}}},"commit":{"parents":[{"commit":"41d9437e24935d18d811e5640573187cde3aa2cf","subject":"Merge \"update constraint for tempest to new release 25.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/41d9437e24935d18d811e5640573187cde3aa2cf"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-24 16:58:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-24 16:58:08.000000000","tz":0},"subject":"update constraint for sushy to new release 3.4.1","message":"update constraint for sushy to new release 3.4.1\n\nChange-Id: Ieb278dff5cd155cdc51d312d9d490411456866b3\nmeta:version: 3.4.1\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta:release:Commit: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta:release:Change-Id: I1c5bb794f75fa5c17f780cc16cadb2348f782686\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0ab52034a436d63b336c1ea6d4d9952903c5c444"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0ab52034a436d63b336c1ea6d4d9952903c5c444"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
