)]}'
{"id":"openstack%2Frequirements~754197","triplet_id":"openstack%2Frequirements~master~I41d647ffdeacb2ee6a2a8d00c32f0b406a3ab63d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I41d647ffdeacb2ee6a2a8d00c32f0b406a3ab63d","subject":"update constraint for openstacksdk to new release 0.50.0","status":"MERGED","created":"2020-09-24 17:00:52.000000000","updated":"2020-09-25 06:13:17.000000000","submitted":"2020-09-25 06:10:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"754197-1601014214118-1561717f","meta_rev_id":"8acbdc0524b204532848807eddb515376c1deabe","_number":754197,"virtual_id_number":754197,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-25 06:10:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-09-25 04:22:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-09-24 17:27:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-09-25 04:22:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-24 17:27:27.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-25 04:22:03.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-09-25 06:10:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7d69e7f79775aed8211e45963c8f67d26952eff3","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-24 17:00:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ce85f2c3be5d754e8e5c3a9aeafaf72edcea1cd","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-24 17:27:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"556894968537fe027513b0cb2e15cffe8c8f4c5d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 18:48:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/df614104d0dd4b04aaa536dad78bd3ed : SUCCESS in 1h 04m 19s","accounts_in_message":[],"_revision_number":1},{"id":"4b61858234849d81b44780af22418f71534dcca7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 00:46:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b379b26c307042ada84432ffe77487ea : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dcb55e4bf824dc18db5927ae52a2199 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/108dfd9425d9481e93400872d019188c : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c30e40162dd5435da18979c177f55347 : SUCCESS in 4m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/db49703d080b4a51aefffe74c1c25529 : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/14ae87e5fb4444f9b2244a1afc2a1b26 : SUCCESS in 3m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9e48e5dffe534405ba3527ab9c1e43b5 : SUCCESS in 6m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a5ff2d7ada0c4808972f68b11085c3c0 : SUCCESS in 6m 54s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/236316896a664e70af1175f5da65812a : SUCCESS in 9m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7c33e13c2d22412d86ae6414df88c281 : SUCCESS in 6m 46s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8a5ad80247264018a7d76c2814a8351c : SUCCESS in 9m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/24a4a41c01584e1a844e0e8638ac3a1f : SUCCESS in 9m 31s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4051aa0f02084ffc96dc11db7ed8e25b : SUCCESS in 6m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/79b86b14813542b3a4b82a692241b02c : SUCCESS in 12m 36s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/888289c9a63144a185127d18442901bf : SUCCESS in 25m 04s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7ac4c7347ece475d9fee9a9fa3fd5cd9 : SUCCESS in 12m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7d534fc99c6d43159dc8d5f28e30c22d : SUCCESS in 4m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bf4cd83e174446bfac60e0b203f39dcf : SUCCESS in 16m 59s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ab3831b1c62d49cca0f3bfcf4981ed7d : SUCCESS in 5m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bb8dbd94040c4c8b946c220998e81107 : SUCCESS in 6m 00s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1bc43d3c3d754110b80512dbe06d77fd : SUCCESS in 9m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/32a00ecca22e45288430d5c9f9c9c41f : SUCCESS in 59m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b378ad451ffd4234ab795fdd6e339b04 : SUCCESS in 1h 35m 17s","accounts_in_message":[],"_revision_number":1},{"id":"fb448c7f5e67743b82d8c6b5fb146487b4506763","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-09-25 04:22:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nI said OK in the releases channel already","accounts_in_message":[],"_revision_number":1},{"id":"6ac335e2d631415ae07e036c4f8e6617a6d96e09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 04:22:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"826ce7a909bfa28e33b778f1097ae029d98bdc55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 06:10:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e95a98ee8b014a1298342db9e52eaafd : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b20bc9c03c048b29fcd9c8786934eaf : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97d6a076183d41e5b5b7e48e99250b84 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec5de0ae89d64825b927fdb8ae0f64aa : SUCCESS in 4m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c967ebc5c1424207bb8619df72e84c7d : SUCCESS in 4m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5f28861ac76c4f90b124beaf0779bb96 : SUCCESS in 4m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/61faf45129334a11975d68cd2cac2b16 : SUCCESS in 7m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f371a8a1e4944a54b22af6ae9e227b2a : SUCCESS in 8m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d8ddb66e97f54891b3f069c7cb2bfb87 : SUCCESS in 10m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7c7e177f15a64a2fa5ec3dcf6eab7bce : SUCCESS in 4m 54s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ac7ef44634554353a79844b6387e582d : SUCCESS in 10m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/719b75c2ebbf4b498e64404a26b5b11e : SUCCESS in 6m 24s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4b904e7d28c24b0f8d6884d8b33a9a53 : SUCCESS in 7m 30s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/042ae3683a554a69a085de69f5be3553 : SUCCESS in 14m 02s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/69fb8bd7574e44f6b15b224e057b4523 : SUCCESS in 26m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4c4013f8f0214110b5129c69e5bdb589 : SUCCESS in 13m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/41902d0ed83743cf8674ae16a8596014 : SUCCESS in 3m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/61f18ea590eb4c4aaeca6e5c9cbb33c6 : SUCCESS in 19m 15s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1e92af0b048b4d94852d5e4cab6d6c24 : SUCCESS in 5m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/16702567ee45434a872eb37ec609daf8 : SUCCESS in 7m 01s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2461ca83f4554e3b867b03512521e3f9 : SUCCESS in 9m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/56870392634b4a108faea18720057b98 : SUCCESS in 1h 06m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5c7ccbf1fb054144ba64c8bc8dfb6604 : SUCCESS in 1h 17m 23s","accounts_in_message":[],"_revision_number":1},{"id":"cb7f12342ac2a0ef9c855b07613818bbfd3e8d28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 06:10:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d993ad126299db3be975021fdbec687f7f744eab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 06:13:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/658b098a8b9e46f0a42b39c16aed52c9 : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"714c7e29f5bf7e9fc11db886e3a2c55f6e121c23","revisions":{"714c7e29f5bf7e9fc11db886e3a2c55f6e121c23":{"kind":"REWORK","_number":1,"created":"2020-09-24 17:00:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/97/754197/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/97/754197/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/754197/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/754197/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/754197/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/97/754197/1"}}},"commit":{"parents":[{"commit":"41d9437e24935d18d811e5640573187cde3aa2cf","subject":"Merge \"update constraint for tempest to new release 25.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/41d9437e24935d18d811e5640573187cde3aa2cf"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-24 17:00:49.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-24 17:00:49.000000000","tz":0},"subject":"update constraint for openstacksdk to new release 0.50.0","message":"update constraint for openstacksdk to new release 0.50.0\n\nChange-Id: I41d647ffdeacb2ee6a2a8d00c32f0b406a3ab63d\nmeta:version: 0.50.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta:release:Commit: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta:release:Change-Id: Iccfa003dba03f7704f1a60b21331f2663bc97d60\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/714c7e29f5bf7e9fc11db886e3a2c55f6e121c23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/714c7e29f5bf7e9fc11db886e3a2c55f6e121c23"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
