)]}'
{"id":"openstack%2Frequirements~755608","triplet_id":"openstack%2Frequirements~stable%2Fussuri~I4a90ffbc9b6fd06547ec14606a61b935b15f8564","project":"openstack/requirements","branch":"stable/ussuri","topic":"new-release","hashtags":[],"change_id":"I4a90ffbc9b6fd06547ec14606a61b935b15f8564","subject":"update constraint for oslo.privsep to new release 2.1.2","status":"MERGED","created":"2020-10-01 18:24:35.000000000","updated":"2020-10-06 08:43:18.000000000","submitted":"2020-10-06 08:43:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755608-1601973798197-2bbe566a","meta_rev_id":"d57d74d427f5b305d9083e9e1611cf4a9d5be99b","_number":755608,"virtual_id_number":755608,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-10-06 01:13:15.000000000","_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-06 08:43:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-10-01 18:42:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-10-02 10:15:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":2,"date":"2020-10-01 20:46:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":1,"date":"2020-10-01 20:46:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-01 18:42:18.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-10-01 20:46:58.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-10-02 10:15:59.000000000","updated_by":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"reviewer":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"state":"REVIEWER"},{"updated":"2020-10-06 01:13:15.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-10-06 08:43:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"675904ae14aaf77f7ecaf6624270bad6e0c26526","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-01 18:24:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"243aa22936ba98048f465184d5ee902cf44fbac5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-01 18:42:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e03fa33f6fd1a274739f2e258081a63f42660eca","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-01 20:46:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"02627cb9f8fd90530cc0068621cd47159f92d989","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 00:28:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1052643807cc460e8fe79a2f3ab4a82a : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f1680558dff424483b9cb0c80cfdd50 : SUCCESS in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/98c50a8900d24d9aa08619c857cd84ac : SUCCESS in 5m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/667f4e5d5b844d93b3a29f26474dc0fc : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d4ad18426d734fda8882ea2ef5165f9b : SUCCESS in 3m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0d7af0677fa24db7814a2f2a3a6863f4 : SUCCESS in 6m 25s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/387fecfbfda540c481812204cc78018e : SUCCESS in 8m 06s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f9781c37e76b4c1d8d2aaa57a3cd820a : SUCCESS in 4m 17s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/51d9b90c84ca443ca3613b409a38b4be : SUCCESS in 11m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e002cc20470d4faa9ffe199d17dd0231 : SUCCESS in 7m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e4ee95a0428b4390917b5141a7cfec02 : SUCCESS in 17m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/19c713c20d24430a9da0059280a3a07f : SUCCESS in 33m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b1b74fe87cb04d0788455d0568bdb9e6 : SUCCESS in 12m 38s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ca89f1db76dc4e628ecb35c1c1a9659a : SUCCESS in 22m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8bd416cfa8c94291b107c774aa4dbe4c : SUCCESS in 18m 41s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d79c27c5161149e980bb1e8e53310645 : SUCCESS in 4m 13s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/a28df4ac7bcb49ac8168e049e5e8ee42 : SUCCESS in 6m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3e08751ca0094fe6b05f707602cd89c0 : SUCCESS in 6m 46s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/726079e497054fd9912ca767b4811807 : SUCCESS in 10m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/38e7a00d24df4ce4a05cea96e76b3f38 : SUCCESS in 10m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b8d1a39560a54c459f23d1dd693c3b23 : FAILURE in 1h 01m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/693fbe853a494d0ab075fcf97a496d4b : FAILURE in 1h 04m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3e8b668c05044faf8f310c14c9d22238 : FAILURE in 16m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/75ea4bf7fefb422ab57b1e3980676db0 : FAILURE in 1h 01m 01s","accounts_in_message":[],"_revision_number":1},{"id":"944d4f2ac2580309f0c223217ae0a6a7ee85de93","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-10-02 10:15:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"381ad8ff489ec91eeda47152e8512ec8f54c5153","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-10-02 10:15:59.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"54a81a5cefa05124d57cc24d9227715b12066751","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 16:32:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f47f7fc978e442e8ba7880d64fc88ba6 : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13aa8621b23b48528f0b8f712f9f0fca : SUCCESS in 5m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ba79243237c44786869e7d844e58adfa : SUCCESS in 4m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/55de286abedc423facf3c57b10c38d40 : SUCCESS in 4m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/48870db41e9b40f4910de4444acfa243 : SUCCESS in 4m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9c9f48999b5643e8a31da748cd433644 : SUCCESS in 8m 41s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/05602fe906bf46cab27979accda4eb5c : SUCCESS in 9m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ff938a7450624a01b1290fbd56649b4f : SUCCESS in 4m 57s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/57b2eb42a35e48918b76c2dd1d25554d : SUCCESS in 10m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7b395a0162a24bef84e625676aa74d2b : SUCCESS in 6m 10s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/eeb145595d33437193356e6de76f0177 : SUCCESS in 12m 18s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/354b6451b0bf4799a9b5f536030b6a06 : SUCCESS in 27m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1b73cdf2af69496783efdb972806f6ba : SUCCESS in 12m 45s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/2848c26ad34b423b8521fc4cc4883afb : SUCCESS in 14m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1cad86dba2ea489eb4411961d584aa9d : SUCCESS in 16m 18s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/7486d05230c54d89b7ecfeff4b6c5489 : SUCCESS in 4m 35s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/fbb3a04f7ccb40f2b2fbe18e854914ae : SUCCESS in 5m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7f10b72defba4b39b9235c065f7d64d2 : SUCCESS in 6m 25s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9f964a0cc1504a779083716661f39fd0 : SUCCESS in 9m 13s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b6ac48989986438e92e04bbc282d68cf : SUCCESS in 9m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2f2d1479783a4b798ee4f725c412ef07 : FAILURE in 43m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/55bfa1b8d3c1484a9c56db2a9e1bbfa5 : SUCCESS in 1h 16m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e4d9e16061874ad9a9e339a655e59870 : SUCCESS in 1h 25m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/19d0bd8473a14ca9bb44f67bff611e20 : SUCCESS in 1h 16m 30s","accounts_in_message":[],"_revision_number":1},{"id":"2056cf998eaf3347d7442ff3d63ab78201c7bed4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-05 15:26:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4f566424134255eaf141f235c492794091c3dddd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 20:34:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4d01745ed3443af818b05ba161e4ffe : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3908e7d2300549af9e3fcc98b7297002 : SUCCESS in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c851e41526c945b8ba26263a1e5a687a : SUCCESS in 5m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/74ed9e36f5c5405b91f6ead582002ea2 : SUCCESS in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6cec2e43700542b3939fd05b281b4318 : SUCCESS in 3m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a95bc047ff1b4456bf30807098cdd56d : FAILURE in 11m 25s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7b2fc28179d747db934b13a4a003a3b0 : SUCCESS in 8m 35s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c69ddb80b9c34715bd490716b1f20713 : SUCCESS in 5m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/636b9cdd50cb49029943c0dd8b66bc7d : SUCCESS in 13m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a0b40f0606e642249d68191b2b9225a1 : SUCCESS in 9m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2de870b198b04200a5b38c0d8aad0067 : SUCCESS in 20m 30s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4ee2546d7bef445ba5e2168684ff448f : SUCCESS in 29m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/14c33ac0c67d4188b09edbe4e2c6bdfb : SUCCESS in 18m 21s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/55c50bda635b4ec79de2ae463e9f64b6 : SUCCESS in 17m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2caf0524f4aa4086aaf240c122b18536 : SUCCESS in 17m 18s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/606f20523e9d4ce2807d08ab1ebcb177 : SUCCESS in 4m 36s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/351a81a996e740c185766453a5b7b1a1 : SUCCESS in 8m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/12c2ab25f5a74e92a5f296806fb2865a : SUCCESS in 9m 26s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/670a5fec9d6d497bb4f74f3f22c065d8 : SUCCESS in 14m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6480ceaf24fd4bc0b12cb2611fd8c9af : SUCCESS in 14m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f097f97beef4c6ba0246d3e517cb1d2 : SUCCESS in 54m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/56f53b9aaefb46e7b1f8d11b29f6818d : SUCCESS in 1h 26m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b445c2b27977441387813e5b1f152849 : SUCCESS in 1h 08m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/87cf0b1aecd441b498004c3be5ad42d4 : FAILURE in 24m 25s","accounts_in_message":[],"_revision_number":1},{"id":"f3870e2f7f237ab42e9dc89bad45569feb58f038","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-06 01:13:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1c1f9793fe844e88969b597f9cff76d03b0df929","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 03:05:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffa3ca39966f45f8aae3f1e4144460bf : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2bffd63262b14149a1fbfd9beee30edb : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d26f3dd414f5458e9dea82fb146af73e : SUCCESS in 4m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9f78192f6a3f48dd86f272e0ebb31361 : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cb21d7599c2848728aa1e5108a8df56a : SUCCESS in 4m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f8463369a3b549dea02dc3a14e92e4f9 : SUCCESS in 6m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/23ba4da8823448c6bca945ec09aa99e4 : SUCCESS in 9m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b21a6e67c8d14d10ae0c3d53565e5b2f : SUCCESS in 5m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4a31c54ccfe44578ae9518758cec39e1 : SUCCESS in 8m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/44c5340a553e4393b69dc64a83fd0ba2 : SUCCESS in 16m 26s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/447645bd75ba4d96b28107e72881c005 : SUCCESS in 15m 08s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/539d9a8d46274a45bff78e31b9574da1 : SUCCESS in 26m 26s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9f93092837d249e7bf5d0e112543b46b : SUCCESS in 15m 20s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/0a45c7545fe049c199a89f6504519ac9 : SUCCESS in 15m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1e5604f52e744983bccede0e923abc07 : SUCCESS in 21m 47s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ef782c54962f40a09e027d979ed88977 : SUCCESS in 4m 45s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/3cd2b9f8391a4bc19db896fea47eabe5 : SUCCESS in 5m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6e15fea8709247b9856545473310983d : SUCCESS in 7m 28s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b76427c2e1bc4260a1adb0a848988805 : SUCCESS in 10m 30s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/873de48403b148188b121c3e72b95368 : SUCCESS in 11m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f28caabf134347abaa2bd98672db7f3a : SUCCESS in 1h 04m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3396edc51e30483998cb84e0f04c78e5 : SUCCESS in 1h 03m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b8e03c2072ed47499250fee111634bae : SUCCESS in 1h 29m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/9d11823af57c4cb9918aa0c56f99ebdf : SUCCESS in 1h 30m 20s","accounts_in_message":[],"_revision_number":1},{"id":"66f552146d23d010dc1ab5ee2570940fb5081dc7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 03:05:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"85844a150870f3ab0531428099c0cc49e814dc0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 08:43:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bc563a61cc044a2886d52399f658378 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e017fa444f744348201c1aa3653e9ee : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/76d52cd354c1454bbc7332a3867025c0 : SUCCESS in 4m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3fe4c7e904c14f36a641c6f90cee7239 : SUCCESS in 4m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c253e3438a9343d3917ee73ad364f329 : SUCCESS in 4m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0264ca8072ca48fca02bc1f49c6d2c50 : SUCCESS in 9m 05s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b0c6e2bb8f3d48ab82ab68441b6e30cc : SUCCESS in 9m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/30d27a54a2ee4ab4aa3557cfda6cd7cd : SUCCESS in 5m 25s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0a78900bf67949dcb7e54db7eef71088 : SUCCESS in 11m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c055c49e9f5f4b97aa50e394b4691a6c : SUCCESS in 8m 13s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0cbd839159734153b7bbdb10742e26c5 : SUCCESS in 16m 21s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1eb22b5ed6594c1fa11f63f29150fd92 : SUCCESS in 27m 20s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4c3482f01ade42419b4e9998c402502c : SUCCESS in 12m 30s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e467db66402d447890776e193aabd809 : SUCCESS in 14m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/624a6d7a89f542b991e3e35459a4d65b : SUCCESS in 18m 26s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3f28e4b249f4474c8977c3cf8e63045a : SUCCESS in 5m 45s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/9468a334656648249f486a671824a127 : SUCCESS in 6m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0ab046c6492d4ea6b2ff754d89328b96 : SUCCESS in 6m 48s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/cdab294ee9914ebfa3e2550ecd1f5eed : SUCCESS in 10m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ea980ada2aa34972904a90ec41eb112d : SUCCESS in 10m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4aec6bea0571482988e0b07b64946b41 : SUCCESS in 54m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/99847433c71e4c2da9c97e70ef12b5e2 : SUCCESS in 1h 13m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d232bfd4e7c8456ebc393dd5b6b0bcb8 : SUCCESS in 1h 11m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/3ca82d4ac51d4a5aa85486ffdac7d9a6 : SUCCESS in 1h 13m 55s","accounts_in_message":[],"_revision_number":1},{"id":"bf0dc1c8a895f806fb56921436f0d6775add933e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 08:43:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"edbf0416474ac694954126118c7990e3dcc8e853","revisions":{"edbf0416474ac694954126118c7990e3dcc8e853":{"kind":"REWORK","_number":1,"created":"2020-10-01 18:24:35.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/08/755608/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/08/755608/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/755608/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/755608/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/755608/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/08/755608/1"}}},"commit":{"parents":[{"commit":"3bb7de97826ce2e972cf51da8afb5b2e1871db87","subject":"update constraint for python-troveclient to new release 3.3.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3bb7de97826ce2e972cf51da8afb5b2e1871db87"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-01 18:24:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-01 18:24:31.000000000","tz":0},"subject":"update constraint for oslo.privsep to new release 2.1.2","message":"update constraint for oslo.privsep to new release 2.1.2\n\nChange-Id: I4a90ffbc9b6fd06547ec14606a61b935b15f8564\nmeta:version: 2.1.2\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta:release:Commit: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta:release:Change-Id: Ida511ad7e58f69bc8da7cb0039d091627dd62d22\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: Ben Nemec \u003copenstack@nemebean.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/edbf0416474ac694954126118c7990e3dcc8e853"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/edbf0416474ac694954126118c7990e3dcc8e853"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
