)]}'
{"id":"openstack%2Frequirements~755891","triplet_id":"openstack%2Frequirements~master~If19ad0b2d07afb94ef8b3ebc4ead4b72cc878e03","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"If19ad0b2d07afb94ef8b3ebc4ead4b72cc878e03","subject":"update constraint for neutron-lib to new release 2.6.1","status":"MERGED","created":"2020-10-02 20:59:18.000000000","updated":"2020-10-04 18:41:49.000000000","submitted":"2020-10-04 18:40:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755891-1601836825080-5cec80e2","meta_rev_id":"dd0175fb205e3913d9b6af842525e84bb7539ab0","_number":755891,"virtual_id_number":755891,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-04 18:40:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-10-02 21:42:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-10-04 15:47:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-10-02 21:43:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-10-02 21:42:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-10-04 15:47:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-02 21:42:09.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-10-02 21:43:20.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-10-04 15:47:20.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-10-04 18:40:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b833bb6283e088771f9ca4dd14261037f175f014","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-02 20:59:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"677177993d1aa3e73fc14d5efda9b2bfcca94e40","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-02 21:42:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5ce8bd5dd4a997128308ba6864b82b959dc0c79f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-02 21:43:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d7461e01a53fc9110383387a00b88649f6876375","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 22:12:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/db62158aa6b04349b2d9a875835338eb : SUCCESS in 1h 07m 53s","accounts_in_message":[],"_revision_number":1},{"id":"e4d99ba43237e4883ef99306a8f428fd3b0e15c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 22:32:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a465ec644b7e44d2bc953b58bb77c692 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb774abfbf974db69f805b51d5e8bd67 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b229957ff9ad4919b2267c17d62d600f : RETRY_LIMIT in 2m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45588e74f2e94b98a5cbd0da9311a020 : SUCCESS in 4m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3f201674e31c4cf89cb14522da15969a : SUCCESS in 4m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/37a38de464154d2d84175894fd1f744d : SUCCESS in 4m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e3540cfe0907402f9695e241372b1f25 : RETRY_LIMIT in 2m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1a4f02c9b8d046aea02b7f07ea3ad5bb : SUCCESS in 8m 56s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f5137fe9212a4643b448c9a6d0373490 : SUCCESS in 9m 32s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a2369fe1eb284c069e1c98efdf75caef : SUCCESS in 5m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/aaa2d6ea47a6402cacabdaffa0471901 : SUCCESS in 8m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5ac62e1078004909887a35038fd897cc : SUCCESS in 6m 43s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/895aa287d10841f99b979d9d5427ea31 : SUCCESS in 7m 12s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/890fa8e163d04a1381c46b4a16d0240f : SUCCESS in 15m 01s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ee5a55f6aa0841b6891bb0df2a3832ab : SUCCESS in 29m 40s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c109fd12f7da4f6b826da1a1777439e5 : SUCCESS in 11m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ec6d7aa3334e4cd2bc86a77d148dcd58 : SUCCESS in 3m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dfef83a632384f1a83d3ec04937a559d : SUCCESS in 21m 56s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/66f8df7a2f414f9995c0184b23c2fd13 : SUCCESS in 4m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/12e9cb93ed6a488ab4f77c3cc2c73441 : SUCCESS in 5m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/56c530eba6d644fcbccf6b94f3a35d0e : SUCCESS in 12m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/95f5c51367884d9ea66cb0247d62508e : SUCCESS in 48m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/89d3443211a94af8a32a4ee220a587e5 : SUCCESS in 1h 19m 14s","accounts_in_message":[],"_revision_number":1},{"id":"cc653a276f65d579e0235d4190c9bb5ba1a1a4d3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-02 22:38:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7750a394677b3f89bd2c87e9d47f22d5e4f3f868","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 23:53:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3e9b59e795e244dc97f68c9f83a31c4e : SUCCESS in 1h 04m 32s","accounts_in_message":[],"_revision_number":1},{"id":"d9e9b7d65b513ec57aece9c50f61cc88885b9325","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-03 00:05:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a69494b15f94b93ab07b50dce8af1dd : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7a1811f31f7478f9f79d95674872d05 : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b8796c542bb48c68d319b653735c1a2 : RETRY_LIMIT in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aec4175db5134f8faa7f7987fe7d34c3 : SUCCESS in 3m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b4e27f9023c64912ad48238878ae462a : SUCCESS in 4m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c55ee98884104c6ebd5fe4daf359a270 : SUCCESS in 3m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2dc39a9ff7df455ab059b4351c72071d : RETRY_LIMIT in 3m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bbb38019109a419b9a825129144e72b0 : SUCCESS in 7m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/bf379b9478aa4f0f8372938cf02efc68 : SUCCESS in 8m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9ed38d838dce4fb7ae5f098bad2ca299 : SUCCESS in 5m 27s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8485d83ac4b640dbbe7cbd2961fbe72f : SUCCESS in 11m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f56b5cac559243329e4731336733a9fd : SUCCESS in 6m 40s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5b3b205834be4bbeb8763dd216b03d7e : SUCCESS in 8m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/272a905ab30842379e9a05267cafed8a : SUCCESS in 13m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/af53dbea2da045a29eeb7c5a198a5795 : SUCCESS in 26m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4f96dd25cd004101b43290e018c41fe7 : SUCCESS in 13m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b2d8edba64534589be265aeda4e34314 : SUCCESS in 3m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e0c5a348c55942ddb7414ee7c4ca6d1e : SUCCESS in 18m 30s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d14b6c5d37454acda741de03979bdd95 : SUCCESS in 4m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9e2970b8ec194fefa95121b0b8fa5671 : SUCCESS in 6m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/763872b8130048fea245cc5ab905fe48 : SUCCESS in 9m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/535216f1e97f4edbb6cd9b9a4a837c23 : SUCCESS in 54m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e7fc43c0961d4b3085238a4258122520 : SUCCESS in 1h 22m 07s","accounts_in_message":[],"_revision_number":1},{"id":"7bf0b3f9186790c527a6538410e5d32dd61491f0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-03 06:44:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4901878460c9f520154149fe752563c3259788cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-03 07:59:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f377db45bf144d14bc06486c82edee9a : SUCCESS in 1h 07m 27s","accounts_in_message":[],"_revision_number":1},{"id":"b8c6c2e079b99ed588f18298a49f25a6c9b84045","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-03 08:27:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd77afdac6d94d7c9f35a297fd870cef : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca363b54804149ebaaad3fbb4b9b1ac2 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68df2fe4e360404ea5fbb17167987a95 : RETRY_LIMIT in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8ff28219d0f47d1ad6034c7205289f9 : SUCCESS in 6m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f641c75bd18a47b08b0d9d81af3925ed : SUCCESS in 5m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/89b4381ce5cc4872852b8f583bb813f9 : SUCCESS in 4m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d55a394f71d241c999a23f116032e8dd : RETRY_LIMIT in 2m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/85757856919a43fcab6cb8c4fe45011b : SUCCESS in 7m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4ac3833cf2fb4eefb4c19f95e52746f7 : SUCCESS in 10m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e3cfbf1d02784fcc9164fc76c06fb772 : SUCCESS in 5m 20s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e4d38b86220642ce90605c2b7d6c6b38 : SUCCESS in 10m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/28d17374b8594c9384e4d3788b026377 : SUCCESS in 9m 43s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0db6e372486b436b9aeca63806a4fb04 : SUCCESS in 6m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4e1dde2fda0740e3baa0e5da04251263 : SUCCESS in 16m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ee6529107fb645e6b782e784c90baf67 : SUCCESS in 25m 11s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f245e9689c1c45308bc922ef232cd0ca : SUCCESS in 13m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/436aee6fe1e24d94b4777e38edaaa58b : SUCCESS in 3m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0e4c251f9e9c465988dda73338652bca : SUCCESS in 13m 01s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b3b81a5525e6459597481e79564e566a : SUCCESS in 6m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9b0f6395ef2e43b78e99992656bb32d6 : SUCCESS in 7m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7c1814f3ec454dc39ae74315cadc735e : SUCCESS in 15m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19750eab671c40fda1688f3acb66bb49 : SUCCESS in 59m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dd7e27cc504b466da430d850f2eb2f45 : SUCCESS in 1h 21m 21s","accounts_in_message":[],"_revision_number":1},{"id":"74c38b2800619d238ed683a336f2109a43f7df59","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-03 19:43:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3e835b05a0fb4fcd9c0efc7c1b19bd44cc46b417","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-03 20:55:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7e0e38dea7fd4d339172345af8f7d845 : SUCCESS in 1h 07m 52s","accounts_in_message":[],"_revision_number":1},{"id":"f3ffd109d51779dcd03d10019fcbf5c6c71fa9fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-03 21:01:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddd2393e685f4d94b4304e2605449975 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd6326710ad2471bb53ee7bff2c376b9 : SUCCESS in 3m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e5c4586f6e7490093c6280577c4d319 : RETRY_LIMIT in 2m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6b6ac4ab8ba4292967d2cc9d6e2fe3b : SUCCESS in 5m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/26fa55ae92ab4e4897f76bc0ba798ead : SUCCESS in 2m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/072b0d52b7b34eb6a77ee22e87cf77c0 : SUCCESS in 4m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/94f428e02a6f4607bd779c624e4da8f6 : RETRY_LIMIT in 2m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/75873119587b45f1979bb61c5a096546 : SUCCESS in 6m 06s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/edb1ab6ddc1f4a78a86b154fddf66a1e : SUCCESS in 6m 54s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/05aae78cc61545da96e1df9da32f3ed1 : SUCCESS in 5m 09s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8bc5f8df4b6d42d8b18265d5fbce4db8 : SUCCESS in 9m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9231dee546a945beba41f792df1ccf21 : SUCCESS in 7m 08s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/74644dcdc3ef4072be9216a7a61e8116 : SUCCESS in 6m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a17d56b3487f428ca138176e18850cac : SUCCESS in 12m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f6b709d0e9234dcd8961f0f32f4357e0 : SUCCESS in 22m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0b8c3ef5e8eb44b1b49b8fd8c7cca554 : SUCCESS in 17m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fa280dae0e864f42803b878f1ebb22b7 : SUCCESS in 4m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/deab3edc84384795a54d82f97ed0bccd : SUCCESS in 13m 59s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6f0b5a01ad54445eac023c48d7be5edb : SUCCESS in 5m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e9af5e3a304a48ec83f06ad1794bb5f3 : SUCCESS in 6m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7da2e3707a8040f299d207792acbfebe : SUCCESS in 9m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a19547ec6748497a8831bcb62ba5a277 : SUCCESS in 51m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/632947c530194754b0e8e9e41547830b : SUCCESS in 1h 13m 25s","accounts_in_message":[],"_revision_number":1},{"id":"1daa746b018aaf6dc743f603b580cea1337d5ec2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-04 15:47:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3bd0abcc1c842b2d4b99dfd2acbe3f92170f4fd8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-04 16:53:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d55101fd630141bd9e454142e814f561 : SUCCESS in 1h 01m 23s","accounts_in_message":[],"_revision_number":1},{"id":"668ceae42101c06b8be0963a8c46db6fb5a2bfec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-04 17:20:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/539f9eb165b24950b347e7b199cc71ae : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4281509cf12e4f9e83d6460e6128b29f : SUCCESS in 2m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec0cf980021649eca8a5e924f5ee9367 : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35b1f74245c9454689a3959ddb12614c : SUCCESS in 3m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c622051de03f466bb929459b30f6495f : SUCCESS in 3m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9430a45194e049f1a5c45fbdb1dee3a3 : SUCCESS in 3m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/04fd4756fd9340b0b971850e2fc80b45 : SUCCESS in 7m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4547fd8ecc3f4d81b2b033fe07d853e9 : SUCCESS in 6m 19s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8e1697adc79248bab1ee940ef2be7181 : SUCCESS in 8m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c952e6c2fcd14844b248f949e51e026e : SUCCESS in 4m 10s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3beb8fa9420c4957bce1c1cf72e823f2 : SUCCESS in 9m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e547809f91384e1080cfccdc8fca0c6c : SUCCESS in 6m 45s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0f1fa55d77e34ed89dda2a8774f2d413 : SUCCESS in 7m 26s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a653c04f7c4c4b7eae2b1d8eec34dfa9 : SUCCESS in 17m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6f09551e1df04168bcf22cd9d9ddfd8c : SUCCESS in 29m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/780b2f8455974c11adca9d50d0d381c5 : SUCCESS in 12m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cc682987716f414aa7c34e3d68843df9 : SUCCESS in 4m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/88a2aa6a431e4c1aa3f399e2b105a6ce : SUCCESS in 13m 50s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/fbf100a33e84489dbd4c8af119310ece : SUCCESS in 4m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c58a55857e774425bac14d72bb7075bf : SUCCESS in 6m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b3459f25190d4d1bb00ecabe7f7988cf : SUCCESS in 9m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/974cf438cdcb47f99ff5ce388cf46506 : SUCCESS in 1h 01m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/800be9fc062a4ab9b12501afc6b6efa2 : SUCCESS in 1h 27m 01s","accounts_in_message":[],"_revision_number":1},{"id":"8427c3720f7657242ddad13834403c525e693c26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-04 17:20:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"691b361bd3ad060685ac233022d6d1222a47c71d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-04 18:40:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42406d41bc65407292c0961a4dad0a03 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40335f9e6a5341eab9ccb66c02e77fc1 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9c29945b30f4c20af835973c623d483 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f04488feb81440ea30a0ec2fd6c4b10 : SUCCESS in 4m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7b57d619c0d9426395b3439ef8d7d6f1 : SUCCESS in 4m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/79e600312635449490fe7ef48bdf3cbc : SUCCESS in 4m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e7017b69f2044c82ab5323ea54a14fab : SUCCESS in 8m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9e2e4af4a43448e4a686416e1b765289 : SUCCESS in 8m 37s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/bc0800746c1c4d989bfb524a2699f355 : SUCCESS in 10m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6f78f09b60c74c1a87f478cc3418bad6 : SUCCESS in 5m 38s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4491280043184b138efe4b6e48f1ccda : SUCCESS in 10m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/031293ed69aa436d9f901b3f3ddc1009 : SUCCESS in 6m 24s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/896dd6df320f4392a1abb4956764e205 : SUCCESS in 8m 29s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a2e98783515146d8adf700538afcf259 : SUCCESS in 14m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b65c760412254e0a8f1c1d515249d232 : SUCCESS in 24m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0bcd04edcb1b4870a834c9ebce75eee5 : SUCCESS in 11m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d1c977bc33b747178ef8af4c79bb0a3c : SUCCESS in 3m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cd0dc7b7aa8948998a3803a7231894e3 : SUCCESS in 17m 29s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/078fd6d9059f47a1a3c9d4c0f375978a : SUCCESS in 4m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3874da894e9949e8b743a49100e38210 : SUCCESS in 7m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/09a9b407f27f48f98a097aef720afb8d : SUCCESS in 9m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ca3de1bc2b3e4e51b6eda6171e77a187 : SUCCESS in 49m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5cbe491e3b834d809ca954c24b6c9c6a : SUCCESS in 1h 17m 33s","accounts_in_message":[],"_revision_number":1},{"id":"a6ec1937b379f3764cbf35d2e2ebcf3fd78e4758","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-04 18:40:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8c714735bad458216e5bf9959914aaa5e5adb339","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-04 18:41:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/094508a2410d4d21947a2f3241d068ff : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"69c7d75ccd640f6ce25bab65708c14f839bea414","revisions":{"69c7d75ccd640f6ce25bab65708c14f839bea414":{"kind":"REWORK","_number":1,"created":"2020-10-02 20:59:18.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/91/755891/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/755891/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/755891/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/755891/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/755891/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/755891/1"}}},"commit":{"parents":[{"commit":"346f659eec09403fcfa1897417347f99bf41b606","subject":"update constraint for tripleo-common to new release 12.7.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/346f659eec09403fcfa1897417347f99bf41b606"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-02 20:59:16.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-02 20:59:16.000000000","tz":0},"subject":"update constraint for neutron-lib to new release 2.6.1","message":"update constraint for neutron-lib to new release 2.6.1\n\nChange-Id: If19ad0b2d07afb94ef8b3ebc4ead4b72cc878e03\nmeta:version: 2.6.1\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta:release:Commit: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta:release:Change-Id: I7fdc133bc0d1883d921c965028ffd363616cf49e\nmeta:release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nmeta:release:Code-Review+1: Balazs Gibizer \u003cbalazs.gibizer@est.tech\u003e\nmeta:release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/69c7d75ccd640f6ce25bab65708c14f839bea414"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/69c7d75ccd640f6ce25bab65708c14f839bea414"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
