)]}'
{"id":"openstack%2Frequirements~755893","triplet_id":"openstack%2Frequirements~master~I5f7bcebc039581be5884ba63f611053fe31db84d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I5f7bcebc039581be5884ba63f611053fe31db84d","subject":"update constraint for python-openstackclient to new release 5.4.0","status":"MERGED","created":"2020-10-02 21:06:18.000000000","updated":"2020-10-04 18:02:10.000000000","submitted":"2020-10-04 18:01:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755893-1601834461555-5c26715b","meta_rev_id":"ca67d417cab0ef8639559c3250a27721d5a525f5","_number":755893,"virtual_id_number":755893,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-04 18:01:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-10-02 21:42:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-10-04 15:47:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-10-02 21:43:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-10-02 21:42:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-02 21:42:11.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-10-02 21:43:37.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-10-04 15:47:12.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-10-04 18:01:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"58b7b9ee943b4af422f0aca6123e77fcd7324ec9","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-02 21:06:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"93d3063bb921e0629fce7b3c5c8184c4909d0366","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-02 21:42:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4047031d234dff618126d5ec2defc544438b3543","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-02 21:43:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c18a1c770eb2ff62e92ba2885541db50c4758e89","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 22:17:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d5a80d6c9a574a75b05590569d000b19 : SUCCESS in 1h 06m 58s","accounts_in_message":[],"_revision_number":1},{"id":"d4176f99151aeafb1ccc3c7502a015d8eedddb3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 23:14:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c6d8553619e4a55bdd763b1eac3af92 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17de63de75544ce0866cb6389b129104 : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a67288d5b1a43cd8af03f7babba176c : RETRY_LIMIT in 3m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/152b0c38b99b48c3a58de94386972a25 : SUCCESS in 6m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8f4522d40ea148018273ab1d6e205da3 : SUCCESS in 4m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1046e958b5c946178e2d51c43be4aced : SUCCESS in 4m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c3df9a826d094256b043b741b456b5d0 : RETRY_LIMIT in 3m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2f2da7b14c9f4f0aa18ee88571f3ad5a : SUCCESS in 6m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f70a941337034f0ca3dcdf12bb67ab4d : SUCCESS in 7m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c37a6bdc0cf64fd8a12f7ac841ade4df : SUCCESS in 4m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a05187081e3a459b96846a0775f3184b : SUCCESS in 9m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/efc25ee196b442d1a6545536f2b50089 : SUCCESS in 5m 56s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5aa800577d6847a4ad537fd299a046de : SUCCESS in 7m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/304523a4dc0848e08d3459c683a6def2 : SUCCESS in 14m 35s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6239ae09d50446cc90d4a8608c9ee402 : SUCCESS in 29m 18s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/13a8af0acda0466fa9f9e31c941b5c50 : SUCCESS in 13m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3ed105192d6748a5adb5d3178c4d523f : SUCCESS in 3m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f206928d768649efba009aefd5f9dd7e : SUCCESS in 17m 17s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d2c8d87bbb4a4a458cce0f9c667860e5 : SUCCESS in 4m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1ed5ae5c2c8346ceb93e7ea15abe2832 : SUCCESS in 6m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/450f42dee9e642f3a3f2c74570007064 : SUCCESS in 10m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a70f95c1c8224972a612d8717e363e3f : SUCCESS in 1h 23m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0d9d1b23708841be8c2fa163e67bbfe9 : SUCCESS in 1h 08m 43s","accounts_in_message":[],"_revision_number":1},{"id":"b1693bf3d72610085e1f953d0920bb4de19310a9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-03 06:44:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c9652bddd7cdaacc3896a870d74cecc7e0252294","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-03 08:10:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/75f7d38cbb174fb7a0ab43e1f20f44b3 : SUCCESS in 1h 12m 14s","accounts_in_message":[],"_revision_number":1},{"id":"31649e6efe034b96e210a075f2faedd8283b7bca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-03 08:28:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a2ae12bc8b34ccdb8af6de8e3b461cf : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/181a4dbf5e8b459bacd887a3088d841b : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29c39d3ec11840038af0b5810c36e3b2 : RETRY_LIMIT in 2m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb9235f20bae4de9b42a9606b307907e : SUCCESS in 5m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4b0255fb095d434f8c164ee4a25ffbb0 : SUCCESS in 5m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/63ae63479a014219ab1aa97ce9b45f25 : SUCCESS in 4m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/828acda21e7246729ac15cb022ca94fd : RETRY_LIMIT in 2m 32s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6bd83c59dd384b12a7648cea153ec6c1 : SUCCESS in 6m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e12931fb36744b71ad5770f76d44af54 : SUCCESS in 8m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/84f78d8858104e558554e36c2579ffff : SUCCESS in 4m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4ae4717094a44b19bbad069b1175fcae : SUCCESS in 10m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/42bb756b4c3145119165155eb60138ce : SUCCESS in 7m 04s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/837f3da24f834c6f87b92cc1656bb6b6 : SUCCESS in 6m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f430613c93784e179092bf8e1d416b49 : SUCCESS in 15m 01s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/82a4f790509c419a9b9a7a8a69bc9ffa : SUCCESS in 34m 36s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/28cf9131386349359d1f007db1bff245 : SUCCESS in 13m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/913596b8499544bfba677363f036be38 : SUCCESS in 4m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/197b3f71ff494fcda5c58e10a10397e8 : SUCCESS in 16m 33s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b41a406d88e345c3b2ccee4787637eab : SUCCESS in 5m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5f60f959636c430593e5c4106fb86f1f : SUCCESS in 6m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/28a5c702897a48778e1bbcca1f0d64c9 : SUCCESS in 9m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f139673b57104261aa34c928482a6aac : SUCCESS in 59m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f026305924e24b5b935af625d38e8616 : SUCCESS in 1h 22m 01s","accounts_in_message":[],"_revision_number":1},{"id":"8f356775bdce9b7772cacb668178a57f86a01998","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-03 19:43:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f4150ae829acbbe4effbfc6614071ccc93740b7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-03 20:50:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d12236f8ed6846b4891c36128576d013 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c028da0ca0644c3fa55e85f6f28683d1 : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b301730cc7d44b44b45f5bf709337b09 : RETRY_LIMIT in 2m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a1987c276f3745cda783140a51c4d6ac : SUCCESS in 6m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ef8b933192634807b700c9e8fdcf1741 : SUCCESS in 4m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/289474abb826469ba4d733809dcc2a89 : SUCCESS in 3m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a0b63b0a524a45b7a1b15a9e8c4ca0d8 : RETRY_LIMIT in 2m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/85cb56f71ec4477ca9c801a13e8a212b : SUCCESS in 5m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/23ed2632099f45fb9dd08d841719cc2b : SUCCESS in 9m 08s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5b4292b9a4334a44b6b4ec45ed865f7d : SUCCESS in 5m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/308f36ae37e747cdaae538797f3d29f9 : SUCCESS in 10m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/de8b1d63315846f2a5745de150b08358 : SUCCESS in 8m 05s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b5113f5d840f415d94539b355e64f819 : SUCCESS in 8m 09s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/355774ae98774b0bbff75479c9ca2e80 : SUCCESS in 12m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b49cc349d48c41c4876d688dbc3d43b9 : SUCCESS in 24m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2bd5ae707658403e9ad4f2fe41400cde : SUCCESS in 9m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1d901a0819554beda21307d6988bdee7 : SUCCESS in 4m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a25c2e76e8c64fe5b0ea811582736c1a : SUCCESS in 19m 06s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/196aa918c0eb4251be18edb9df1097f9 : SUCCESS in 3m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e16f5794fd5a4f35a79520cd416b5a8e : SUCCESS in 6m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/346e5dacb4424cbb813125cd34b63f83 : SUCCESS in 9m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00af18379b664454b69261be2dc316cb : SUCCESS in 1h 01m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cbc0f7a71d77471c92e16184848ef506 : SUCCESS in 57m 59s","accounts_in_message":[],"_revision_number":1},{"id":"5b8081a9d5e2f336c3d2b34741d78fcd4b016a5a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-03 20:55:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/bd09bc07d0194eb5b06125f190b6dd85 : SUCCESS in 1h 08m 27s","accounts_in_message":[],"_revision_number":1},{"id":"1c35546d1ac4036f8cc504924744009945c53c17","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-04 15:47:12.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d889b59457c4a450cbc161a44f6355ea91105285","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-04 16:44:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48b4d9f5060d4581979393fd38980134 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8eec6a8069f14fb48a8424e27dd4b4d0 : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f37edc6b8f04d9eb1ceac91901725a0 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4bc3202b9e648a7b43675ef3c62fe07 : SUCCESS in 4m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0d7bc3e28e2f4eb5a689cde46ae788dd : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8689ce895c8c4f44b969f5c4f0e7a125 : SUCCESS in 3m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/34e3d41170bb4c548c3e262f2a84cc63 : SUCCESS in 6m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d83fb75a13234f6eb3761afb1ddee046 : SUCCESS in 8m 51s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/92aa7116cc5c4ca0995e5e96aaa81ebc : SUCCESS in 8m 43s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ecb778fd615a44c9b8857deefbade28b : SUCCESS in 4m 21s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f24911c6062149779480a174472aff3a : SUCCESS in 9m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/01b62e2605c64477b1f97d4d7f5672f3 : SUCCESS in 7m 46s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8ebf7cfba27440cc925ce84241097c05 : SUCCESS in 15m 13s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fce631ce8ca34575b158752c284a2041 : SUCCESS in 11m 36s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e83bbf6986494544b5df94bf54f4fe0d : SUCCESS in 26m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/87b6e81cfd3d4cc48405edf187da2823 : SUCCESS in 12m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f43ee34ee5bc48899a2ba9af546b41f4 : SUCCESS in 3m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/de793a14f4214b9fb5aede46820457c7 : SUCCESS in 16m 19s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/58b663fc15964b91baba080fd1830914 : SUCCESS in 5m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/02413b7a494249f780b3a93264b7d4cc : SUCCESS in 6m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e89d479cc50745b7b0e98979a6bbc76b : SUCCESS in 9m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3551a7a63c63417dbf1529bea677fea8 : SUCCESS in 48m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e8d581a94b454f92bb783db03595a348 : SUCCESS in 55m 52s","accounts_in_message":[],"_revision_number":1},{"id":"9ed7ff180ffd2e200755c034f8e580e5ae327dbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-04 16:44:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6720b9b157d784ffe5aeae5167d5c86b4bb4ac57","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-04 16:53:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ce395542a2304a189e8de62d1c3aa947 : SUCCESS in 1h 02m 48s","accounts_in_message":[],"_revision_number":1},{"id":"e70dfa2dc3cdf477892d90a4ceff0af9e8304423","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-04 18:01:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd708e67d5524e02a7330c5f360ea336 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1b2826c711747598cb5d621f06a99c8 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a34fba7468ab463897336ff349b5fba6 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59b9bf63be7b48f585aca4df9ea0b0ad : SUCCESS in 6m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c532867897484f5f8e922dd010762fd3 : SUCCESS in 5m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5e7bcc2ebf9841fa904a0318c3fb8446 : SUCCESS in 3m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ecba513bf4b64b7baf493931f3ed7f52 : SUCCESS in 6m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/443e994c80db4d1c8de330a04f14c4f7 : SUCCESS in 6m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e40c87b5a18e486bbc0061c0974597a2 : SUCCESS in 9m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/887ef4fd500e45569541cf28bce5da1f : SUCCESS in 5m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/41362d489931448da5b39c8c6f54231d : SUCCESS in 8m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6feb8a95b18e44af80070ad959656d6b : SUCCESS in 7m 10s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/40c5b09a62b54a86adb2b2ae25d2eea9 : SUCCESS in 6m 28s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/029e75ba36d14a6ab67d65d5d2493895 : SUCCESS in 14m 02s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e89ba32969c24ac78f9159de1c24b7fd : SUCCESS in 25m 37s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b846d73a05614fe7bb238f269089b13b : SUCCESS in 13m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a4add39b7cf54dde95165b76c75c225e : SUCCESS in 4m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8ea9026b7aef491d9c3a68135ca33331 : SUCCESS in 17m 11s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/23429bb8d6ff49b9977bbcb3c1d7f0d8 : SUCCESS in 5m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cb0f5208b305410f8c1276789a02c281 : SUCCESS in 6m 40s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f3356768302f4ceca33c330f38095002 : SUCCESS in 10m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f89c89bda1347df8f05e25c00b0fe50 : SUCCESS in 53m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b7ecaa59af89421c87b536e04b4e87eb : SUCCESS in 1h 09m 47s","accounts_in_message":[],"_revision_number":1},{"id":"e25174efee606fd2d0be2754370a2b8428328340","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-04 18:01:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7725c9638104bf7a529d598634942f42099eb0b9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-04 18:02:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c965c8552d349d5b2fcf64abed46e24 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7d52a955a9e6d9cd14a479a07eba7d75c7394635","revisions":{"7d52a955a9e6d9cd14a479a07eba7d75c7394635":{"kind":"REWORK","_number":1,"created":"2020-10-02 21:06:18.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/93/755893/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/93/755893/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/755893/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/755893/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/755893/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/93/755893/1"}}},"commit":{"parents":[{"commit":"346f659eec09403fcfa1897417347f99bf41b606","subject":"update constraint for tripleo-common to new release 12.7.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/346f659eec09403fcfa1897417347f99bf41b606"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-02 21:06:16.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-02 21:06:16.000000000","tz":0},"subject":"update constraint for python-openstackclient to new release 5.4.0","message":"update constraint for python-openstackclient to new release 5.4.0\n\nChange-Id: I5f7bcebc039581be5884ba63f611053fe31db84d\nmeta:version: 5.4.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I45570ff270ad9c2d9134fd088d355af90e974832\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7d52a955a9e6d9cd14a479a07eba7d75c7394635"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7d52a955a9e6d9cd14a479a07eba7d75c7394635"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
