)]}'
{"id":"openstack%2Frequirements~756136","triplet_id":"openstack%2Frequirements~stable%2Fussuri~Ib17ea2b9941fd086b1d9326a483e063115eed69d","project":"openstack/requirements","branch":"stable/ussuri","topic":"new-release","hashtags":[],"change_id":"Ib17ea2b9941fd086b1d9326a483e063115eed69d","subject":"update constraint for tripleo-common to new release 12.4.2","status":"MERGED","created":"2020-10-05 17:19:57.000000000","updated":"2020-10-06 06:36:55.000000000","submitted":"2020-10-06 06:36:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756136-1601966215074-3de226de","meta_rev_id":"ac30925d919e20e73513e57411bf9d11919a5606","_number":756136,"virtual_id_number":756136,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-06 06:36:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-10-06 01:09:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-10-05 17:20:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-10-06 01:09:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-05 17:20:22.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-10-06 01:09:56.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-10-06 06:36:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3fdbac4aaec93f9eab5497a4fad728e47b7fae5c","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-05 17:19:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"11f1f8159f6677f14848182b579c81d124af4607","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-05 17:20:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ac3ee64f2218efcda20f1147e8fdf678d42a13f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 23:15:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9170a0c81e1c421da6bebdc290cbda98 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67c4445ed68d451796ad563d7b374c82 : SUCCESS in 5m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/46a8d3064b194e7897db4ba5676cee74 : SUCCESS in 5m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3208f2a8483e4e84a77c776942c13b5f : SUCCESS in 3m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fd7a3dd0827a4f75b283c7e8f4bec5fc : SUCCESS in 3m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/620dcf5afcaa46a5bf8e8b93346aa754 : SUCCESS in 6m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cf364a5f27bb4a56aac8644308cc1242 : SUCCESS in 11m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ea29f0499edb4804b6944e76c122d6d7 : SUCCESS in 5m 36s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9ab0b3f2b83e4277bd7ce54be7101ce4 : SUCCESS in 8m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/78e72a20900841c782b5865da77659b0 : SUCCESS in 6m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/382ff541f9b54fd2a3095684a68d98ee : SUCCESS in 14m 24s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/607c37a1b9624df9a43495da5cede5f1 : SUCCESS in 25m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/003cf43d8f1f460e80a31425315ee362 : SUCCESS in 14m 49s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/11b36b8f92e540cc9f6f58e8ca5209b9 : SUCCESS in 14m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/af7266f2f907487c9777f1d58e2bf020 : SUCCESS in 15m 56s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a92786ddebaa4ac5b8c9d11d47acd6dd : SUCCESS in 4m 19s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/78d17454845e447aa1411d3c5f9ac4b2 : SUCCESS in 5m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/41604896de2e4e76b0105c862dff5eec : SUCCESS in 6m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b65f9e34fbd749308070d241f1d9ad08 : SUCCESS in 31m 27s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/81024600b6b84e279a3ad0967d29b1a1 : SUCCESS in 11m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8b4484a4cf774bb790ae3e26c86d3627 : SUCCESS in 53m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b3592584ceec40319887c82078c4348c : SUCCESS in 1h 28m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0849fa4b4fd14015ba5023637cf8109e : SUCCESS in 1h 13m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/a965e2ca1ed14cb980ebb19c8c77801d : SUCCESS in 1h 26m 09s","accounts_in_message":[],"_revision_number":1},{"id":"96ab906353985a84f4c89226c50699eda5a2717f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-06 01:09:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"09a41630b50359e6a72b4afdc14840c50bdd7198","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 01:10:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"568d294ecf7eddd9d8a7dff5bb08dfa273d014d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 06:36:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a0894c7753c40b09546a33dec4e324e : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96a5437d381f48768be0c175af5f40f0 : SUCCESS in 5m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6217240bfc94437ba85e5901e5bc2d16 : SUCCESS in 6m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1e101eeff3ac4ce9907d3580ca838b6b : SUCCESS in 5m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0edf71e08ea24bcdbf0cd85eee235b26 : SUCCESS in 6m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1c5a58238dbb4b52b361f44f3dd81cca : SUCCESS in 9m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9668cb0918284025ae2da132bfe26656 : SUCCESS in 11m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/18cd0b197b8347068b83bd9f4ee742c9 : SUCCESS in 6m 44s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e0711da73f4840818e55351b4e86bfff : SUCCESS in 8m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dccc4c86e57c4198abdf4cced1ab4c40 : SUCCESS in 6m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a80e8283e68048d68b4689f95327be32 : SUCCESS in 13m 53s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cce854badd824378a9b9dd59ec5a42dd : SUCCESS in 19m 04s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/004a5b38e97b45c38c8788369523403c : SUCCESS in 14m 38s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f1db088d5a064587868945b9e831d455 : SUCCESS in 14m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ee7deeefea944c609c1a094aa2ddda4d : SUCCESS in 13m 35s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6d587717a93e4501ba73242fbd52dc7e : SUCCESS in 5m 33s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/f8c72705cc8643e4b232855c5a0f5d5a : SUCCESS in 5m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c734a00d23714b63a40dbb6de41306ef : SUCCESS in 7m 09s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/058a8e09208a4ae2bd37f2610bf11462 : SUCCESS in 10m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ecd32869ceb24cd2b90b599b7fdbab4c : SUCCESS in 9m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/69c22e03bfe64eccb20f53bdf77d4b89 : SUCCESS in 51m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/008e710f0ab04c8f81bd811e051883ea : SUCCESS in 1h 14m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/22e59fa77c3245f9a14a644271135cbf : SUCCESS in 1h 15m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/8c44cb4f1b9e41bf919318e60b4aae32 : SUCCESS in 1h 09m 33s","accounts_in_message":[],"_revision_number":1},{"id":"50bc3de1d2433dd309c0e71f2155f80e6d20c48c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 06:36:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eed2f17abd07d166f691b8a4949fee57f9548360","revisions":{"eed2f17abd07d166f691b8a4949fee57f9548360":{"kind":"REWORK","_number":1,"created":"2020-10-05 17:19:57.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/36/756136/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/36/756136/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/756136/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/756136/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/756136/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/36/756136/1"}}},"commit":{"parents":[{"commit":"3bb7de97826ce2e972cf51da8afb5b2e1871db87","subject":"update constraint for python-troveclient to new release 3.3.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3bb7de97826ce2e972cf51da8afb5b2e1871db87"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-05 17:19:52.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-05 17:19:52.000000000","tz":0},"subject":"update constraint for tripleo-common to new release 12.4.2","message":"update constraint for tripleo-common to new release 12.4.2\n\nChange-Id: Ib17ea2b9941fd086b1d9326a483e063115eed69d\nmeta:version: 12.4.2\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: MSA \u003cmarios@redhat.com\u003e\nmeta:release:Commit: MSA \u003cmarios@redhat.com\u003e\nmeta:release:Change-Id: I803eafba77e45fecc00d15ab2c7e72294363ca27\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Alex Schultz \u003caschultz@redhat.com\u003e\nmeta:release:Code-Review+1: yatin \u003cykarel@redhat.com\u003e\nmeta:release:Code-Review+1: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta:release:Code-Review+1: wes hayutin \u003cwhayutin@redhat.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eed2f17abd07d166f691b8a4949fee57f9548360"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eed2f17abd07d166f691b8a4949fee57f9548360"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
