)]}'
{"id":"openstack%2Frequirements~756137","triplet_id":"openstack%2Frequirements~stable%2Fussuri~Ibde9ae732142df0ab274c94ffc3ac3bc44b13f16","project":"openstack/requirements","branch":"stable/ussuri","topic":"new-release","hashtags":[],"change_id":"Ibde9ae732142df0ab274c94ffc3ac3bc44b13f16","subject":"update constraint for paunch to new release 7.0.2","status":"MERGED","created":"2020-10-05 17:20:56.000000000","updated":"2020-10-06 18:04:22.000000000","submitted":"2020-10-06 18:04:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756137-1602007462814-8f1aa1ae","meta_rev_id":"379311f6cd9530f898b0822a2e691be062a656d5","_number":756137,"virtual_id_number":756137,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-06 18:04:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-10-06 01:09:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-10-05 17:21:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-10-06 01:09:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-05 17:21:40.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-10-06 01:09:58.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-10-06 18:04:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cf22417c0d2c6ad6e08e45249415c78bc6da0bfd","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-05 17:20:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8124b229b9051cb0c997c09cc18d25f8d6dddec","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-05 17:21:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f9128291a39c21c25b093a790ffa26d9b5695c02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 23:18:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03dec4098ba94cbc9228e9b20dbc07a7 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d8745071e3b497daf26dc742fe8c5c8 : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/418aba8c234042c394bf1e3a469e6298 : SUCCESS in 4m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d0097700a47747148b51375e9cdb970c : SUCCESS in 3m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3f68da01f811493e95decc59347bf492 : SUCCESS in 3m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d053cfbffcc3424e815e5e4a8530eff9 : SUCCESS in 8m 09s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5c67220b0209462abe8cde2a888b96f5 : SUCCESS in 10m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8d8375633965495b9a5027259eed283e : SUCCESS in 5m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a834e2f17fa147d9a1140b6be3e28464 : SUCCESS in 11m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d7dfd97605414566a6f57fbba4260521 : SUCCESS in 6m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/775f0a04119e4d3da4db8719a68ddbee : SUCCESS in 12m 59s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/af84f110df334694afe6765a9d93ca8e : SUCCESS in 27m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/062a160b6580464cb9ee37df78d6a6ba : SUCCESS in 15m 50s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/15d8720316914cacb3a2c2de6e11d352 : SUCCESS in 14m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/12b2079a8e4f498f8dd92efb8a9f198a : SUCCESS in 20m 16s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/05964b32c2d4469fa5dc25756eedbf72 : SUCCESS in 5m 37s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/228c253c65db49e8bbae30d12a00e054 : SUCCESS in 5m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0799fe1cdb834a17bb80d461fc85d148 : SUCCESS in 6m 43s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0c35c3f700d04616aac11803e3283aa3 : SUCCESS in 11m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8f35142ec98541fbb935941c3809a063 : SUCCESS in 10m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b5f8317e2d254cb791f2c065ced2dcaa : SUCCESS in 48m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9b6329f124f04acea2b9ead3d5579ac7 : SUCCESS in 1h 20m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/708d174ac09d43ee99f125d2eb8845bb : SUCCESS in 1h 25m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/b214fc97ca224a10a9dd7147be15eca0 : SUCCESS in 1h 16m 06s","accounts_in_message":[],"_revision_number":1},{"id":"a02834877910a93d154c384a2d3895c8a57a06da","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-06 01:09:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"28eec5768f3cc21dbe3eed0b3a86069fa8549418","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 01:38:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"93cbe499a9f388d7e3ffee91b37828756e7f2f5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 06:36:56.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01502947ce444a8db55e3e35b11e7e4c : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be5ffab15db5403b83227c0253260c76 : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9a0ece64b41b4387a2e7ab0eded46e12 : SUCCESS in 4m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/95999a839ff14a94b550fb9baf4216a7 : SUCCESS in 4m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1be07ea00f9845ca9a005a441f5498ca : SUCCESS in 5m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e5f3152906124df7b426a60e5bec1a87 : SUCCESS in 6m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e8fe262ef0a846bdbba68fe8ff19b9c3 : SUCCESS in 10m 08s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2c3e4fe5f56b456caeee1622d6857ad4 : SUCCESS in 4m 19s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0d44dc69fd5a4a53a489fde82ae248fa : SUCCESS in 8m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7653baa6a0e84f7682e4f31c2c58418c : SUCCESS in 5m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4df4136efcd641159d537a21343c9172 : SUCCESS in 14m 35s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/972cc7ad160b4bbda0674089af0ab6ba : SUCCESS in 29m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8895a1a0616f499ca1aee3faf25e54d3 : SUCCESS in 14m 10s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4f82da6c0fca4c35839841ebf12e6359 : SUCCESS in 15m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/275a21c2dd4e404a860c61a3a9df4bbb : SUCCESS in 14m 20s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8e2209070b0f4f7e887cd3230c38fc96 : SUCCESS in 5m 00s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/9ccd7155a47d485288747bbd136dfad4 : SUCCESS in 6m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9b1282059b0d4360a7693d0eacb3385e : SUCCESS in 6m 51s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d369893211ff443c97c392a26f622b64 : SUCCESS in 14m 15s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/639efbefc5894b47860880cbb99167ea : SUCCESS in 12m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/82344f117da4464491d0311e8b47ce57 : SUCCESS in 51m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8dafcf163c414ade9b6d58dbb9e31f60 : SUCCESS in 58m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/84aae9b00dc04dedafccd65000c759a7 : FAILURE in 1h 05m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/e04ebe07fb064f1692b2d854baa9ede9 : SUCCESS in 1h 12m 25s","accounts_in_message":[],"_revision_number":1},{"id":"29db04b1cc9ed987a989fab02ff30579872d861e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-06 06:42:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d19a15c7f81ababff79bf215d8c5b3466832a95d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 11:23:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0846f1c54d7249f7b50c2ebdb6a47ee0 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a482ed5dadc4c4d8f61773c6c7b1d29 : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2042faaf09224fff9803955d1037922c : SUCCESS in 4m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/145362c2c45541f5ad65b871249c14d7 : SUCCESS in 4m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5e7c5a6fbd4245e49c679e1217dae312 : SUCCESS in 4m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6e719b87b226414a9352a89ed4bceaa4 : SUCCESS in 8m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/66acc8425acf4dd9adfb3768be258c8e : SUCCESS in 7m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5b1b0d518b6d41b7b77e86d2e8a69e9c : SUCCESS in 5m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4a8b21708fae454683847ecc0c1c53cd : SUCCESS in 10m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/74e72b00792347fd9d4b49dff348ac76 : SUCCESS in 7m 49s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/15ab519d99fc40ee99d3911b42769ec0 : SUCCESS in 13m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/46f524cbdc3a42f293b533af8af48550 : SUCCESS in 25m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/df2dae9e153d4f5e9da0e16ca0ea81cd : SUCCESS in 12m 13s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/2f6930dfbe994f108e617abd91603e0d : SUCCESS in 13m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e2dc8b2707b4420aa40f983dcef3ba15 : SUCCESS in 16m 30s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c5fa7d2bfce24ed485b73ea0d867e612 : SUCCESS in 5m 07s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/d0110c9e01b7409abd7ac31c6f07e928 : SUCCESS in 5m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e2ac8e09325b4b93af85b34cca3ec1d9 : SUCCESS in 8m 03s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/96e093d8c7fd4cb0a6e49fd8e817c16e : SUCCESS in 9m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d4c9438954f94dd782f905cc687ca1ea : SUCCESS in 10m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/da41a74c288d4f2199bf1ef1a8b3e697 : SUCCESS in 57m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/81df940188f94f068b1f0277cb994c9a : SUCCESS in 1h 14m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/35bb06e0b4a14768b700ec365dce36b3 : SUCCESS in 1h 16m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/70662b5639994b0a9a647b7cfdf3050a : SUCCESS in 1h 19m 48s","accounts_in_message":[],"_revision_number":1},{"id":"953ecd5944d6be22e53bf9c44079f8a94702465c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 11:23:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fe34c22c3444dfc4ab88c5518bade97d1639f785","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 18:04:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"06e35ff6f0633c92899ea39c8dfb71c7b43cb7fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 18:04:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c47449a47df45a6bbfdb67574fe4724 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e3370d0aeac4898a2e0f7fd584deab8 : SUCCESS in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bea5f4bc0d6a46e7b0a1203bb59422b9 : SUCCESS in 4m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e759388a784c4d3b99aa98b87c8d0f36 : SUCCESS in 6m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9aa49cb70af54742b3f6a7b40b5d8139 : SUCCESS in 6m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/eaab59ffeaa0432f85e237942f13721d : SUCCESS in 9m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fd009ae519e647189ec0eb08817973ee : SUCCESS in 11m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/41cd4fa0e03f4d5faf71f549ed5d6073 : SUCCESS in 6m 11s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e4580b0f177c4e0e84229d9b08af9f16 : SUCCESS in 12m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/54759876c16d4875a108d20e73d4a03e : SUCCESS in 9m 13s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b8fe5f4b20b7466481c67dae2c8f3dd4 : SUCCESS in 15m 58s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c7f6021f897848feacc226fd6934299a : SUCCESS in 28m 43s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2afe2863348d445ab3d41e99ed2283be : SUCCESS in 13m 34s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/abb315221f5c4e6d809509f48b1069a3 : SUCCESS in 13m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/58b478e124444f499baee21d7bc8d94d : SUCCESS in 17m 14s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/282fbcc3a1834c0790402ccee9de678b : SUCCESS in 5m 32s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/0a81dba2f69446b0bb839602b2fd3bad : SUCCESS in 5m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e331bb6328134230b3e3b826c28c1784 : SUCCESS in 7m 48s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8c54066a590441169d598a5646ab747c : SUCCESS in 10m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/99be34e2cc694931baf001fbc2f07cba : SUCCESS in 11m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dce1fced3c6145b294f3b63c91e19de6 : SUCCESS in 56m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7e1a4e2d965a4b57a519930bda7b3110 : SUCCESS in 1h 20m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f8f2d78b691b4a6e9277fabff3ed2d01 : SUCCESS in 1h 13m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/53e5f5c6d86140c0a8edc1d5e551d335 : SUCCESS in 1h 13m 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d0c923ccc4dd854ba707ce92bc06ec413e36dd2e","revisions":{"d0c923ccc4dd854ba707ce92bc06ec413e36dd2e":{"kind":"REWORK","_number":1,"created":"2020-10-05 17:20:56.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/37/756137/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/37/756137/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/756137/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/756137/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/756137/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/37/756137/1"}}},"commit":{"parents":[{"commit":"3bb7de97826ce2e972cf51da8afb5b2e1871db87","subject":"update constraint for python-troveclient to new release 3.3.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3bb7de97826ce2e972cf51da8afb5b2e1871db87"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-05 17:20:55.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-05 17:20:55.000000000","tz":0},"subject":"update constraint for paunch to new release 7.0.2","message":"update constraint for paunch to new release 7.0.2\n\nChange-Id: Ibde9ae732142df0ab274c94ffc3ac3bc44b13f16\nmeta:version: 7.0.2\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: MSA \u003cmarios@redhat.com\u003e\nmeta:release:Commit: MSA \u003cmarios@redhat.com\u003e\nmeta:release:Change-Id: I803eafba77e45fecc00d15ab2c7e72294363ca27\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Alex Schultz \u003caschultz@redhat.com\u003e\nmeta:release:Code-Review+1: yatin \u003cykarel@redhat.com\u003e\nmeta:release:Code-Review+1: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta:release:Code-Review+1: wes hayutin \u003cwhayutin@redhat.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d0c923ccc4dd854ba707ce92bc06ec413e36dd2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d0c923ccc4dd854ba707ce92bc06ec413e36dd2e"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
